)]}'
{"id":"openstack%2Fkarbor~755211","triplet_id":"openstack%2Fkarbor~master~Ibd28376f97c7e3be897af5d60dfe0bcff4f08961","project":"openstack/karbor","branch":"master","hashtags":[],"change_id":"Ibd28376f97c7e3be897af5d60dfe0bcff4f08961","subject":"Bump py37 to py38 in tox.ini","status":"ABANDONED","created":"2020-09-30 07:36:48.000000000","updated":"2020-10-09 03:31:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"45ca1cc76037d964ca7c112a08dcaa09dfa2c8ff","_number":755211,"virtual_id_number":755211,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-09 02:25:54.000000000","_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-30 09:18:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 09:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebda3cbbd1f256e201db5dd4a0362ea57f8b6536","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-30 07:36:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5aaf92075958e5322762efeeddfab59aec494d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:18:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a68b02f9b9254aebbe61a352b4e57532 : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3ca1d7304284e70b8e598ec7f67d34d : FAILURE in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ce30882d3664c82b831fdc1ec3b8a6d : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c332fbafd91450287b0b973fdef63f8 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e826601d70ea4837b8384e7599de59b9 : FAILURE in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a65482b2594e46d09f2a7657b259b2b7 : FAILURE in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdfd0c95b6a47f6875d97314ce7bc79 : SUCCESS in 5m 53s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/55fa13a997334c0ebe9dce7a03130e60 : POST_FAILURE in 1h 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb74995b6fd07a9b8d46a0bb47d2316576ee9dc3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-30 09:45:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cff3dd6d35abebed49a7da1e8d0b49c03102938a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 11:23:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e705cd077fd49d6bfd0deae57a8202c : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5db8a674870c4f5d94ae3635f8079b35 : FAILURE in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cd9b8e442de4284a0c94ac1147ecbf2 : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf93f397c294eb0a1fdc21fe181035c : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/324de9e9ebb847c2b611208322289b14 : FAILURE in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dd7adf1df774b8ab134ade091c7ed76 : FAILURE in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6a768850ad4bb79d5e93835eb4037a : SUCCESS in 4m 30s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/6f3d5d307bb040ef9435f735377bbae1 : POST_FAILURE in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb91df5f2ea1f12b15eb3a03e29498bb52870167","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-09 02:25:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f567580584fbcdc2bee5f513af02e94b4367c1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:29:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0fa68d24cbdd480ab5bb9292ad154f7b : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3a1d1a592f84ac4b18c71afdb442214 : FAILURE in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/737ffdbe25d94b05bd97889071995f53 : FAILURE in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbce93b09954e9d80e0cf122cd26ef0 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b0c49e64ae43f3ba29d9f634a223fa : FAILURE in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9352d6c5cc714909b6cede046bee5bdb : FAILURE in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b192efeccb314034ab052744a22f53a9 : SUCCESS in 4m 46s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/e56c0c618b3843b79110c2c3aaaf43eb : POST_FAILURE in 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45ca1cc76037d964ca7c112a08dcaa09dfa2c8ff","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-09 03:31:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4e13f911b48ca89c85f9268cb843d9ab6254639","revisions":{"b4e13f911b48ca89c85f9268cb843d9ab6254639":{"kind":"REWORK","_number":1,"created":"2020-09-30 07:36:48.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/11/755211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/11/755211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/11/755211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/11/755211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/11/755211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/11/755211/1"}}},"commit":{"parents":[{"commit":"5e80d0c07b8f0a4064b141129406b0f3a0057448","subject":"Replace assertItemsEqual with assertCountEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/5e80d0c07b8f0a4064b141129406b0f3a0057448"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 05:56:19.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 07:17:01.000000000","tz":480},"subject":"Bump py37 to py38 in tox.ini","message":"Bump py37 to py38 in tox.ini\n\nIn wallaby cycle, we should test py38 by default.\n\nChange-Id: Ibd28376f97c7e3be897af5d60dfe0bcff4f08961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/b4e13f911b48ca89c85f9268cb843d9ab6254639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/b4e13f911b48ca89c85f9268cb843d9ab6254639"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
