)]}'
{"id":"openstack%2Fkarbor~766118","triplet_id":"openstack%2Fkarbor~master~Iadc4150b5bff6d24fd67e1b3e4307ac198988cd1","project":"openstack/karbor","branch":"master","hashtags":[],"change_id":"Iadc4150b5bff6d24fd67e1b3e4307ac198988cd1","subject":"fix log format error","status":"ABANDONED","created":"2020-12-09 02:42:58.000000000","updated":"2021-06-09 22:46:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77d596702e5de11dc5ca7cab6ba146b0488956f3","_number":766118,"virtual_id_number":766118,"owner":{"_account_id":32245,"name":"fudunwei","display_name":"devfu","email":"dunweifu1997@163.com","username":"fudunwei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-09 04:08:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 04:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88dfde5d0dafee3389ffafdbc559ad4f42b08901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32245,"name":"fudunwei","display_name":"devfu","email":"dunweifu1997@163.com","username":"fudunwei"},"date":"2020-12-09 02:42:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ab849d56b673f9553d3876dbb6058078b6c0e66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 04:08:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/617f9721533c4fb2a8d80c88ea2dfdfe : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/744a1d83a77442fab4eb8ea6864c48f0 : FAILURE in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533336d636844729b1e688c8916b35f9 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b970fdfa847e4b22b9f3b5164e3d061c : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b4f0f92b25744f89a7ceac150756e63 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aabdd7f11a3a4a99a9cd01c9664396a3 : SUCCESS in 4m 20s\n- karbor-fullstack https://zuul.opendev.org/t/openstack/build/23ec80c177d9418b929ae117b4910fe6 : FAILURE in 1h 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77d596702e5de11dc5ca7cab6ba146b0488956f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-09 22:46:24.000000000","message":"Abandoned\n\nkarbor is retired now http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f0eb2b19d80fd314dc6eb17744711e763df297b","revisions":{"3f0eb2b19d80fd314dc6eb17744711e763df297b":{"kind":"REWORK","_number":1,"created":"2020-12-09 02:42:58.000000000","uploader":{"_account_id":32245,"name":"fudunwei","display_name":"devfu","email":"dunweifu1997@163.com","username":"fudunwei"},"ref":"refs/changes/18/766118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/karbor","ref":"refs/changes/18/766118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/766118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/766118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/karbor refs/changes/18/766118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/karbor refs/changes/18/766118/1"}}},"commit":{"parents":[{"commit":"fcca3935af26a8b1b78eabfdd345840a3a2dd8c0","subject":"Merge \"Switch legacy Zuul jobs to native Zuul v3 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/fcca3935af26a8b1b78eabfdd345840a3a2dd8c0"}]}],"author":{"name":"fudunwei","email":"fudunwei@kylinos.cn","date":"2020-12-09 02:42:35.000000000","tz":480},"committer":{"name":"fudunwei","email":"fudunwei@kylinos.cn","date":"2020-12-09 02:42:35.000000000","tz":480},"subject":"fix log format error","message":"fix log format error\n\nChange-Id: Iadc4150b5bff6d24fd67e1b3e4307ac198988cd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/3f0eb2b19d80fd314dc6eb17744711e763df297b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/karbor/commit/3f0eb2b19d80fd314dc6eb17744711e763df297b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
