)]}'
{"id":"openstack%2Fkayobe-config-dev~1003282","triplet_id":"openstack%2Fkayobe-config-dev~stable%2F2025.2~Idb827125dd1dc26f0b9f1287a131cf32f5e8beec","project":"openstack/kayobe-config-dev","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-03 08:44:26.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-03 11:02:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb827125dd1dc26f0b9f1287a131cf32f5e8beec","subject":"Adds aio-network-connectivity check environment","status":"MERGED","created":"2026-09-01 14:03:04.000000000","updated":"2026-09-03 11:02:39.000000000","submitted":"2026-09-03 11:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003282","cherry_pick_of_change":986140,"cherry_pick_of_patch_set":2,"meta_rev_id":"2b5c4c6952a363e317b895863372734439d3dd2c","_number":1003282,"virtual_id_number":1003282,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 11:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-03 08:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-02 08:51:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-03 08:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 14:03:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-09-01 14:47:58.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-01 16:43:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 23:37:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"a77654a8d083010db2543572944f88636a6aaf2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-01 14:03:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c282bad055e33aa12681b0006e3e2929d2e4db6d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-01 14:47:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b828ca6de39e98f5db2eb5d963a395ec165836ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 16:43:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/959f325fa93b4e57834039dff35cf7b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c3716871bc4742bb5d614742c9c5be : SUCCESS in 2m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d1f005d712714518a0de3bc1a5de4413 : SUCCESS in 16m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74f2e246e1db485aa3056065ee5fc279 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0e905f808435490aa29fbbde9c946a03 : SUCCESS in 14m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81401f96ef2f4d198dd86839552b7692 : SUCCESS in 22m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2822c17d7ed94ac6b2bc2f7e7d3f3e43 : SUCCESS in 1h 04m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5f047bd3e5b341929680c617c4388de2 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4e0efba72a20437584a3c42091131511 : SUCCESS in 1h 04m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce3cd86aee5f4c99b5d01252208a55c4 : SUCCESS in 1h 45m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3595159c4512453fa4a8338827043bb9 : FAILURE in 1h 49m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e462a9af6e14209bc26b06ebf17801f : SUCCESS in 2h 37m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/290535616d264845beb58177578da453 : SUCCESS in 21m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5e50d808a00f4bd4b446c49d67cd00e7 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/559b4e9adf854c208e26e5113c4c9458 : SUCCESS in 24m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4dead5f271c64d6c99989eca2581b67e : SUCCESS in 13m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/734b6dfab679421984c3d0c548c33f82 : SUCCESS in 28m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe72be3f8fb54c95a448946e15e6e911 : SUCCESS in 23m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e1f2df1af204f3c9febd3687adf4ce0 : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8409922e5e87e1a647e0ae570dd51c17069b07c0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-01 16:46:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"431fdd7847c447879f04f1434567c614c9fc63bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 19:24:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40a0cd78a1a34afc9ab1c34e3db77270\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b4613da55047e9bd785ac6ae2daf23 : SUCCESS in 2m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/63da9b6caee24341ac31af93fec95c20 : SUCCESS in 10m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d89ef4778e87447dab6f554c515edc81 : SUCCESS in 12m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a689045ab12649a79a5caa2c661f9cc1 : SUCCESS in 24m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e84b595e440842ef8af46d138c4126a5 : SUCCESS in 15m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f04a75f789384a9e848b5050c8a5cca7 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ebcd7497c29143c4af7956a901523cb2 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5229b74e4d9a4562aaf769ebae47efbe : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f72cbe2ec52a409a9c0714b8c5e4006f : SUCCESS in 1h 34m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/120898fa17f9465e8d88b1929246998c : FAILURE in 1h 34m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60804eb761934029b2e47a9c4298928a : SUCCESS in 2h 31m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/272f1485d697426cb51b31fafccd240d : SUCCESS in 15m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/59f64821144b41a49f9652f5060ba884 : SUCCESS in 25m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3fd296fc8d84785b425d9c0c8273db1 : SUCCESS in 10m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd611bdc5bbc47e690dffbe3e302deb1 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/179839c79d4c4ca2a2c1ef22a8c7b455 : SUCCESS in 21m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/537363b359ea4cfcaa47190ac6647d01 : SUCCESS in 23m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b560edeb12ae47c28b7824d6083d9a1c : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"96e1d644021d36c248b0f4b246065d72883e1cfa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-01 23:37:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04e910666a83310df629be46a1c34637d839f8ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 02:12:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ab2d6007701408b9d1a670e153d6f99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a883c0564f1490ba3afac01958e2e4b : SUCCESS in 3m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f020e0285bcf408aa15b9bb1a48e36b6 : SUCCESS in 13m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/193274a4f3394a07af426016f9ed064e : SUCCESS in 7m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bd5e7d054eab40cfa609cd4995260569 : SUCCESS in 16m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39c0ebeb428644b3b06602d14259dbd8 : SUCCESS in 15m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e179922c26804f12941e71cfce25ed32 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b275807df0d34deea718f3433bb94bb7 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/22fff996bc8f4bda97f4d9be454fdffd : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/551c53c96c2d4bfdbea36c852dc4ef1f : SUCCESS in 1h 43m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1112c1b0d2fe4c25821773db32cf1efe : FAILURE in 1h 30m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26822b8f471e4a0c9e3633964f88e778 : SUCCESS in 2h 33m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ba9cc197f0fb4472a8ec64ccd00e4f46 : SUCCESS in 14m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6766854a78ae426c8ee3b5758874e70c : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/453134cf79ef438cb874edbcc9708220 : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d4b3908da3bf46318689f8733c5c7379 : SUCCESS in 16m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4d814211ae148b6a2fe2f4e56e5ffae : SUCCESS in 32m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1af991a84d7e4e87ac345a188e4012bf : SUCCESS in 21m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ffa6e0287594d838079ceed0c6f2681 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e5e8b77175bfd391fdefbd63a441dedaf2a1757e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-02 08:51:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f66361fde05be242066ad2b67bf5deb9d48105d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 09:27:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb64a64830e4bf0b95bd64b2a03f49a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80998e43c6fe458091d8faeeb9be0966 : SUCCESS in 2m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c67f0f03df34700898c8dbd2c5526cc : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcaf3c2d1821486dae8a3f2f0e1453d9 : SUCCESS in 6m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bf77c0f582604998b87a565690af7aa4 : SUCCESS in 23m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5782e63c4939428bb4cad39997202dd1 : SUCCESS in 29m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/925f81b5d8bd4ecc82fe13c84e556998 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f57833801be40bf8e4d8b9fcd80a9d4 : SUCCESS in 9m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/161df4ffbdb2417c847f04fa9ded15f4 : SUCCESS in 26m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a46332b4c7c442e38b734b37bf545a0c : SUCCESS in 19m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e35d47c6aa64513836a29171fca3c8f : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":2},{"id":"393479e092b28acfac8ebaf94f230f5a0215c27e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-03 08:44:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca5e1d5bcdde51af7122b1cad1261cf86dce11b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 08:45:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a8e93015fbdfcfdfe65689d190be8ff7f643788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:02:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/164fb3da0e384eb0a6398765446e99da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c609dbf88bac47e29e77e70af68a8e0f : SUCCESS in 2m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6c4333b0eb754ea1a3a51e400fc42aa0 : SUCCESS in 9m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acf504db9fbe4ae085e71a11c417fb01 : SUCCESS in 12m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5a2574e027ad4b5c89ef41db1eaa6436 : SUCCESS in 18m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8ceac2528ab44331be78e22733582f7f : SUCCESS in 10m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfad68f3f91443f5b05d958098168482 : SUCCESS in 20m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a147e6b3c33941b981fe08ec5c79db03 : SUCCESS in 9m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27bb87d6f48449e7a74a73e008962826 : SUCCESS in 18m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8901f6a597d64feab614d8d23b2913aa : SUCCESS in 10m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e738748602484295998dbcf48acd6666 : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":2},{"id":"2b5c4c6952a363e317b895863372734439d3dd2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:02:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eaf33a36130f95db1886b908933433550938e975","revisions":{"39270ef48a8c41b756f2a35aa970d9d59364d035":{"kind":"REWORK","_number":1,"created":"2026-09-01 14:03:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/1003282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/82/1003282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/1"}}},"commit":{"parents":[{"commit":"93379cf3f614a2967b2a656914f647823ef0b508","subject":"Use CentOS Stream 9 IPA deployment images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/93379cf3f614a2967b2a656914f647823ef0b508"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:39:04.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-01 14:03:04.000000000","tz":0},"subject":"Adds aio-network-connectivity check environment","message":"Adds aio-network-connectivity check environment\n\nAllows testing network-connectivity with an AIO.\n\nChange-Id: Idb827125dd1dc26f0b9f1287a131cf32f5e8beec\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit bbf45b2ba6db1436056176f4f2497babaaf1744f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/39270ef48a8c41b756f2a35aa970d9d59364d035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/39270ef48a8c41b756f2a35aa970d9d59364d035"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"5ab35589fcfc22671d22969c3fa2bdaa4e025776","ours":"93379cf3f614a2967b2a656914f647823ef0b508","theirs":"bbf45b2ba6db1436056176f4f2497babaaf1744f","merge_strategy":"recursive","contains_conflicts":false}},"eaf33a36130f95db1886b908933433550938e975":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-02 08:51:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/1003282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/82/1003282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/82/1003282/2"}}},"commit":{"parents":[{"commit":"d59e8b8c246e23e89e8db8f5eb67bcf05010c615","subject":"CI: Disable overcloud jobs temporarily","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d59e8b8c246e23e89e8db8f5eb67bcf05010c615"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:39:04.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-02 08:51:59.000000000","tz":0},"subject":"Adds aio-network-connectivity check environment","message":"Adds aio-network-connectivity check environment\n\nAllows testing network-connectivity with an AIO.\n\nChange-Id: Idb827125dd1dc26f0b9f1287a131cf32f5e8beec\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit bbf45b2ba6db1436056176f4f2497babaaf1744f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/eaf33a36130f95db1886b908933433550938e975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/eaf33a36130f95db1886b908933433550938e975"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase","conflicts":{"base":"93379cf3f614a2967b2a656914f647823ef0b508","ours":"39270ef48a8c41b756f2a35aa970d9d59364d035","theirs":"d59e8b8c246e23e89e8db8f5eb67bcf05010c615","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
