)]}'
{"id":"openstack%2Fkayobe-config-dev~1003925","triplet_id":"openstack%2Fkayobe-config-dev~master~I79ae3213561a5f7c1aaaf8823e1eb75b82306cba","project":"openstack/kayobe-config-dev","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-10-01 16:37:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I79ae3213561a5f7c1aaaf8823e1eb75b82306cba","subject":"ironic: Set pxe_filter sync_period to 5s","status":"MERGED","created":"2026-09-04 09:02:01.000000000","updated":"2026-10-01 16:37:31.000000000","submitted":"2026-10-01 16:37:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003925","meta_rev_id":"4c5285790f658005cd1a6c58c3539eafbff86299","_number":1003925,"virtual_id_number":1003925,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-01 16:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"all":[{"value":1,"date":"2026-10-01 15:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-10-01 13:34:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-10-01 14:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-10-01 13:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-10-01 14:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 11:05:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-01 13:10:45.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-10-01 13:34:32.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"REVIEWER"},{"updated":"2026-10-01 14:24:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-10-01 15:43:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"8f1bf8358fcfdf649ddfa8133c891bb95fd66d60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-04 09:02:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"638b5f3b822e536406d66d257ef9b671b6d5863d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:05:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2203296277349e69cad38eb277320fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8dbfae263664013a0d00bc19706abe0 : SUCCESS in 2m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e2be318db3714441ab623374c2d05b2e : SUCCESS in 18m 26s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/1da389564507457c84cb7774bc78772d : SUCCESS in 14m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4b12cc01bba4ef5a545dd6ed9503e9e : SUCCESS in 20m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c043d505d95d43ce8c7162a67fe98050 : SUCCESS in 18m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/30534e50c42946b181b0c83205219516 : SUCCESS in 22m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf48317c0bef4f4a8b07a3a2f4400f0d : SUCCESS in 22m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8e9cca05ec7544558191c2d531b80f3d : SUCCESS in 25m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68c286158c6e4e989c089a8b813dd731 : SUCCESS in 20m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c7f6b55878bc437c981473025cfcead1 : FAILURE in 1h 12m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b23a288adca742cdb4d881226dd6777f : FAILURE in 1h 16m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5a68a192af534caa9c11b37bf65c0a33 : FAILURE in 1h 29m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39ec7f55803d4fb59707dd5217178110 : FAILURE in 1h 15m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08b6a5ea6fd24871ae12fe6448401622 : TIMED_OUT in 2h 02m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/667802fa7ee14c95b0900d1f7a3339f9 : FAILURE in 1h 16m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83c12519aa8a4ca2a8a82efa70ccd3fa : FAILURE in 1h 21m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0620c50d76f9471d89072efda252a59e : SUCCESS in 17m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b1c5cfad00743a5aa11fc9b3d8f5ea4 : SUCCESS in 30m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e2a8c99de8f4203bca9ffa6835d029d : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e073a4396104bbd97dad245dd381c5c : SUCCESS in 24m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7f2e593262c94d0bb7532ec7fb373590 : SUCCESS in 41m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d867b3ec80964419a52c7f331dd33ec4 : SUCCESS in 40m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1774ebca11cb4856954ff7aa7251a63b : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2185080acbfa4e93b48d59a1f460fed4 : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"50ad0f3765d2d876666fd7ca7fc4fcda9761242f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-10-01 10:34:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ac897f5d340c43a9e3fa8a0bdfcfeb0eb5e73d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:58:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3641cbf5f064ecbb3d0a194311488ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bdeb23ddc94e3c9416c6c744bf0cbb : SUCCESS in 2m 40s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/23a1f14ac0974077b2f802d6f35a0dca : SUCCESS in 29m 29s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/fd685ed3c5874a91bb9358ec2dd384c3 : SUCCESS in 32m 55s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7b68d26d0584ca39a65cf2479ec5d31 : SUCCESS in 34m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d8516c33a364d1591cfa5cf8e944ec7 : SUCCESS in 34m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5a1698b14e4d4e138708396c50be7436 : SUCCESS in 35m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0138871996246bebf8ed095369379f8 : SUCCESS in 37m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8fa12d75691446ca9ef84b5899e75fcd : SUCCESS in 32m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cf2a756954d4735b68d9174f587132e : SUCCESS in 44m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7146f0acb6524de98a354eb46338a670 : SUCCESS in 1h 35m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4337fd7633ce4175bf60f08c9334b6a3 : SUCCESS in 2h 13m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/93cccdb9deb14362a5ec2b569c83a2bb : SUCCESS in 1h 54m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bdd7a8fc1eb45ac8dda7fff8cdb55a9 : SUCCESS in 1h 21m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8fc39bf4df15418cba989e590b77402c : SUCCESS in 1h 41m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b04e4cee5004f989fa93bcab94ca708 : SUCCESS in 2h 09m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/437302168bd547d68cd3697ce50f7831 : SUCCESS in 2h 23m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/63e2d15d048a42d481e3481a7de98a4b : SUCCESS in 39m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ae8b3a1dcd8e48a0b7ace52a1e9426ce : SUCCESS in 36m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16b4f108a6134d7eb682abbc4efe8010 : SUCCESS in 30m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/834528c59a844e57ab4ab922eaa07da6 : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2b5725e301134687bb2328382e207e1f : SUCCESS in 45m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c23c4466ba004c58ace069084feacd28 : SUCCESS in 56m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f453a23a3fe549f4a11c6d45f479e8a9 : SUCCESS in 37m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a905a7c35dc247dc953c12f90bdd8911 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":1},{"id":"358e406f961324616287c84b0d89d361bc63617c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-10-01 13:10:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59d26b8c8108414828c0f8d07e5064dd57b42b64","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-10-01 13:34:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb2267faf563bb1748568e5471dc6ae8b73cdf04","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-10-01 14:24:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69f249458902dfa38a8814ff96b68cdb2b562313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 14:25:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6b71064b12a9c8953bd34646864326805f5191d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-10-01 15:43:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1ea0b60223ce31460d25fe9ac5fa83dc8f74e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 16:37:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3780dadc11ca4e788ef5e6d0eda92081\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ceb7a39cb74dbc83373851b36807e2 : SUCCESS in 2m 18s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9efc125675394cb980fa67aca4a65013 : SUCCESS in 13m 44s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d94e5a1eefb415d9346676cd07dee28 : SUCCESS in 18m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8808fb157b934560a85e31a213ae56af : SUCCESS in 23m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3fe005970074d6d9b9845c46d28e2f7 : SUCCESS in 21m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e4883c55bf7d4542b01e942cdd0d3b3e : SUCCESS in 16m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b0b2f9b48064c6398d620eda1e2ea95 : SUCCESS in 27m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dbf18ff697934dc1a33bc96ae8d60790 : SUCCESS in 59m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d639d2f6ae924d0b8688e9be1e354fbc : SUCCESS in 59m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f05a6d4404e94e689205f6146aaeb263 : SUCCESS in 1h 10m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eb8a9e744574a039d29fab57de8b7f3 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4ac4df11832431993993e601d71dfa7 : SUCCESS in 1h 08m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2d58877594224b5daae747f9b357f785 : SUCCESS in 1h 53m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/524303490b6040b1b0b43a6c4c4502c5 : SUCCESS in 2h 11m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/313bb3a12d8a4f89af389e71f800f3f9 : SUCCESS in 28m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5fdd8408613641beb9ac207c3724fab3 : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e2bbe09fb354e85b23122fcde47ad7b : SUCCESS in 24m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c9dcdfb683d47a1b4753aab47650f51 : SUCCESS in 23m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0821399889e64369aacaa12499d6f530 : SUCCESS in 39m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e5dc2a71d9d41eeacf00159431abb33 : SUCCESS in 35m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/93903caef72c4e2f933ebe40f568240a : SUCCESS in 17m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d2526ea4734225843115a0caf3ee99 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4c5285790f658005cd1a6c58c3539eafbff86299","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 16:37:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d522cacec497e0d8b5abed491d3741bb88894c26","revisions":{"d522cacec497e0d8b5abed491d3741bb88894c26":{"kind":"REWORK","_number":1,"created":"2026-09-04 09:02:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/1003925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/25/1003925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/25/1003925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/25/1003925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/25/1003925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/25/1003925/1"}}},"commit":{"parents":[{"commit":"db075691bccb7d496c1ebc00f81a884d7b0e9819","subject":"CI: Synchronise jobs with kayobe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/db075691bccb7d496c1ebc00f81a884d7b0e9819"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-04 08:59:56.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-04 09:01:50.000000000","tz":60},"subject":"ironic: Set pxe_filter sync_period to 5s","message":"ironic: Set pxe_filter sync_period to 5s\n\nIn virtualized CI environments, VMs boot very quickly during\ninspection, which can cause them to send DHCP requests before\nironic-pxe-filter finishes its periodic sync (default 45s) to unblock\nthe node MAC in ironic_dnsmasq.\n\nReduce sync_period to 5s in CI to avoid this race condition.\n\nChange-Id: I79ae3213561a5f7c1aaaf8823e1eb75b82306cba\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d522cacec497e0d8b5abed491d3741bb88894c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d522cacec497e0d8b5abed491d3741bb88894c26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
