)]}'
{"id":"openstack%2Fkayobe-config-dev~767436","triplet_id":"openstack%2Fkayobe-config-dev~master~If2e351ebe638e4591197a071aecc44e98d3d865d","project":"openstack/kayobe-config-dev","branch":"master","topic":"retire-karbor","hashtags":[],"change_id":"If2e351ebe638e4591197a071aecc44e98d3d865d","subject":"Remove Retired Karbor Support","status":"MERGED","created":"2020-12-16 23:20:55.000000000","updated":"2020-12-22 16:14:23.000000000","submitted":"2020-12-22 16:14:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767436-retire-karbor","meta_rev_id":"1ffd4ae65fe232a3705a9902720d8f74e0b014ef","_number":767436,"virtual_id_number":767436,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-22 16:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-17 10:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-22 14:31:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 00:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 10:50:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"4daa92337ce5b1f965c80e77a973e43fff81b701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-16 23:20:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d8044f3b9d8e9f22ed3119d9cf284cec87ad294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 00:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/219fbcf84c4d4ecea08554abfd532097 : SUCCESS in 2m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a585526242a64bde9b7a66cbc174d04a : FAILURE in 50m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8cbc8b16d52d4663ab47a79f7225beb2 : SUCCESS in 14m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/93f806c8dcde4dc79003f025219226fe : FAILURE in 46m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a7cfcf692ac947b686abae96c6ce9fd0 : FAILURE in 4m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68c37be2c0b7460b990985bc077408ff : FAILURE in 25m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3481053c9e3c45409c3e52dd73871cb7 : FAILURE in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"833b1984671ef533fcd08d0521a42fbbe26953fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 10:50:09.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks. CI currently broken...","accounts_in_message":[],"_revision_number":1},{"id":"65360a0d6567d97f3fc7f2cf81862bb380533087","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:19:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd3de9551029d79dd3da640c19b8e80f6a7c8dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:03:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75161fbe3c649cbb48bd0fc23beeb27 : SUCCESS in 3m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5685fc9ab6dc4f9a9cfd7f904ca5227f : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a1497441df3547159e03550ba5c0964f : SUCCESS in 16m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b4a21f5cc59d4f40b4987acb64a54c99 : SUCCESS in 1h 42m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6e61e5c7661f4d7f8f306bb86fad29f5 : SUCCESS in 15m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/de74103ce3d146ecb1a1394ce49b232a : SUCCESS in 23m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e1837d6c7d1e4600b595fe83dd20d1f7 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8c8d354309b9363347706922ee122c5bd724b852","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 14:31:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdb6481ca06d88445fd8c401a77b25a83e9c7ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:31:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fff9563867fa54ebf5cd455bec4f744597402f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 16:14:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518e8b010f3a41e0af722dd294e38fa6 : SUCCESS in 2m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/632e65bdec35492292f5b884325ebf8c : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3755001980a24ee38420bf8373b7057f : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/82d464cac22a47b398970cb13b40e634 : SUCCESS in 1h 37m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9f7035c185ba4786a3baded63e91cf5d : SUCCESS in 15m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e01420bd097f41a999185e52faa9f636 : SUCCESS in 23m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0c062df42d3a42c49a60307072a0f154 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1ffd4ae65fe232a3705a9902720d8f74e0b014ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 16:14:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef7c5fdf5a3b867e6a0205b28866a6f896534fbe","revisions":{"ef7c5fdf5a3b867e6a0205b28866a6f896534fbe":{"kind":"REWORK","_number":1,"created":"2020-12-16 23:20:55.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/36/767436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/36/767436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/36/767436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/36/767436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/36/767436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/36/767436/1"}}},"commit":{"parents":[{"commit":"21f972fb85140f683c94bc0d7de033498c305c26","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/21f972fb85140f683c94bc0d7de033498c305c26"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-16 23:20:07.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-16 23:20:07.000000000","tz":-480},"subject":"Remove Retired Karbor Support","message":"Remove Retired Karbor Support\n\nAs announced on the openstack-discuss ML[1], Karbor is retiring\nthis cycle (Wallaby).\n\nNeeded-By: https://review.opendev.org/c/openstack/karbor/+/767032\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018643.html\n\nChange-Id: If2e351ebe638e4591197a071aecc44e98d3d865d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/ef7c5fdf5a3b867e6a0205b28866a6f896534fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/ef7c5fdf5a3b867e6a0205b28866a6f896534fbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
