)]}'
{"id":"openstack%2Fkayobe-config-dev~792673","triplet_id":"openstack%2Fkayobe-config-dev~master~Ib82da83681f153f8ddccb311123abadc42c2eb3b","project":"openstack/kayobe-config-dev","branch":"master","topic":"sync","hashtags":[],"change_id":"Ib82da83681f153f8ddccb311123abadc42c2eb3b","subject":"Sync kayobe-config-dev with latest kayobe changes","status":"MERGED","created":"2021-05-21 16:26:35.000000000","updated":"2021-05-25 13:28:13.000000000","submitted":"2021-05-25 13:07:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792673-sync","meta_rev_id":"1c98982222fbece3191948749d54d6845c17c96e","_number":792673,"virtual_id_number":792673,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-25 13:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-25 12:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-05-24 09:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-25 12:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 17:12:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 09:12:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-25 12:20:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9bc5fdbc4751928f7ca745eb923d0b13ecfd027f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-21 16:26:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65285e9296bc8fa25cb6e7329d2d8b4e0c240b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 17:12:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0963d060542f4b3c958d727a9c801316 : SUCCESS in 3m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f8290bba0d84b5ca4b09f8c8b68be1d : SUCCESS in 43m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b80323999afa4accad4ffa3ba9dddee3 : FAILURE in 13m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27a4cded3f79490bb8bf7e7b7c79c179 : SUCCESS in 24m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c6f9eb806b74cd792daf3fbce698cb8 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8788ee7844ac98b28bef14dbc26facfb45ddbc0a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-21 19:41:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ea4b60e1b19b90fc38bc706101f3a5d3001a8af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 20:27:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abb926bb9ae94e509813f8a344e19ec9 : SUCCESS in 2m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1abbbe5e0c94e0ebab2d1a23434a68b : SUCCESS in 42m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4e37d6f0c74fbd9dd8d180e9e51dbd : FAILURE in 11m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac0333d78619446fb17883eb4393ec47 : SUCCESS in 28m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f4f93f5789149748e9ebc1d1e7b97cf : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fd50d3574144049395d5559bfcd029cb6e3db10a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-24 09:12:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8dbc67cdc414e13290fd59fad76a12c18ee7fc73","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-25 10:36:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"20c8e2a04e549fcbbc4c19aff222a587266d32f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 11:22:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ac1ee5200c414a8bc393b7aedb9522 : SUCCESS in 3m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7e8afa9b1f346ce9a168721bbb2217b : SUCCESS in 43m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a13ea95ae548dcb7e9e2216f95a883 : SUCCESS in 21m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c37b5e2082c4676ab9fc508472fed12 : SUCCESS in 22m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bea3c14bb585448489ade0d48d3d3520 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1f90aa10cccaed3035a76faac48ea30961057236","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-25 12:20:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f72057c90e0fd4651e42b5780dcecf4d70a90055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 12:21:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40f2a9c4457ec54f4b0ef525d03fbc47b96a8109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 13:07:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de468ef5ed34fb281b0c325c0d99109 : SUCCESS in 3m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f191d8601aa447288d265e9c2aaeb5e : SUCCESS in 45m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84bb19e4f553413e9e6f0700ee856d5f : SUCCESS in 13m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16e80caefee441f5ab96fe91979812ea : SUCCESS in 24m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d97da4da5747d3aaabcf88cbdef94c : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e95bac2d42740788aad40f659c729ec41f379d62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 13:07:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"158b5c3f79a6317b63ca02e3408cf72682015a4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 13:22:05.000000000","message":"Patch Set 1:\n\nI wonder why only the ubuntu jobs ran","accounts_in_message":[],"_revision_number":1},{"id":"180ef87129c60b176aa73adb16b45ae5bef62034","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 13:25:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I wonder why only the ubuntu jobs ran\n\nAh... stream. https://review.opendev.org/c/openstack/kayobe-config-dev/+/792974","accounts_in_message":[],"_revision_number":1},{"id":"17684f07645dcc2cc86579c00afd7e9194aff5cf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-25 13:25:36.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I wonder why only the ubuntu jobs ran\n\nBecause the centos8 ones have been renamed to centos8s.","accounts_in_message":[],"_revision_number":1},{"id":"1c98982222fbece3191948749d54d6845c17c96e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-25 13:28:13.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e I wonder why only the ubuntu jobs ran\n\u003e \n\u003e Because the centos8 ones have been renamed to centos8s.\n\nhttps://review.opendev.org/c/openstack/kayobe-config-dev/+/792975","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9210ff39c1e446cee188348fd7a9522e9bf8b83","revisions":{"d9210ff39c1e446cee188348fd7a9522e9bf8b83":{"kind":"REWORK","_number":1,"created":"2021-05-21 16:26:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/792673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/73/792673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/73/792673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/73/792673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/73/792673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/73/792673/1"}}},"commit":{"parents":[{"commit":"8d8e9bdba36fd90946bad47fdf628058be0ac68d","subject":"Merge \"Set os_distribution automatically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/8d8e9bdba36fd90946bad47fdf628058be0ac68d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-21 16:26:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-21 16:26:06.000000000","tz":120},"subject":"Sync kayobe-config-dev with latest kayobe changes","message":"Sync kayobe-config-dev with latest kayobe changes\n\nChange-Id: Ib82da83681f153f8ddccb311123abadc42c2eb3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d9210ff39c1e446cee188348fd7a9522e9bf8b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d9210ff39c1e446cee188348fd7a9522e9bf8b83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
