)]}'
{"id":"openstack%2Fkayobe-config-dev~795983","triplet_id":"openstack%2Fkayobe-config-dev~master~Ie3eec95ac7dadbfc613e913c09371d61b7e5eb2a","project":"openstack/kayobe-config-dev","branch":"master","hashtags":[],"change_id":"Ie3eec95ac7dadbfc613e913c09371d61b7e5eb2a","subject":"Enable deploying monitoring with default inventory","status":"MERGED","created":"2021-06-11 10:43:13.000000000","updated":"2021-06-11 20:37:04.000000000","submitted":"2021-06-11 20:37:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795983","meta_rev_id":"0e225bc8f3b40917aad68ddfcab07d1ea408c244","_number":795983,"virtual_id_number":795983,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 20:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-06-11 19:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-11 13:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-06-11 19:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-06-11 10:43:50.000000000","_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 10:43:50.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-06-11 11:36:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 13:25:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-11 19:34:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6ea9e43ff5c6fba30f25be280f4c54264479000a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-06-11 10:43:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f18af437972c17048128cdfd8853d644e8b8ace3","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-06-11 10:43:50.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"75c0eba989ef9741586fd2deffe7bc11a9e05465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 11:36:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18efffe4a3b144b29f10159ddeeebe44 : SUCCESS in 3m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c9462501be14a3189f05268d56e9b3f : SUCCESS in 52m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8312d2b709894e73b14a12810419a459 : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47e43f7baf0540c9a923d552820dabfd : SUCCESS in 28m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ed7c028ecef4e8d8f1eb43442f4e4d7 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"831599453e3209ec03d78be390768c7680d7f570","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-11 13:25:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db25875964d9869bfea639e9979ee8a64f5a3e6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-11 19:34:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"327bad4739a76ad11e93c09f7724bbf68879fa24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 19:35:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2673f4d9282f219e5d8e96c389f1681289d00fab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:37:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af4f921c3ed44eabf25f600570ac691 : SUCCESS in 3m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d664e5757595438db85a002cc76dc371 : SUCCESS in 55m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4506a4799714584a169ecc18e8c2868 : SUCCESS in 14m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53881d37ef5c404a8fae28384e6b2fff : SUCCESS in 27m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be09a76efca44e9eb5293a5644b61390 : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0e225bc8f3b40917aad68ddfcab07d1ea408c244","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:37:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a286890b83f61c7418c791231933fea2bdc045a","revisions":{"5a286890b83f61c7418c791231933fea2bdc045a":{"kind":"REWORK","_number":1,"created":"2021-06-11 10:43:13.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/83/795983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/83/795983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/83/795983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/83/795983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/83/795983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/83/795983/1"}}},"commit":{"parents":[{"commit":"d9210ff39c1e446cee188348fd7a9522e9bf8b83","subject":"Sync kayobe-config-dev with latest kayobe changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d9210ff39c1e446cee188348fd7a9522e9bf8b83"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-06-11 10:40:27.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-06-11 10:40:27.000000000","tz":120},"subject":"Enable deploying monitoring with default inventory","message":"Enable deploying monitoring with default inventory\n\nThis makes enabling monitoring possible without any modification\nto the default inventory.\n\nChange-Id: Ie3eec95ac7dadbfc613e913c09371d61b7e5eb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/5a286890b83f61c7418c791231933fea2bdc045a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/5a286890b83f61c7418c791231933fea2bdc045a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
