)]}'
{"id":"openstack%2Fkayobe-config-dev~815272","triplet_id":"openstack%2Fkayobe-config-dev~stable%2Fxena~I9a7cd0c4f99d14ec51b2cbc18d833a9d68f695bb","project":"openstack/kayobe-config-dev","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I9a7cd0c4f99d14ec51b2cbc18d833a9d68f695bb","subject":"Update .gitreview for stable/xena","status":"MERGED","created":"2021-10-25 10:02:43.000000000","updated":"2021-10-26 10:50:37.000000000","submitted":"2021-10-26 10:50:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815272-create-xena","meta_rev_id":"a1ad1e00a749dda88eb6b6a9528d5a4418018185","_number":815272,"virtual_id_number":815272,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 10:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-25 10:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-26 08:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-10-26 08:48:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 10:35:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-25 12:07:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 08:48:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"7eaad6ffdc2d9265c9b80861d88dba5fae51e824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-10-25 10:02:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"836bb8c03ab6f985ddf13040acfc7cde123a708d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-25 10:35:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb6dde77f65a6e820b0a3d2d46587a1f9f9e32b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:07:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbffa890eeb4e6690832f2650ec6a3a : SUCCESS in 3m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c8b9a0da8914c3dbf69015bfbb21f52 : SUCCESS in 58m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e9b3b5253a424f93e8703b6a52a359 : SUCCESS in 41m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ac7cb32df5c54e5f884b759e946b9d33 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e9378fba1ec942c3b7cdb22467c76214 : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edfbd3c5d6cf446c9d051e6d7369b1b7 : SUCCESS in 15m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9a2d4a1bb134f129f3b142bd78fa8af : SUCCESS in 1h 58m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d3110de4e0704e6ab429721e3cc723ea : SUCCESS in 33m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/debd131a6e6c476393f0404216d91bd4 : SUCCESS in 32m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e537f85e8603406fbd686422f8c0d42f : SUCCESS in 29m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/65e0f8d7e6f04cc6a39130a2bd2d6128 : SUCCESS in 21m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d814d7ad517414c935e78631257e034 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5e71b0fb4ec1f1dbbf5a231ebd30fb8b445bcdee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-26 08:48:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7051ce185d965a24471be578416ac5beef28116b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 08:49:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d62b116c50f556c368f3c1abe1cdb3cd8a735842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 10:50:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ea0b9dec114a9ab7321c42f5371a9d : SUCCESS in 4m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8f249906c3f048428bb443e37bcf18d3 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf2625e67e44e01a7f6ff66364b0a8c : SUCCESS in 50m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d032bf6e7de342be9fd11cb2425f6470 : SUCCESS in 1h 15m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c910a79eb800411687255eb56ca386d9 : SUCCESS in 33m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29a65a07eb5f4c14871b05ee6c749bbf : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf5a5b6071964df3b6708ea363291fdb : SUCCESS in 1h 46m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6e6202af5522466bb30c111f23129c23 : SUCCESS in 36m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f6d2c5928334a06b9f15852e744b22d : SUCCESS in 29m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca5ca9b23d8a4710bd244b77d53ed1a1 : SUCCESS in 32m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1d85716729704dc9acf82121eb3493e7 : SUCCESS in 24m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd36b98f6c294ecd91bb7c8d80f8dd93 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a1ad1e00a749dda88eb6b6a9528d5a4418018185","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 10:50:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7bb262c95a8701749f45613cc63115aa75931ea","revisions":{"a7bb262c95a8701749f45613cc63115aa75931ea":{"kind":"REWORK","_number":1,"created":"2021-10-25 10:02:43.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/72/815272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/72/815272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/72/815272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/72/815272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/72/815272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/72/815272/1"}}},"commit":{"parents":[{"commit":"14ba88cd7c10b6aaf45863e4ddf13004f4b9db04","subject":"Sync kayobe-config-dev with kayobe changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/14ba88cd7c10b6aaf45863e4ddf13004f4b9db04"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-10-25 10:02:42.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-10-25 10:02:42.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I9a7cd0c4f99d14ec51b2cbc18d833a9d68f695bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/a7bb262c95a8701749f45613cc63115aa75931ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/a7bb262c95a8701749f45613cc63115aa75931ea"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
