)]}'
{"id":"openstack%2Fkayobe-config-dev~824424","triplet_id":"openstack%2Fkayobe-config-dev~stable%2Fussuri~I5dbaba68e24c1dec8fa37ca312ae36cdb2d63c5a","project":"openstack/kayobe-config-dev","branch":"stable/ussuri","topic":"ci-stop-testing-non-stream-centos-stable/ussuri","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-28 23:52:19.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-13 10:12:12.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I5dbaba68e24c1dec8fa37ca312ae36cdb2d63c5a","subject":"[CI] Move testing to CentOS Stream","status":"MERGED","created":"2022-01-13 09:01:15.000000000","updated":"2022-01-28 23:52:19.000000000","submitted":"2022-01-28 23:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824424-ci-stop-testing-non-stream-centos-stable/ussuri","cherry_pick_of_change":823268,"cherry_pick_of_patch_set":1,"meta_rev_id":"c9910c76d9444bf45ca8f2fad2b46f95262e0878","_number":824424,"virtual_id_number":824424,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 23:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-13 09:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-01-13 10:12:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-28 22:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-28 22:00:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 09:01:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-01-13 09:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 09:13:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-28 22:00:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"2f092a692c7f914cbad93d6fab62e1c3c6f05c2d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-13 09:01:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"9050bfbea8552905e23a2ec7cd9474e1100527a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:02:07.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe-config-dev on branch stable/ussuri.  The error was:\n\n  while scanning a simple key\n  in \"openstack/kayobe-config-dev/zuul.d/project.yaml@stable/ussuri\", line 6, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/kayobe-config-dev/zuul.d/project.yaml@stable/ussuri\", line 13, column 1","accounts_in_message":[],"_revision_number":1},{"id":"5a26da5769ad27f3280bb08407fc2d0aa72728cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-13 09:02:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c3546d59244894048f09120bd17c2711e41f7f4c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-13 09:02:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"accb7d20b7e4790d9c890ee81511327a64754cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-13 09:02:49.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e9f3fde578dec9a0490b876154aeb9a0aae6f729","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-13 09:13:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8b3702dd2c2ec5755895c6b11eaea2ef0839c97e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-13 10:12:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3a21ce2c09df10b569aae91630ab9b23af499663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 15:03:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c25e584d1614794bda5653cdc9092f3 : SUCCESS in 3m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/60a991ec5b7e49438e37d80733937ae1 : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/22dd9df918144e78a71b7dbf0670acce : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/14347dcaeb1a41ff92a545d93e2a8f4f : POST_FAILURE in 2h 55m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/228c50aa220e43389e8187e97f44077c : SUCCESS in 29m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/066e135b976548a3b3965fb329018cb2 : FAILURE in 18m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bee70387dd0d42d9a715092510efe74b : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5dbc0859e09678fb6fda6ceb6681da96ef3b2900","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-18 09:34:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c04e0b7ab034387c52aad4bca96102fd37e91c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 14:50:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1fb5df9cb842ee849571683c3c7138 : SUCCESS in 4m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ac17f8d8a3ab4a8a88ee840528563337 : FAILURE in 26m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eee6e2ddc4cb4cad8094c697503bba2a : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65369ce4f36744529204f32b7c30404c : FAILURE in 2h 08m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/89427c8d6dee48feb5b3b5bc43d3274c : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ad41dc13c0b64ef7939bdc12f8de9966 : SUCCESS in 30m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2b0c7da6be949eeb53d9619224dcb49 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":3},{"id":"4fe47286de157d17f6510ef95271c6579fa9aee8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-28 14:49:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"856107ff5a4a88fffae1ab0691a57959264d6257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 17:06:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e608a1134204da2a81398ad782228fd : SUCCESS in 3m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5647163348ed4202802e614547cc1740 : SUCCESS in 1h 09m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f8626e9aa9f64cad938794e8a2adcc31 : SUCCESS in 15m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b6c879d3e18406d9ca8f860a6c15feb : SUCCESS in 2h 14m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b536ae31d9314ab4a0d066189f2880bf : SUCCESS in 31m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df173f0729b24279a35ab9e955f7af4e : SUCCESS in 33m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2a848b8af31440d2a2e6c177375e03a7 : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3414f8a82609b5632376e04ad90898bc6399ff9b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-28 22:00:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aa7eea22645c61194d6aa915d7bebefb0bdb76fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 22:02:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"69b695d290671c0d83723673734e8384a33f0e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 23:52:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd4075d713c463e9b6c4a94e934eb81 : SUCCESS in 3m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef845f7c6b86418bbd8fc55d7f7a8527 : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/98dcb60b546d426d9025c1c30a9e373d : SUCCESS in 17m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/830cafb5831645299223e6c9935d8228 : SUCCESS in 1h 48m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f61d0222c720486cb10c3477e96ee9a7 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f274368d240948f3aa457c3d92646f25 : SUCCESS in 29m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a06b7ab8aec046cb966c387524b146f7 : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c9910c76d9444bf45ca8f2fad2b46f95262e0878","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 23:52:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f478265c027b6795fee562b0836c081fe7ec47d3","revisions":{"d225905f7aa457fd169695310e5293a9660b1d34":{"kind":"REWORK","_number":1,"created":"2022-01-13 09:01:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/824424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/24/824424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/1"}}},"commit":{"parents":[{"commit":"cb096de2edc8a4eaab6b9fb1061a6e218ef238e0","subject":"Sync configs with kayobe @ cf8869eecd669df411696e78bd90d0b9a02f429f","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/cb096de2edc8a4eaab6b9fb1061a6e218ef238e0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 11:22:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-13 09:01:15.000000000","tz":0},"subject":"[CI] Move testing to CentOS Stream","message":"[CI] Move testing to CentOS Stream\n\nRequired by the relevant Kayobe change.\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nNeeded-By: https://review.opendev.org/c/openstack/kayobe/+/823266\nChange-Id: I5dbaba68e24c1dec8fa37ca312ae36cdb2d63c5a\n(cherry picked from commit 18cd3ed7a137195f9d1d35c1b13459e101854f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d225905f7aa457fd169695310e5293a9660b1d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d225905f7aa457fd169695310e5293a9660b1d34"}]},"branch":"refs/heads/stable/ussuri"},"3f635bdde6799e35487e01889810f67ea81cd166":{"kind":"REWORK","_number":2,"created":"2022-01-13 09:02:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/824424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/24/824424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/2"}}},"commit":{"parents":[{"commit":"cb096de2edc8a4eaab6b9fb1061a6e218ef238e0","subject":"Sync configs with kayobe @ cf8869eecd669df411696e78bd90d0b9a02f429f","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/cb096de2edc8a4eaab6b9fb1061a6e218ef238e0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 11:22:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-13 09:02:21.000000000","tz":0},"subject":"[CI] Move testing to CentOS Stream","message":"[CI] Move testing to CentOS Stream\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/824309\nChange-Id: I5dbaba68e24c1dec8fa37ca312ae36cdb2d63c5a\n(cherry picked from commit 18cd3ed7a137195f9d1d35c1b13459e101854f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/3f635bdde6799e35487e01889810f67ea81cd166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/3f635bdde6799e35487e01889810f67ea81cd166"}]},"branch":"refs/heads/stable/ussuri"},"f478265c027b6795fee562b0836c081fe7ec47d3":{"kind":"REWORK","_number":3,"created":"2022-01-13 09:02:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/824424/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/24/824424/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/24/824424/3"}}},"commit":{"parents":[{"commit":"cb096de2edc8a4eaab6b9fb1061a6e218ef238e0","subject":"Sync configs with kayobe @ cf8869eecd669df411696e78bd90d0b9a02f429f","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/cb096de2edc8a4eaab6b9fb1061a6e218ef238e0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 11:22:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-13 09:02:49.000000000","tz":0},"subject":"[CI] Move testing to CentOS Stream","message":"[CI] Move testing to CentOS Stream\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/824309\nChange-Id: I5dbaba68e24c1dec8fa37ca312ae36cdb2d63c5a\n(cherry picked from commit 18cd3ed7a137195f9d1d35c1b13459e101854f8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/f478265c027b6795fee562b0836c081fe7ec47d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/f478265c027b6795fee562b0836c081fe7ec47d3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
