)]}'
{"id":"openstack%2Fkayobe-config-dev~881716","triplet_id":"openstack%2Fkayobe-config-dev~master~I2a31e374a6d59c365e1132882236d6abb0c5b62b","project":"openstack/kayobe-config-dev","branch":"master","topic":"test-kayobe-seed-images","hashtags":[],"change_id":"I2a31e374a6d59c365e1132882236d6abb0c5b62b","subject":"[DNM] test kayobe-seed-images","status":"ABANDONED","created":"2023-04-27 11:31:54.000000000","updated":"2023-04-27 23:08:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a26070aed32586b8e933e054ef3e8ef0a0152187","_number":881716,"virtual_id_number":881716,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-27 12:38:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 12:38:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc5af35381444614686a83ceaba5ccfdcc47015f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 11:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"441999d386a1618579e4bdf599b4cf6c08623e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:38:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88f10e16a414de39ad40c035d7047df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ec95d3865448f1a041fb944943ec2e : SUCCESS in 2m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/87efa8ab0a75420da9883a209decdec0 : SUCCESS in 45m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a11ab5f931a242e6be1b7313a58fa218 : SUCCESS in 1h 05m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae25be6783f041b9b7fb1ef827c4824a : SUCCESS in 48m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fed14da2f618432392330fe055531b0a : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8af067f386ef450abbde2e89b2bc8400 : SUCCESS in 15m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/210f3e7e20f54686a824655428447587 : SUCCESS in 13m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab3e86da64304180b7f8f3e5437df01f : SUCCESS in 14m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/73ff6f563f3b4b7488118d35263bd38a : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ce0072b07ce4205805f9e5af8b50587 : SUCCESS in 13m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f53bd06da71c4c3097f49fabc767f11c : FAILURE in 30m 46s (non-voting)\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77154eb2f6b44248b5388e275c068072 : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd4acd1f1bc54a588486e31f07af1922 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c6a912a0f2147608576b08a39d5adc8 : SUCCESS in 16m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/759faee3c76e49c484708fd864012977 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a26070aed32586b8e933e054ef3e8ef0a0152187","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 23:08:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b91e5d95cd86f03a0609569f5226314023ce294","revisions":{"4b91e5d95cd86f03a0609569f5226314023ce294":{"kind":"REWORK","_number":1,"created":"2023-04-27 11:31:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/16/881716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/16/881716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/16/881716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/16/881716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/16/881716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/16/881716/1"}}},"commit":{"parents":[{"commit":"12cbf39eed5df35a2f303f6589c8e04a78e32ebb","subject":"Use custom cpu_models on Redhat family only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/12cbf39eed5df35a2f303f6589c8e04a78e32ebb"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 11:25:19.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 11:30:26.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images","message":"[DNM] test kayobe-seed-images\n\nChange-Id: I2a31e374a6d59c365e1132882236d6abb0c5b62b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/4b91e5d95cd86f03a0609569f5226314023ce294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/4b91e5d95cd86f03a0609569f5226314023ce294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
