)]}'
{"id":"openstack%2Fkayobe-config-dev~907940","triplet_id":"openstack%2Fkayobe-config-dev~stable%2F2023.1~Ib61a62e126c66d4133d8e80dff5643ebc6966bd1","project":"openstack/kayobe-config-dev","branch":"stable/2023.1","topic":"aio-network-manager-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-07 15:37:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib61a62e126c66d4133d8e80dff5643ebc6966bd1","subject":"Use dummy1 as bridge port instead of eth1","status":"MERGED","created":"2024-02-07 07:03:08.000000000","updated":"2024-03-07 15:37:52.000000000","submitted":"2024-03-07 15:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907940-aio-network-manager-stable/2023.1","cherry_pick_of_change":904044,"cherry_pick_of_patch_set":2,"meta_rev_id":"1f63aff470c613ad5094365a9a3c98d8837b3b3e","_number":907940,"virtual_id_number":907940,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 15:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-04 12:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-03-07 13:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-07 13:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 08:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-04 12:08:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-07 13:40:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"700f7899c2d45b7ca4bad9b341beca3c2fd2091f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-07 07:03:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9d46bd4b475bb9ab1e18d2e817afc82eab7ababc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:12:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3458bf7e604f01953ed349ec2b6b1d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e3160253e064a42aa1f02557ae97887 : SUCCESS in 2m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b96a5ee2ad9045aab8532496063461cc : SUCCESS in 58m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4fe1802e18894c358932ef7948ed9622 : SUCCESS in 51m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb53b2205caf47b2b6b4f695146b2d22 : SUCCESS in 51m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d05c84f2adc446ff958aeeecd900448e : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9cb0da5ad59d43a09683a0927a94b371 : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/04b37b889f404bbcbbbd3d17fb09b6d1 : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d4f21b1b341473f800799957df1b270 : SUCCESS in 13m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c0b7ff783d5c4469b81242f1ea009e1e : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8ad5da68440407c8e2f2e751ef70e5b : SUCCESS in 14m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d810632a8f2a4c408f5a9b9877032031 : SUCCESS in 43m 07s (non-voting)\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b21d61463302473682d462c80ea0ae86 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cc0715210cf402b8c1dd171e5b32c6f : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a92b13b0b5ab4e70adda26d1facf034e : SUCCESS in 19m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9b23cfe829f492eb372d56dadf3d23a : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e77996fd2beadff3cd4fea6264ba312e46103353","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-04 12:08:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8307a532f04f6145a1eb98a4b13c348548c127a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-07 13:40:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66df77f79ac65182534fc5a220a937964209c683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:40:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41fff981fea520f79610e0c564f740d0d34c2f2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 15:37:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9591580c14a44f6bad5d7130b0190151\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9006b612aa414b76aa4302c360baef5c : SUCCESS in 2m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b3e867a6430246cd9d7053de9dec8dee : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae14ebfff0ce49269a9a7f70f5902242 : SUCCESS in 59m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1ed7d0971ec84ff8bffbbbbe11a5d217 : SUCCESS in 1h 12m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0fc552a8e5564ec499ff56244f178070 : SUCCESS in 22m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8dc91aa80ff4a9cbea25345f68d1a94 : SUCCESS in 15m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b2475438d76e45dfab7f32ea2bd81a51 : SUCCESS in 19m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ded4100f9c04e2cbdfdd367086e22a6 : SUCCESS in 18m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef039b1a228b4394a56295fb87bbc732 : SUCCESS in 15m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf811e8c7d5a420e87ea6dde2729f9a4 : SUCCESS in 11m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/620aa7eb6f194b838f840a0b2e54ab2d : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f303c793469648daa5d8206c65085a83 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1f63aff470c613ad5094365a9a3c98d8837b3b3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 15:37:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"537e8f894699280c067b738176ee4fe94f93d8d0","revisions":{"537e8f894699280c067b738176ee4fe94f93d8d0":{"kind":"REWORK","_number":1,"created":"2024-02-07 07:03:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/907940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/40/907940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/907940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/907940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/907940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/907940/1"}}},"commit":{"parents":[{"commit":"98721b81e0e68493db53c559c94cb591b6ae401a","subject":"Rename disable_selinux_do_reboot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/98721b81e0e68493db53c559c94cb591b6ae401a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-20 08:36:46.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-07 07:03:08.000000000","tz":0},"subject":"Use dummy1 as bridge port instead of eth1","message":"Use dummy1 as bridge port instead of eth1\n\nWith NetworkManager, the MichaelRigart.interfaces role only creates an\ninterface with the dummy type if it is called dummy*, otherwise it is\ncreated as ethernet.\n\nThe default dev configuration was using eth1 for the bridge port, which\nwas causing issues on Rocky Linux 9, for example when following the\nautomated setup [1]. Switch to dummy1 to ensure a dummy port is created.\n\n[1] https://docs.openstack.org/kayobe/latest/contributor/automated.html\n\nChange-Id: Ib61a62e126c66d4133d8e80dff5643ebc6966bd1\n(cherry picked from commit baf5c9f7191bf155b52072f494301474bd4b6755)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/537e8f894699280c067b738176ee4fe94f93d8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/537e8f894699280c067b738176ee4fe94f93d8d0"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
