)]}'
{"id":"openstack%2Fkayobe-config-dev~910017","triplet_id":"openstack%2Fkayobe-config-dev~master~I9899ca0542b81d6e3752c464bdebfb14cbf0ed48","project":"openstack/kayobe-config-dev","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-04-29 13:55:11.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-04-25 11:00:33.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I9899ca0542b81d6e3752c464bdebfb14cbf0ed48","subject":"Replace docker group with container-engine","status":"MERGED","created":"2024-02-23 17:23:51.000000000","updated":"2024-05-30 09:52:44.000000000","submitted":"2024-04-29 13:55:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910017","meta_rev_id":"b573c5e4a78b95de4e0634563c57841187edf77a","_number":910017,"virtual_id_number":910017,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-29 13:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-04-08 10:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-04-29 12:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-25 11:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-29 12:47:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 17:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 17:18:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-04-08 10:26:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-25 11:00:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-04-29 12:47:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1a5557b5353ff1522c845a1ffdc6151ea2596274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 17:23:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"775a6d6af680d4a0fe1b99d4d3a2eefe0f5e4d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 17:50:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db9c5f0b0ce2446a81a42026d5409456\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87cb3539acdf4f4b9dbaf578ebc92fca : SUCCESS in 2m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/656fff4843a947c688d37a40ea905d7e : FAILURE in 16m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/147a9dd5a9564dfd9771a740f8bee61d : FAILURE in 12m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce0ca63428e14af9a13b1620d7d1d718 : FAILURE in 12m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0306583091d5466caa9668f7cfe53b57 : FAILURE in 15m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5ed2fbb27c7a49ed978e74d23826a627 : FAILURE in 15m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e971305e5c34499c9a319e6d9da3555b : FAILURE in 16m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e023a81a76a144d08ad0bcc049d7776c : FAILURE in 13m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cead466b061e4853820229324eeb03d8 : FAILURE in 10m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdfffebb51264bfa98eabf3683cccf3e : FAILURE in 10m 29s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/de7d7075415341ca9b60c590be46a573 : FAILURE in 10m 35s (non-voting)\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/27f92b7c8f214a20aee8570fa2fde7d7 : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c85ea668d49143a380386d39ceeee730 : SUCCESS in 13m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adb258e74c64474aa8ad81e427373369 : SUCCESS in 18m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/879b66d66d854c3ba814c3779d440090 : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"88c894f4249723536481169692a80f1adcaaf347","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-01 17:18:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a373e2d7c0e247848be07bae9fbc0979ef584cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 17:44:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dda1bcbe1ebd4233b792adc12654585a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9548c03868dc44a1a1f249c17c5f8c5f : SUCCESS in 3m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7821879a336c4913ae8c9e8dd7923ee7 : FAILURE in 13m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e84a55bc64454dbeb069f891ca027414 : FAILURE in 15m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f025112393d44514bef4c2bc375fbcfa : FAILURE in 11m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ed11e01a47c74b99af9c8b7c7453e2d4 : FAILURE in 15m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/90f8766a03d14071b5fd7745243be742 : FAILURE in 17m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/aa4f2083f85c4c97928a6f26a705a72f : FAILURE in 17m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ffb127bbed9401fafaebe18a0e450e7 : FAILURE in 18m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6a4927888dcf4b1591edaeb7740ef4ae : FAILURE in 10m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f31502f8871427eb3e1a24d63baab18 : FAILURE in 9m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/70a492a1a5ff4912a01285d2886da92f : FAILURE in 10m 45s (non-voting)\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02857851cd0841a9afaffd5b87d7e197 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08f6105bd8814d768122e7e3c6bb09fb : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2ab9366af54f48b0a4405dabd0282fa1 : SUCCESS in 15m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/894308083def458ba862ef3551331f88 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e744192687ca85503b94c9a21e013e2f8be50d4b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 15:00:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"359e393c73093326c9f5aca283b6d9f970f572b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-18 14:49:01.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"779b578536c80444163cf981d9bc1ad60ce732dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:15:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0e8ed9e793e421aacddf35814e2cbc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd76cfd5f2f485eb8ecd994eafdece7 : SUCCESS in 2m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d57dad3853524ac8892284369ccc1d30 : FAILURE in 12m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/13e572f8ecbd43e5b915dee4f18cb4d3 : FAILURE in 11m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4fc10667b644efd871110a32371ca4b : FAILURE in 9m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9e804e7013f1450090e69554526f9d8b : FAILURE in 16m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c1d817ba647243b8bc13c934b08a5e02 : FAILURE in 15m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e8d61259b8274d96ab71b6e5dae2e7bb : FAILURE in 15m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ec9237d3cf64737898636261c5a55ee : FAILURE in 12m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1c176d79ef9545cdb6e9a30634b1a6a2 : FAILURE in 10m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2462c47158c146289b28651e40ec442a : FAILURE in 7m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fba329591b0249dc8f3b8e55e211f21c : FAILURE in 11m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82c9f560958044c1bd5200f853b81acd : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93ae900434264f949934ab6e6851f27b : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/39163b0a30064a54904b96caf1d48e38 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5332c1e80174b22a1e86e2d956204e2 : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1b77d79dbf96422b1ed5ecd54dcbecd805ca2206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-18 16:46:24.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bcef0267f8ce126237401b51f99f5c3d630d6d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 18:05:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d95a7be5d441bd9bc0b7cee432acee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3214cff65994c44b4ff1c377433d2dc : SUCCESS in 2m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1a737d86601441b2b9c606e68f6e9229 : SUCCESS in 1h 13m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6c31aaf5edeb4749bb0619adf533ed7e : SUCCESS in 59m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc2e5c2d79ce49dc81782c3acd7eedb8 : SUCCESS in 52m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cdb6e1bcf2774861b9998f101d43b56a : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0f21c94d8ebd453bb971015c8bcd8658 : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1c16c99d72454fa58352d687c18eeff8 : SUCCESS in 15m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d15fac20feb4973858e324dd5c58e9d : SUCCESS in 14m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3d93659c7ddb4dbe98f4f3a5ee2030bb : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3160d218d7546b5a663e27ac07d0b2e : SUCCESS in 16m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6309d337b84e449abe55f24cbee2e87d : SUCCESS in 41m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a93bff6ee0694924a9a6f8201d866dee : SUCCESS in 17m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a36bc6ccb3ce4779b5ca727555002da0 : SUCCESS in 12m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f49119f265247ba8cdf34e0d45926f4 : SUCCESS in 13m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38accd5f385f44c0a28ab7a2495a8bb7 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c6f7b7a26677619acd331dffcd790b385de02eb9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-08 10:26:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2a6920bf96b2f40492ec8610ecd10a188e111eda","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-25 11:00:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d4c1b76690453651491f7c4a94c499acfe563eff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-04-29 12:47:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9e4ee05db94a3fe615464fa1c0070588c775a9ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 12:48:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b52cd2bfac944740f0c89412029d7b71a2e6f45b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:55:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699c34f0549d43b78694545f6f7163f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce936e8db17d4896a7a9177638498cf6 : SUCCESS in 3m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd2908b057b64a02af9a7560ceafe357 : SUCCESS in 58m 08s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80ca6fff5d214ceb9d13430c9cdb8b53 : SUCCESS in 58m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/234d6682f9d848b69a9e3a33dd26deb5 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b5db0fa8ea64447da56ea9edb9878195 : SUCCESS in 15m 39s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8814efa06abd4076957f9e2798a8abf2 : SUCCESS in 13m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7331c9194ba7483bb30c75fcadf8e155 : SUCCESS in 15m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c8740ab4c7448d19a817cbed4459ccb : SUCCESS in 13m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/512409f89c8a4b47bb9adbb5a0a5391b : SUCCESS in 12m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/048be51b057943cda85a78c58a1f77a6 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c056ae41fc7b4561b6c4e3bf3564c8c9 : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/393d5f5a1c0046cc9f6f416ddbe6b4b8 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"57b1e29a4e1003f8a4c8085a07aee29e03a8de7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:55:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b573c5e4a78b95de4e0634563c57841187edf77a","tag":"autogenerated:gerrit:revert","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 09:52:44.000000000","message":"Created a revert of this change as I501589f04a0d9d1b52f88c841b3d28984eb8b0ed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a129cdb8ef30b28200a3636fa4223fad97f27a6","revisions":{"224e08ddfbe7306692a3240354258941d413f36a":{"kind":"REWORK","_number":1,"created":"2024-02-23 17:23:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/17/910017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/17/910017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/1"}}},"commit":{"parents":[{"commit":"baf5c9f7191bf155b52072f494301474bd4b6755","subject":"Use dummy1 as bridge port instead of eth1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/baf5c9f7191bf155b52072f494301474bd4b6755"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 17:21:49.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 17:23:48.000000000","tz":0},"subject":"Replace docker group with container-engine","message":"Replace docker group with container-engine\n\nThis is part of the move to support podman in kayobe.\n\nChange-Id: I9899ca0542b81d6e3752c464bdebfb14cbf0ed48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/224e08ddfbe7306692a3240354258941d413f36a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/224e08ddfbe7306692a3240354258941d413f36a"}]},"branch":"refs/heads/master"},"0ade8d77cea7225b33514304435bbf1f49336752":{"kind":"REWORK","_number":2,"created":"2024-03-18 14:49:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/17/910017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/17/910017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/2"}}},"commit":{"parents":[{"commit":"baf5c9f7191bf155b52072f494301474bd4b6755","subject":"Use dummy1 as bridge port instead of eth1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/baf5c9f7191bf155b52072f494301474bd4b6755"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 17:21:49.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-18 14:49:00.000000000","tz":0},"subject":"Replace docker group with container-engine","message":"Replace docker group with container-engine\n\nThis is part of the move to support podman in kayobe.\n\nChange-Id: I9899ca0542b81d6e3752c464bdebfb14cbf0ed48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/0ade8d77cea7225b33514304435bbf1f49336752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/0ade8d77cea7225b33514304435bbf1f49336752"}]},"branch":"refs/heads/master"},"4a129cdb8ef30b28200a3636fa4223fad97f27a6":{"kind":"REWORK","_number":3,"created":"2024-03-18 16:46:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/17/910017/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/17/910017/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/17/910017/3"}}},"commit":{"parents":[{"commit":"baf5c9f7191bf155b52072f494301474bd4b6755","subject":"Use dummy1 as bridge port instead of eth1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/baf5c9f7191bf155b52072f494301474bd4b6755"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 17:21:49.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-18 16:46:24.000000000","tz":0},"subject":"Replace docker group with container-engine","message":"Replace docker group with container-engine\n\nThis is part of the move to support podman in kayobe.\n\nChange-Id: I9899ca0542b81d6e3752c464bdebfb14cbf0ed48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/4a129cdb8ef30b28200a3636fa4223fad97f27a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/4a129cdb8ef30b28200a3636fa4223fad97f27a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
