)]}'
{"id":"openstack%2Fkayobe-config-dev~920811","triplet_id":"openstack%2Fkayobe-config-dev~master~Ib3f82028b275130eac31c1a297b2598dadea5fb2","project":"openstack/kayobe-config-dev","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-05-30 11:52:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3f82028b275130eac31c1a297b2598dadea5fb2","subject":"[release] Synchronise with kayobe for 2024.1","status":"MERGED","created":"2024-05-30 08:10:33.000000000","updated":"2024-05-30 11:52:47.000000000","submitted":"2024-05-30 11:52:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920811","meta_rev_id":"71b9e66aa2c7bf216f03bd0b414371d6233039c7","_number":920811,"virtual_id_number":920811,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-30 11:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-05-30 08:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-05-30 08:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-05-30 08:43:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-30 08:39:10.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-05-30 08:43:26.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-05-30 09:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53668da5be9ef1d2006b0640f6049625771a63c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-30 08:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a06ff6186eac24fbf34bb39e19a2214b471669b7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 08:39:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8764f2e7a9fff767588efdb3c5fba52898c4c393","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-30 08:43:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5073afc990af82c973aab28d448f80aa59fdd539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 09:36:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a356072d9184fe8a3cc8a1c6ea8dcf1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2711ee21b8914e56821bcc32ed16a834 : SUCCESS in 2m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8a30da56800843568b8d897b85ced780 : SUCCESS in 1h 04m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/83892edb98254f52899d31caeefd6927 : SUCCESS in 1h 08m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fe11b666ae041aaa148e1ed33515380 : SUCCESS in 50m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/17d0812b493a4e9caeeb76338cb2454c : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d3a5156ef01049be8972d328b5c40d83 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/23b71a2abe184deea15fe51cc5dabbbe : SUCCESS in 18m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8bcd88313e44202aeb1142a13bca095 : SUCCESS in 12m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ac5fcad1cfac4d1ba131ae2c39d07014 : SUCCESS in 17m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96fa52e3d35c44eab87284e9220fcc9b : SUCCESS in 11m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/87c45396b0ba42a8880ff991059ab42f : SUCCESS in 38m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76fae07380f84626bceaaaa1aa1e5745 : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab4875b3b6374ef6b4bb04dc611174ba : SUCCESS in 9m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f92b9e0af6eb4286b53899e4786588ce : SUCCESS in 13m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19fd6219f9bd44238d878d3dc1c61ccb : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e08c6f9cbbb01ed65e6ebe0046bf3b9de54df759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 09:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a683b62fddcca2cc1176c40756fad826968c0f1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:52:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e5486065104031aa628bfeebed7ca1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6482da6ae2b14bcbb96e7a869d047dec : SUCCESS in 2m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fda823eb84e148d9a2c5f0bc92646ca2 : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfbdb25f0b97454dbf65b0714960ebb3 : SUCCESS in 51m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c75ebe558c024d6a95a455634f8b2775 : SUCCESS in 1h 12m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1f21d3d1b70646a8b25b081407383389 : SUCCESS in 17m 17s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08401e9d66024b69bceb2affcba1529f : SUCCESS in 13m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/54b6075032d144feacef0a9e2ac1aa04 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f68cdc5822904b3a86131f042ea2884d : SUCCESS in 17m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/044e784f2f3b4e92b4f7fdd66448947a : SUCCESS in 11m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2914153a5b64468a4cbe8310b7434c0 : SUCCESS in 9m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cee6c146172d4398bdf5a5974415a80c : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97ceca429de549a7bd1bada0c018306b : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"71b9e66aa2c7bf216f03bd0b414371d6233039c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:52:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"772832ed5ae1a667a0da427aa7a4f49f728ce278","revisions":{"772832ed5ae1a667a0da427aa7a4f49f728ce278":{"kind":"REWORK","_number":1,"created":"2024-05-30 08:10:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/920811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/11/920811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/11/920811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/11/920811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/11/920811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/11/920811/1"}}},"commit":{"parents":[{"commit":"4a129cdb8ef30b28200a3636fa4223fad97f27a6","subject":"Replace docker group with container-engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/4a129cdb8ef30b28200a3636fa4223fad97f27a6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-30 08:07:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-30 08:07:07.000000000","tz":120},"subject":"[release] Synchronise with kayobe for 2024.1","message":"[release] Synchronise with kayobe for 2024.1\n\nChange-Id: Ib3f82028b275130eac31c1a297b2598dadea5fb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/772832ed5ae1a667a0da427aa7a4f49f728ce278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/772832ed5ae1a667a0da427aa7a4f49f728ce278"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
