)]}'
{"id":"openstack%2Fkayobe-config-dev~929602","triplet_id":"openstack%2Fkayobe-config-dev~stable%2F2023.1~I2ade9585a0f23472d016c311b70c3ee71bdb2726","project":"openstack/kayobe-config-dev","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-17 15:36:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ade9585a0f23472d016c311b70c3ee71bdb2726","subject":"[2023.1-only] Synchronise with latest kayobe","status":"MERGED","created":"2024-09-17 08:19:16.000000000","updated":"2024-09-17 15:36:38.000000000","submitted":"2024-09-17 15:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929602","meta_rev_id":"a3140b82152d3aad3a149f1024d599226b18aeab","_number":929602,"virtual_id_number":929602,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 15:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-09-17 14:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-09-17 08:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-17 08:46:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-09-17 14:15:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 08:42:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-17 08:46:52.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-09-17 09:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 14:15:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"ca517d7ffcdb50b35397004931e86e978c561b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 08:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7d24d628f0348494b76b120fe9ee0ee276c4648","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-17 08:42:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1db9ad10b144709994a157163e0410b9f26b9c49","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-09-17 08:46:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05230d78e094a380634af8ad29cd3f12cd6e620b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 09:37:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f73d75bbd7a947439929abb8c7a15c19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1487fc8c0243fa8a35c1ba4d156e4a : SUCCESS in 2m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1691260eb8754c808462ca1f96f3f1ce : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2cef696e598642d79d6a765df090c418 : SUCCESS in 58m 18s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ad37d42fb6748048aa5500ed767389b : SUCCESS in 33m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0626cdfa125c42f6ae6ee6b451981776 : SUCCESS in 47m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/67acad47dcea4b7c9dfb7c9a8051fd60 : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a682d67e441f4dd092a2f64e6ec3e089 : SUCCESS in 10m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e29c1cd7240c47ad9c588de56e3af5f6 : SUCCESS in 13m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0a30dac125b44ad08e03ed50f75b30b3 : SUCCESS in 22m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d1f3e5100704579acfe62e497b04b84 : SUCCESS in 15m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a64f374f425746d4851ff37cd05c6c77 : SUCCESS in 26m 00s (non-voting)\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c1fd814bab26480e9a36679c197111a5 : SUCCESS in 10m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/038f06a5002c43c383a09046d9dd7386 : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45dc636ca6c64c7abdb9f59dc1c52e79 : SUCCESS in 22m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69ff87f72e7d43eda1a7fb2812c300c7 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ab316767606f8bf13fcf9234998ca4f6574d0f2a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-17 14:15:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9a149708da44a315fda9b851d7439826e47b8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 14:15:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6072f567e7a09848a13eff3055e1e3ab7e171975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:36:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef06f4d7c0f44a9befc96e07a5b79d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e694c30f9a40dd8c9a7bba7a06b7c8 : SUCCESS in 2m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e1a3e885ee4e4c1da5365fc63f17d213 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa8f4e43a52e4b65a658dbc065c12230 : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7bd5096bd2b041c7bb92b14cbf3615af : SUCCESS in 1h 13m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9802d0a7bc0947a3a830237bef3c603e : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c03dbacb3254818b77938a97bfcc8c3 : SUCCESS in 14m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e2689300565e4ff9b69becf1d0bad03b : SUCCESS in 19m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18d15ba8bf504f898529d2ebf8ce081b : SUCCESS in 17m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2643e47a05fb478ea776dfc5478ebdc1 : SUCCESS in 16m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02893a7d026748d59effd4ba9e25704d : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52a41dce323d4fc49d6e3d4dee8f1dc0 : SUCCESS in 21m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8917c1005fa49acb0cf7ff2ef426c90 : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a3140b82152d3aad3a149f1024d599226b18aeab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:36:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d78c7e97704636712cb3de4b5f45d4270382e05","revisions":{"8d78c7e97704636712cb3de4b5f45d4270382e05":{"kind":"REWORK","_number":1,"created":"2024-09-17 08:19:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/929602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/02/929602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/02/929602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/02/929602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/02/929602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/02/929602/1"}}},"commit":{"parents":[{"commit":"537e8f894699280c067b738176ee4fe94f93d8d0","subject":"Use dummy1 as bridge port instead of eth1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/537e8f894699280c067b738176ee4fe94f93d8d0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:18:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:18:45.000000000","tz":120},"subject":"[2023.1-only] Synchronise with latest kayobe","message":"[2023.1-only] Synchronise with latest kayobe\n\nChange-Id: I2ade9585a0f23472d016c311b70c3ee71bdb2726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/8d78c7e97704636712cb3de4b5f45d4270382e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/8d78c7e97704636712cb3de4b5f45d4270382e05"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
