)]}'
{"id":"openstack%2Fkayobe-config-dev~929603","triplet_id":"openstack%2Fkayobe-config-dev~master~If3180d009d34c9c99fb1b237f0e32e928f04f6d2","project":"openstack/kayobe-config-dev","branch":"master","topic":"ipa-c9s","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-24 10:55:02.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-27 06:03:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"If3180d009d34c9c99fb1b237f0e32e928f04f6d2","subject":"Use CentOS Stream 9 IPA deployment images","status":"MERGED","created":"2024-09-17 08:30:29.000000000","updated":"2026-01-27 09:09:42.000000000","submitted":"2026-01-27 09:09:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929603-ipa-c9s","meta_rev_id":"79013b09fe22b30a844e966a8c05fc2bbc9b2826","_number":929603,"virtual_id_number":929603,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 09:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-26 16:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-24 10:55:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-26 16:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 16:39:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 08:41:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-17 10:41:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 16:00:35.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-01-26 16:39:12.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"49657a3ec27db1e3f97512a9298149d9cf1907c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 08:30:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d15da3aee87c7287a2e819f6fd3e6b6a748eac47","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-17 08:41:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"22bb7124e2f5a7e8d6117ca506c25f48b9239988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 10:41:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a92e974f5d842e88f0377bf83a974ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6834584deb640aaa0257ef5aa373f9c : SUCCESS in 2m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/350ba170317f437a8751558773e1ecff : FAILURE in 2h 01m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7b33e3afd73749fa946f869be55b97f1 : FAILURE in 2h 02m 54s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19e25615d32c41048747b43614561593 : FAILURE in 1h 46m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3bc494a2bb304977bd83b5b24e0ca173 : TIMED_OUT in 2h 02m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e09fe01417624dc082c1e16b8b147e09 : SUCCESS in 21m 17s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e70a857955d643049794a3a50b906d84 : SUCCESS in 16m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86899eae1c0c42bcaba63f761c40119e : SUCCESS in 12m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/22543fbfaa86410eae3cb06cfb7a326c : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be0bf4035e3942e2b2a015ff7ddd7f0e : SUCCESS in 14m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bbc2f6cd09104b79ab7bae98f933f9c6 : SUCCESS in 41m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/439edbb93b0d499c886eccb01038f778 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63a0e9f94db840a1b72024e210de793c : SUCCESS in 7m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/737dc056396f4f3e8c9a70b1f15252a7 : SUCCESS in 19m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52f2f066f2ee4b8da9cf69f23521658d : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"57ae848e7d8a094d4c334a508a8f49e2e157c62e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 12:39:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"88bf5d0c6273f0d2fa9b15d0e5568cfe5f90f921","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-06 15:32:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"749bb84dfc7bb413a4ca2cd968847dbaf39fc3a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 17:18:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e391db38b004fa0aa438f897797b659\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c4b8de165d4a16b39a3ee3403b5e80 : SUCCESS in 6m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ba1f00d9fd5f496caa2b4bc53e92d535 : FAILURE in 1h 35m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/051a8d4c5c3c4625965771cdacc33b91 : FAILURE in 1h 37m 42s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fe87c3e53e164c8886e44109535eb7c4 : FAILURE in 1h 44m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/81be41517ac745ff9b395ef40f7c6d5c : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/df1ef62cb8ed4e3c856fc22c3990b3b0 : SUCCESS in 36m 29s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/30ad8a67b1d04a77bb99cad7d6be1978 : SUCCESS in 19m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d99b44f0f6544f9bb5314725aa9c353e : SUCCESS in 44m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f58de45cb20f468ebcfcc82d685d247c : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e46d9b100e84483a6946785c909055a : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0c1361060de4f3c2673e98f5378ac3664f294749","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 17:06:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dbf4bfbe914410876e080390e44927bd21456eaf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 17:10:10.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a53c84129757ab62af67bfa2bae458f03b61fd16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 17:11:22.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"916e8a848cc13267adc4bd54cbc2d4cb3275de7f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 17:33:14.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"0ccfe99bae07cf02e4f4c6c16c6888876475e59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:47:49.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/974490 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"a435138f5586b3a58f416f799fef9bbd1ff2028f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-24 10:55:02.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e3799cba2770aded88b1d9921798a8115fe7ab83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 13:02:02.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/573caaa25f084f14a64981bfa795bc9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adc623344074c8d9b1fb4c66b48b758 : SUCCESS in 2m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/947b02e8ffe5458fac1fa8abe1e2d5b9 : SUCCESS in 18m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5331c00a169e4008bca94c8a0ed41424 : SUCCESS in 12m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6c0ebb7ec37c45a5b49a738ad3722b32 : SUCCESS in 24m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71d08e4771704312b120b8e1c6e3dbc7 : SUCCESS in 15m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0c32f36df08c45abbdc86b3ecb2d0c9a : SUCCESS in 50m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/609d9be8b4e04681be7d8406f6981848 : SUCCESS in 51m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f465a46cbbad425c97fc9ba0b0911aca : FAILURE in 2h 00m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88608c8a90184e83a328957260fa8ddf : SUCCESS in 50m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4bcd3526f7c448368db140c017239ea3 : SUCCESS in 48m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93945040c22440f898f89464ffe8de43 : SUCCESS in 1h 28m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1fc4e622ebc243c5bb30cd5f7516727b : SUCCESS in 21m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/170baeff663f405096fe2b76f1d871be : SUCCESS in 21m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/076775880d0347c49f2b3c8311850364 : SUCCESS in 19m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f54300d9fb55429da9654813d90dfc06 : SUCCESS in 8m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94196470fa4f49d5b766dff95c15ef18 : SUCCESS in 25m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/09c85f214a0f4805bcdaf4e710b5cb69 : SUCCESS in 18m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9c8840ce34447d1a92e2cb403e55b52 : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":4},{"id":"c72f089f138c518fe650966f21287bd516696183","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-24 19:11:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"81ec0344598ae319413a7c9550be31896f2fad8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 20:35:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68678529ea44486da499cea948d9ab50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d78a340a154be7a8b44e5235d18d02 : SUCCESS in 1m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a449482260164321a9e1337460789440 : SUCCESS in 8m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/218ae17458864151a62dc68e02fb9ca5 : SUCCESS in 7m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f7f25232db9e4a9981dda87ba1ea05ac : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/019517e567054aa5bd742c21a33ce392 : SUCCESS in 14m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4d46609d8295414d8f7e3a1d15b20651 : SUCCESS in 47m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f4f783d6d8145d88b3dafdd3f840f54 : SUCCESS in 48m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/17ba93ed766644d2be63b6fba231eca5 : SUCCESS in 55m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e20a9eb1c234481b8f77d5d601f0cc16 : SUCCESS in 47m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/296fc9478b964c77818ce835cd6ec33b : SUCCESS in 46m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/823578a675eb4264b8fbeaf80f6739a9 : SUCCESS in 1h 21m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/afa194c93afc48b09d91d6f3f8e1bdee : SUCCESS in 21m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e13a91b26844f47a290e22755d92039 : SUCCESS in 9m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9745c032b71b4c41b8f61a6e5a76d5b1 : SUCCESS in 17m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0306914fab1047d9840a1ec7ebbcdae7 : SUCCESS in 7m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a5c9d0486ae4ebf98df464ffc88b1b5 : SUCCESS in 27m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fcfd0f06f244953addbfa5f89a37351 : SUCCESS in 9m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4698d8efc0a04553ba8f5b115d7713bb : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a37e9d43d621e5689c78ce76ad4930ac38c8b540","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-26 16:00:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5350ee54875324b9526f705b3602951d6ae9dd96","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-26 16:39:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aad445378dc6f60497058130e7017458531f7137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:39:14.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31fa953afa3247df989774a11b5dd968\n\n\nWarning:\n  Change 974490,4 in project openstack/kayobe does not share a change queue with 929603 in project openstack/kayobe-config-dev","accounts_in_message":[],"_revision_number":4},{"id":"9f3641b4047a9187d4a294e9191f4bc0bd6c81b3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 22:36:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c740203a7c55ea271e5aa5f3b08a1aaa21026213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 00:04:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581ecb6867834fd29400554c9fd93410\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24290c449d14125a6cf946fcd6f0e13 : SUCCESS in 2m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0bcc2c6423a045609d1f0d426059b264 : SUCCESS in 9m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0304a6ac8e79465ab8c3794a7ead63c6 : SUCCESS in 9m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/79289e929f534f989a89cc102858a48d : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc2f266db28d4052bf0c141bfc6058c8 : SUCCESS in 16m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/457830b5b63a428c9cd094b5ca1b8e66 : SUCCESS in 50m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3b1dc7aa0c39418491f3a8630ffece80 : SUCCESS in 50m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/489d8cd038e240d6a05e268d9327dec5 : SUCCESS in 58m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/338ebd8ece964c7a9569514dfe7b4af6 : SUCCESS in 50m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d7ecec2b6c1491d9dfde06932c62c58 : SUCCESS in 47m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/005adef9f21f494a844c39714a78ede5 : SUCCESS in 1h 27m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d61997c1b11246b18c2f949d623f7416 : SUCCESS in 1h 24m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a5ef55751a954c6ba3ff4ec14d1ccbd4 : SUCCESS in 9m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1cf481b71d2246edb42a16b824531f68 : SUCCESS in 9m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a86e4534435743f78402aa9effceb800 : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d95d98c4d129474596f12bc25f0560a3 : SUCCESS in 14m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2244672bfed54b5598f2a55b258749d4 : SUCCESS in 14m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0d56adb9bee4af4a3aefa2cce51770c : SUCCESS in 27m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/25fecb64e0f04ce0b0691168f80dee1a : SUCCESS in 8m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70ce505fabfc4c4f9d8c43c723752498 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":4},{"id":"8f8a388aa0a8153ae2a1c200a4148080b5699a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 00:04:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f2fcef996e3c1551c44a639089d3ea3de45da58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 01:33:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/238c1f11cb3f4275b3269021c4dcb65e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c2649595c146ce95d9d363bb641a3e : SUCCESS in 2m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45bc27fe3c4c4793bfcc99969c1ea6be : SUCCESS in 12m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2896fd3cc8774f8ea58d91cf1de5ff3a : SUCCESS in 7m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fdac2d65edae4fbbb704bf3cacc84488 : SUCCESS in 23m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebe116f78e3d4b3ebf7cf70b5671c606 : SUCCESS in 8m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d9f6340d22344f55bf9a88a9cf616553 : SUCCESS in 50m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa9afe5a98794157bba84cf9505881bc : SUCCESS in 51m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/780577380b904536b829159321e96fde : SUCCESS in 58m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c4ec936246742399d1742333858d738 : SUCCESS in 50m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7546d92402c4764a0942a7d6958a2cd : SUCCESS in 50m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1dd3cd99226a40a4b76efdb04978421d : FAILURE in 16m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d818d6917df84d4a919e4986966668b1 : SUCCESS in 1h 27m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/83242bdf01a84d3ea7870a092b3699f0 : SUCCESS in 10m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e39ec2e75f7c46bcb083410788472c83 : SUCCESS in 9m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2cd8b7789714cdfa882f116a09a54d3 : SUCCESS in 11m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f4a1b41e0ea546669498e65b817bfb1c : SUCCESS in 7m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b8623ec88d194ab787ceabc539e4c3ed : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b53c78bae1f4d96b35ca55ad9d260aa : SUCCESS in 15m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/364ae7631c7040908de8d51cd72b2277 : SUCCESS in 18m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a79cf89376e4ab6a5724062a6d5c81a : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":4},{"id":"46f32a51e03f21372ceb717b3fca1c3e2e992006","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 06:03:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3b6fae039c0805d48848a5c691a47493afe96935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:34:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b394ca8f114646738ea1e5065154ebdb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4c6010dfee4d09b5a90815fd647881 : SUCCESS in 1m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/30c6455dbfaa43c1b1ab660e1370b8ab : SUCCESS in 8m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e06ff97a1a34568bfd0f4018392abf2 : SUCCESS in 6m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2d1aa5ec316d41b591f899c0fc02f97d : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45dab6efa8d9423a9161db70dceb24b4 : SUCCESS in 15m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6edbf40c37204264b36db25d8bf4fdb6 : SUCCESS in 52m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/09301f04c75c44eeadafaf6123ec4662 : SUCCESS in 51m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/06ce18894dff4632a08811ff79787a74 : SUCCESS in 58m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a960d2c02cf45c3bbed2d669410dac0 : SUCCESS in 49m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/90aab0ea159644e780ae6c5eca271c9b : SUCCESS in 52m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/62338372ea2645299e7faf914c063a2b : SUCCESS in 1h 29m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0f50815b4514f9fa657f93e91c097a5 : SUCCESS in 1h 24m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3fd6b658dbd745ae873ebf05e487d502 : SUCCESS in 9m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/09ec9e8e3f73475eb7d1116ede55d317 : SUCCESS in 9m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6176603c76f64a4faf944a4f6c087af5 : SUCCESS in 13m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/67858900f68748b09aced2752f88bcbf : SUCCESS in 12m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b56acd2f5df461cab821150c39dd228 : SUCCESS in 24m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0427e16d726a449386b2a6223988d41a : SUCCESS in 25m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f5d483e17934fb08aa74876bf8ac981 : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1062dd1529de41a2a3967e3ceb3b567a : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f546da0089f32572943529ce7dbc646b15ae2a24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:34:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"94fab1dd837c18de4add64960cfc903cc78ba480","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:09:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34e5ab218394e63bc2ee909270e5d31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40dcc28dd51b4fd88ea4d18625b7cf6a : SUCCESS in 2m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e42bcffbed6349b3abf6647a610bc5f4 : SUCCESS in 14m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9cf617b47744e6db33b3700e492441b : SUCCESS in 11m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/61682c73ac3346e99172ae0e65f7b050 : SUCCESS in 24m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c22f619591d249f7b78b70c1566487f1 : SUCCESS in 12m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/937bb0bf46e547a4b3c66ea40c2bcfa3 : SUCCESS in 51m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/076a0862892d4a3699874bf1c8542f9d : SUCCESS in 52m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b6b1230b1268479a96c2b750a7106974 : SUCCESS in 58m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0e8d743c48b47cfbd1790cd43719786 : SUCCESS in 49m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a77cc9404e5940dcafe2954ed2cf0912 : SUCCESS in 50m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a97fbf0c05ef4da9a3f524b75dab7fc3 : SUCCESS in 1h 34m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d360044d8a30432d9f1e14ef611ba5c1 : SUCCESS in 1h 29m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f67d607b15540d7b309d6ea56f6f5dc : SUCCESS in 21m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f2bba754c3d4da7ad4d2f55539ec037 : SUCCESS in 20m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02dce7c0870d4a4dab0a5da2505afeed : SUCCESS in 20m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf64f0d0a6144cc8b54ace4dda970118 : SUCCESS in 14m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/21cfbf7cf32e4bacb5db4a9afe66ec05 : SUCCESS in 21m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f69fb480f2b4e789a1dd7c71a483666 : SUCCESS in 28m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9e457b5f2d1e4c559213a482d764df58 : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33e1f0b224a94e4fb92314116dbd555a : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":4},{"id":"79013b09fe22b30a844e966a8c05fc2bbc9b2826","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:09:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b4d887772c9448ab31cc401aec35a33d3909794d","revisions":{"546ee82afcd393abb43f2a47f04588605ddfa597":{"kind":"REWORK","_number":1,"created":"2024-09-17 08:30:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/929603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/03/929603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/1"}}},"commit":{"parents":[{"commit":"75047ec8c03a8da2b568f4f0f36d5e83bd1e988b","subject":"Sync with kayobe for 2024.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/75047ec8c03a8da2b568f4f0f36d5e83bd1e988b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:30:05.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:30:05.000000000","tz":120},"subject":"Use CentOS Stream 9 IPA deployment images","message":"Use CentOS Stream 9 IPA deployment images\n\nThe TinyIPA images are broken in master and 2024.1 due to a parted bug\nin Tiny Core Linux 15 [1]. Switch to the default CentOS Stream 9 IPA\nimages instead.\n\n[1] https://bugs.launchpad.net/ironic-python-agent-builder/+bug/2068936\n\nChange-Id: If3180d009d34c9c99fb1b237f0e32e928f04f6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/546ee82afcd393abb43f2a47f04588605ddfa597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/546ee82afcd393abb43f2a47f04588605ddfa597"}]},"branch":"refs/heads/master"},"d02608ef6923c0a9809e3446335c677efbb445be":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-12-06 15:32:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/929603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/03/929603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/2"}}},"commit":{"parents":[{"commit":"7d423c45ae835b2167081a82dc0022318d22a3e4","subject":"Store ironic-inspector data in database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/7d423c45ae835b2167081a82dc0022318d22a3e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-06 15:31:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-06 15:31:53.000000000","tz":60},"subject":"Use CentOS Stream 9 IPA deployment images","message":"Use CentOS Stream 9 IPA deployment images\n\nThe TinyIPA images are known to have issues. Switch to the default\nCentOS Stream 9 IPA images instead.\n\nChange-Id: If3180d009d34c9c99fb1b237f0e32e928f04f6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d02608ef6923c0a9809e3446335c677efbb445be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d02608ef6923c0a9809e3446335c677efbb445be"}]},"branch":"refs/heads/master"},"d0c276eb7a422f0254994a7039366c7759646af1":{"kind":"REWORK","_number":3,"created":"2026-01-23 17:06:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/929603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/03/929603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/3"}}},"commit":{"parents":[{"commit":"b20eba7f25d483ccdd355735fe8af3604d6952f9","subject":"Synchronise configuration with master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/b20eba7f25d483ccdd355735fe8af3604d6952f9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:06:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:06:39.000000000","tz":60},"subject":"Use CentOS Stream 9 IPA deployment images","message":"Use CentOS Stream 9 IPA deployment images\n\nThe TinyIPA images are known to have issues. Switch to the default\nCentOS Stream 9 IPA images instead.\n\nChange-Id: If3180d009d34c9c99fb1b237f0e32e928f04f6d2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d0c276eb7a422f0254994a7039366c7759646af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/d0c276eb7a422f0254994a7039366c7759646af1"}]},"branch":"refs/heads/master"},"b4d887772c9448ab31cc401aec35a33d3909794d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-23 17:10:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/929603/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/03/929603/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/03/929603/4"}}},"commit":{"parents":[{"commit":"b20eba7f25d483ccdd355735fe8af3604d6952f9","subject":"Synchronise configuration with master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/b20eba7f25d483ccdd355735fe8af3604d6952f9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:09:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:09:57.000000000","tz":60},"subject":"Use CentOS Stream 9 IPA deployment images","message":"Use CentOS Stream 9 IPA deployment images\n\nThe TinyIPA images are known to have issues. Switch to the default\nCentOS Stream 9 IPA images instead.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/974490\n\nChange-Id: If3180d009d34c9c99fb1b237f0e32e928f04f6d2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/b4d887772c9448ab31cc401aec35a33d3909794d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/b4d887772c9448ab31cc401aec35a33d3909794d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
