)]}'
{"id":"openstack%2Fkayobe-config-dev~936168","triplet_id":"openstack%2Fkayobe-config-dev~stable%2F2024.1~I5d6489b092350078f1da4b8f740a097ea900b249","project":"openstack/kayobe-config-dev","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-11-25 17:54:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d6489b092350078f1da4b8f740a097ea900b249","subject":"Store ironic-inspector data in database","status":"MERGED","created":"2024-11-25 15:25:42.000000000","updated":"2024-11-25 17:54:28.000000000","submitted":"2024-11-25 17:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936168","cherry_pick_of_change":936123,"cherry_pick_of_patch_set":1,"meta_rev_id":"3d43ee9cca750328be085fb6f1ba18af060d975c","_number":936168,"virtual_id_number":936168,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 17:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-11-25 15:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-25 16:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-25 16:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-25 15:25:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-11-25 15:28:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-11-25 16:33:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-11-25 16:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c310ced827e512eb351022d7695a1109db80c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-25 15:25:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e9ecfb53f2527134381eed9efded37a2c6cff48c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-25 15:28:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4e3ae4982c4c0de9accb541f10dd7dac2813738","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-11-25 16:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7d7eb9d36001d9056722a39d14e2e781ce03549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 16:44:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2654ca941529477683176b213ca0dcf9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3639f969b58c4929aaf274d5a91b6b68 : SUCCESS in 3m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/35c3c660b00b47ef91f43baf67480f02 : SUCCESS in 34m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/656b97020c2249e0a67ac975fbac4f26 : SUCCESS in 59m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c5b1dda38ae4754ad9580edfd3e58e9 : SUCCESS in 50m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/85f7960def864665b9d72716039dc433 : SUCCESS in 40m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/86033bab952141f9a4ba2ae7d3d9f303 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6b1b95e8106b4d7787e625a1b3f733a8 : SUCCESS in 16m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4448e910b3648188dcc4f6d8ea834c0 : SUCCESS in 16m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8cd899e3db1a4ee2a5d3492bc533a59f : SUCCESS in 21m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c120168a11e742dc8d740e5b15365f73 : SUCCESS in 13m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/caab094d94364995b6ec7570210c3028 : SUCCESS in 40m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a4ca5e9f8e24cf4ae91b320ca63ab5c : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c10d31ba11db4819b37a1707caa2156d : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c9c65ec3e4042c2b884684b151720a3 : SUCCESS in 21m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e69bab30e5a421cb2262bfe8c230341 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3f421be392073a21f004598ed50f842b46d67b08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 16:44:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56365ff04c230bc15a576829c292c2c87119ffc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 17:54:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e8d1e68d9e74a079dd9736a1392e41e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4ba770f35f476089db642acead62c4 : SUCCESS in 3m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd3c4a3ccce74221b8cec66fa3967317 : SUCCESS in 35m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85f0fe30f255416c8e122f3da9b9af73 : SUCCESS in 38m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2936e6c66a204e6bb1c3451bc79ce6c4 : SUCCESS in 1h 08m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dba65a40149b4fd5a486eb92627e2b54 : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da44c85a35314bd4bb62a3d1e954bf73 : SUCCESS in 13m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5aa3c216d8cd482a8f5a4cd1428534a5 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae46d220731d43f3b05db157204b93d6 : SUCCESS in 12m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0fa42fdef8f3441d8083c0f9da5e1e1b : SUCCESS in 14m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e858ef1c2954414a9acc3745f2399c4b : SUCCESS in 9m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84280cd871d3447693d9577909ca430c : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8ec4863b86c4ab4ab35454217335227 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3d43ee9cca750328be085fb6f1ba18af060d975c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 17:54:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21bcf21f0fb0ec0bcdc6026a6536e1e37b3a171d","revisions":{"21bcf21f0fb0ec0bcdc6026a6536e1e37b3a171d":{"kind":"REWORK","_number":1,"created":"2024-11-25 15:25:42.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/68/936168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/68/936168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/68/936168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/68/936168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/68/936168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/68/936168/1"}}},"commit":{"parents":[{"commit":"30a7970e6dd7d7fd33ed7b5e5ba297396981fe62","subject":"[2024.1-only] Synchronise with latest kayobe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/30a7970e6dd7d7fd33ed7b5e5ba297396981fe62"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-25 09:13:41.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-11-25 15:25:42.000000000","tz":0},"subject":"Store ironic-inspector data in database","message":"Store ironic-inspector data in database\n\nWe\u0027re observing a lot of CI failures caused by DockerHub pull limits.\n\nChange-Id: I5d6489b092350078f1da4b8f740a097ea900b249\n(cherry picked from commit 7d423c45ae835b2167081a82dc0022318d22a3e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/21bcf21f0fb0ec0bcdc6026a6536e1e37b3a171d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/21bcf21f0fb0ec0bcdc6026a6536e1e37b3a171d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
