)]}'
{"id":"openstack%2Fkayobe-config-dev~937259","triplet_id":"openstack%2Fkayobe-config-dev~master~I9e3fe774107fd7fe2bdfeffa5e95dceb8d3d522d","project":"openstack/kayobe-config-dev","branch":"master","topic":"pin-tiny-ipa","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-10 07:12:37.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-12-10 09:30:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e3fe774107fd7fe2bdfeffa5e95dceb8d3d522d","subject":"Pin TinyIPA images to stable/2024.2","status":"MERGED","created":"2024-12-06 15:38:55.000000000","updated":"2025-01-25 08:38:48.000000000","submitted":"2024-12-10 09:30:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937259-pin-tiny-ipa","meta_rev_id":"0e19ddf6c85925cffbdafb9636fafe6a49629290","_number":937259,"virtual_id_number":937259,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 09:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-12-10 08:09:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-10 07:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-10 08:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2024-12-10 10:15:43.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-12-10 08:09:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-10 08:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2024-12-10 08:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 16:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 16:39:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-10 08:09:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-12-10 08:16:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-12-10 10:15:43.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"082e6eda9e3a603e5bb31e3cb22555c669c10593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-06 15:38:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87bd2576e11206a32b6c1ef383287a90735079f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 16:25:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17d6e5eb927a480cba78f54c2285eea7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a107e646c634baa8a82a0ce72dd8968 : SUCCESS in 2m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/963eac14d5b542608dbaebc27548e7a3 : FAILURE in 21m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/04a74348b2ad409a8c3158c3e625aa98 : FAILURE in 19m 58s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3927f7a0df0549ec9981eb4581cdb4fa : FAILURE in 26m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dab4042b13a441d4b2aceaa61fed182e : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2d888a93889b4451870ab7c83b5a2d05 : SUCCESS in 18m 35s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/be2e463dd8884790a4fa422950b1ae09 : SUCCESS in 20m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8da6bcefa4944eff97e0882cc1297967 : SUCCESS in 40m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/22fd3fe038e847e69c9f427f9666f288 : SUCCESS in 17m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8054c662d104a758ee0cd87a872b3d7 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5e1eb8b6218098692df38ffd113ca71d94512647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-09 20:39:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"635aa357a19cd3c343f01d55797b2a68201d062a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 22:08:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af440307a47b4f6198ba1e6a4aec5f0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e627972d30492b90ae5d92f06464d6 : SUCCESS in 1m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0f5a1dbb692440cea0c93d722d28524f : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/92f50a1c6ce44c5cbb02af907d11e4e4 : SUCCESS in 58m 30s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bc18f29c5e8d4ad4bf935fd681fb5880 : SUCCESS in 1h 20m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6db3dd87f0aa459d8e3bb5a1fe12bc39 : SUCCESS in 20m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4556fc477c3746c192de4490ea9df631 : SUCCESS in 20m 44s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3e8731b364a74094bf336577742749a7 : SUCCESS in 22m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6e56f0d6ae90483a875b9b620f3e7d40 : SUCCESS in 49m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8bdb151d519b4938a245124657a269e5 : SUCCESS in 17m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b86feb6ef52c47c3a2e077db2fb66e62 : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c25a0461342a8d482b41e0a80e2e48f506d6ef5e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-10 07:12:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"43e5e2b1d918679b06eb0a8acfed921ce061e82d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-10 08:07:28.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"874f93b9332fe719c7ba26bacd593e9b929f40b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-10 08:09:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8d1dfbbeb6934ce63f0f700b3ea9f482a067552c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 08:10:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f8356c5d80d4e427f0fadfb7add49cf2a4cb19a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-10 08:16:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f3e1fcbfe7badbde64cb6286d6df8db6d9a00d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:30:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b406c509f02c451686c61e4f94fccbf1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84f9874d65b43108c983d8b7da918d8 : SUCCESS in 2m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8bf324b30112482993cc7ae9666153b2 : SUCCESS in 1h 12m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b0e588e86004214826ba088b7f9221d : SUCCESS in 1h 14m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6bba4eee46ee4c138ab269d9021dfd93 : SUCCESS in 18m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e59f5702f4774248a87d88d546d8b903 : SUCCESS in 20m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/112b52fa3ee8471aa31455700975ac66 : SUCCESS in 16m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/67bf7929948545129c7fcb90b9706022 : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":2},{"id":"66ff909aedf300fcf5f14a7699549fa5baa2583f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:30:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"69f9ca43c3f25c5c639bd54c499cc8fec88e33be","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-12-10 10:15:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e19ddf6c85925cffbdafb9636fafe6a49629290","tag":"autogenerated:gerrit:revert","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-25 08:38:48.000000000","message":"Created a revert of this change as Ibef45fad608eecf76c398339a9c21a5cf7ba3b15","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2a9daaab9474db50219d27d8ba43e1ef00a279b","revisions":{"501fff88b9372dc6de255612ec055a6a904e536e":{"kind":"REWORK","_number":1,"created":"2024-12-06 15:38:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/937259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/59/937259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/1"}}},"commit":{"parents":[{"commit":"7d423c45ae835b2167081a82dc0022318d22a3e4","subject":"Store ironic-inspector data in database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/7d423c45ae835b2167081a82dc0022318d22a3e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-06 15:38:05.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-06 15:38:05.000000000","tz":60},"subject":"Pin TinyIPA images to stable/2024.2","message":"Pin TinyIPA images to stable/2024.2\n\nTinyIPA master branch images are failing with:\n\n    traps: ironic-python-a[495] trap invalid opcode ip:7efee73f58a2 sp:7ffcb6eaf7a8 error:0 in _rust.abi3.so[7efee7112000+631000]\n\nChange-Id: I9e3fe774107fd7fe2bdfeffa5e95dceb8d3d522d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/501fff88b9372dc6de255612ec055a6a904e536e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/501fff88b9372dc6de255612ec055a6a904e536e"}]},"branch":"refs/heads/master"},"f2a9daaab9474db50219d27d8ba43e1ef00a279b":{"kind":"REWORK","_number":2,"created":"2024-12-09 20:39:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/937259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/59/937259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/59/937259/2"}}},"commit":{"parents":[{"commit":"7d423c45ae835b2167081a82dc0022318d22a3e4","subject":"Store ironic-inspector data in database","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/7d423c45ae835b2167081a82dc0022318d22a3e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-09 20:39:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-09 20:39:17.000000000","tz":60},"subject":"Pin TinyIPA images to stable/2024.2","message":"Pin TinyIPA images to stable/2024.2\n\nTinyIPA master branch images are failing with:\n\n    traps: ironic-python-a[495] trap invalid opcode ip:7efee73f58a2 sp:7ffcb6eaf7a8 error:0 in _rust.abi3.so[7efee7112000+631000]\n\nChange-Id: I9e3fe774107fd7fe2bdfeffa5e95dceb8d3d522d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/f2a9daaab9474db50219d27d8ba43e1ef00a279b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/f2a9daaab9474db50219d27d8ba43e1ef00a279b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
