)]}'
{"id":"openstack%2Fkayobe-config-dev~955454","triplet_id":"openstack%2Fkayobe-config-dev~master~Iafe63667512ffc60514538dad21bd3980faf453e","project":"openstack/kayobe-config-dev","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-08-20 13:44:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-20 13:44:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-20 15:55:16.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-20 15:55:16.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-08-20 15:55:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafe63667512ffc60514538dad21bd3980faf453e","subject":"Synchronise configuration with master","status":"MERGED","created":"2025-07-21 07:55:27.000000000","updated":"2025-08-20 15:55:16.000000000","submitted":"2025-08-20 15:55:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955454","meta_rev_id":"06f2b2a66fafdecf1e1a09152af7257dbf4e5a3e","_number":955454,"virtual_id_number":955454,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 15:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-08-20 13:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-08-20 13:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-08-20 13:44:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-21 09:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-21 11:30:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-21 12:16:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-07-21 12:16:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-08-11 15:17:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"c4f8e721f4e1c3ac200419f109073ada7382650b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-21 07:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4db900f22f51c18f74666207449f5dbdbc8ae8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:35:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c25c9d54a1c9468db4fc07846b571fdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac1b217e7904b8ca64c8f48b2dd9bed : SUCCESS in 2m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c75346d18c6e4f72bd6c1b8f515736ef : FAILURE in 24m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/36ce936771d34678be586c85dc828d79 : FAILURE in 18m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a8e98cc51340aea7ee1b8c48d82bed : SUCCESS in 57m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3172668782e4453b89a37b1f74dce1d5 : FAILURE in 33m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/45daec4807cc417da2fc96ce7276b161 : SUCCESS in 12m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/81a2a1c08f0d4ca2867cd890d84384c1 : SUCCESS in 18m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/992dbe12ffbc45f18c3c18eb17644d31 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7a2cc3e822e143f586efd823e975d5a4 : FAILURE in 1h 32m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c4a54069b9ec4d54876e943a696e94f6 : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da2cbd3a39714994a732da5cc94a063a : SUCCESS in 15m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dc7d1760a7e641e589cfe35fe5ba4636 : SUCCESS in 42m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c3f59b448e0a4bef9adac7d09fa6743d : SUCCESS in 30m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/165013fd0fef454a96a6e15d0075614b : SUCCESS in 11m 56s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/300102c1582b4c5faa4d28ea6d3e9f0f : SUCCESS in 9m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d662e4a67849629479efa87f571f9d : SUCCESS in 8m 11s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d67789ed7049446392e8f38709b89904 : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc259e3aba714fcd9de44519a4f22353 : SUCCESS in 13m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaa6e797ecc44e32be2066a08d87e0da : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c24216088ea1a6739009db9f25197505a3e98d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-21 10:24:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"29aa104168f874cdce03d94f112b7dd2952784b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 11:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c5f7711a37549f588c6b969bb173ecd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca56806133ff42c7a5366a737161e16e : SUCCESS in 2m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1c12038ed5d419399b70d79bdd440c2 : SUCCESS in 52m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2cb421652fb04d6782ee5eb00500a059 : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e090bf6e27364045b6c4514b26f08f1b : SUCCESS in 16m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86ab09912ddf4fe28b2b93838bcd9c2f : SUCCESS in 12m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3cdee12486a94f55b0d84a92d01dd9f1 : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/346f712e5dd9472481303b5637b3336b : SUCCESS in 15m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/573198c2424e48f69e53a8f34a2b7070 : SUCCESS in 35m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a2c7879cc474d95ab130a91b9474b58 : SUCCESS in 26m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/85091d882f1a468898332e3a7601cf16 : SUCCESS in 12m 14s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/83c52dc9616e427f929ccdaeef6d342f : SUCCESS in 12m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e2e8a3b7b0840dfbe2ee10b7469a80c : SUCCESS in 10m 25s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/077b3db3d7b14b7ba57c837a361bd653 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82f07628c7054d29b3498fc660cd005d : SUCCESS in 15m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3457467a55f04290a68197fe6b7cf029 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3f11b78231016e363035ad65a156445b9eb3de7b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-21 11:30:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b8d3ddd277c09b6b1a6cbf2fe9d05f3dc678fe1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-11 15:17:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44bae5fcf313d9ff3ec10bfbb5adac00db8165d4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-20 13:41:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af36686782f6f2f564f06a68e9d1becf065055ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:41:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40e72dd028359306a7d7f5193dd4d76d5e9e8257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-20 13:42:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8e0e71e892262b6bd14228239ba40b57be73184a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-20 13:44:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"50d1635d555edcbc5df187d9b75e9cc5b4d27ae0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-20 13:44:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"39d0c24e02eb374d47bd6a86cc9dbadf3ff42464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 14:47:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc8456c3ccaf457182fbc102036d3aa1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5760ce4b2f614e06bca2ed57ede8a77a : SUCCESS in 3m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62092d9fcbdb4ebdb591c0cf7161c373 : SUCCESS in 35m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21c57ad143594b85aea7e10c1ca2405d : SUCCESS in 30m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a6135d39972448293d2bad5f9d64764 : SUCCESS in 22m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/497725e988944bbfb97780f0a123a9b1 : SUCCESS in 9m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33c767e0f5754c3dad9022451ee4725d : SUCCESS in 1h 03m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e4064e33fc2340d2adf1be5a4163727c : SUCCESS in 8m 39s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d7c0d8b1f46475ab1a051800f4d3900 : SUCCESS in 14m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abc1c35f0749408994e90c38293fcf54 : SUCCESS in 13m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9841746a150412fac3b154d6cfa9d73 : SUCCESS in 11m 33s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a9e84d23c7eb4367a0d376828fe91590 : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25229d8f5c2f4828bee09de37fbdf82b : SUCCESS in 29m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/611fd9ab161d4484a64d95fca6f5ce75 : SUCCESS in 9m 03s (non-voting)\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/972de7c77bea4c34891bd5cf280f9214 : SUCCESS in 11m 18s (non-voting)\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c43f419a3a0a4c858cf967e3485b78f6 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/27b43dbce766452d820b762b23753916 : SUCCESS in 14m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/65a18b14d86c46e39a4056863b27c0b8 : SUCCESS in 13m 20s (non-voting)\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/127492931a3a4183a8e09233d50d3a3a : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":3},{"id":"68e59b64b21349e76ae7832d6e2037f839df26eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 14:47:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e08f861ef3478652e0b5e2506d46b94ac3955ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 15:55:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f674008f48749939af9b4a91f94927e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1e9893d9c24b439a0414d387a5a0db : SUCCESS in 1m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/824db728a3984304a9c064098926c539 : SUCCESS in 1h 05m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6c46c8c9d7e488db289d823705dc33a : SUCCESS in 31m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/940632371aae4796860c250e4ff9ce76 : SUCCESS in 21m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d21b86651d6a462b9e2c41890c2352a0 : SUCCESS in 9m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef9a08e4c24428ca462de3ade36726d : SUCCESS in 56m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b6175fe82297470d9ad6adf5aa9c22f8 : SUCCESS in 10m 35s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b90435ab64e473cb2a0668462cdd86b : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cffcd4254ed14516a993ba52d871e23e : SUCCESS in 11m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e629d71b2cd4e3095db4ad75d61df34 : SUCCESS in 15m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/36446f79c89041e7b592978772287b48 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87e1a105b84f496d9876ffde5acc9427 : SUCCESS in 22m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41083a45f97f417fbef78c74e8651725 : SUCCESS in 9m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/639644a46ab245da885cd418c3697e34 : SUCCESS in 8m 50s","accounts_in_message":[],"_revision_number":3},{"id":"06f2b2a66fafdecf1e1a09152af7257dbf4e5a3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 15:55:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"faa8fccbfee20162a72dddb28cdad0c6104ba8b2","revisions":{"feecbf2d5e37bb85f8bd1f7673f4b88d5f462bce":{"kind":"REWORK","_number":1,"created":"2025-07-21 07:55:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/54/955454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/54/955454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/1"}}},"commit":{"parents":[{"commit":"724af9bed30402311923e8c5dcbfca349150711f","subject":"[release] Synchronise with kayobe for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/724af9bed30402311923e8c5dcbfca349150711f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-21 07:54:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-21 07:54:12.000000000","tz":120},"subject":"Synchronise configuration with master","message":"Synchronise configuration with master\n\nSynchronised with kayobe @ 4d24fe7e97e6fb02185c0a2f1e44896718c4ac6c.\n\nChange-Id: Iafe63667512ffc60514538dad21bd3980faf453e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/feecbf2d5e37bb85f8bd1f7673f4b88d5f462bce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/feecbf2d5e37bb85f8bd1f7673f4b88d5f462bce"}]},"branch":"refs/heads/master"},"e4e690ae531a394a6a273127f7516b6315c17fd1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-21 10:24:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/54/955454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/54/955454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/2"}}},"commit":{"parents":[{"commit":"88560a1937c3ef2527fa93b972b063a435a59792","subject":"CI: Remove failing CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/88560a1937c3ef2527fa93b972b063a435a59792"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-21 10:23:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-07-21 10:23:54.000000000","tz":120},"subject":"Synchronise configuration with master","message":"Synchronise configuration with master\n\nSynchronised with kayobe @ 4d24fe7e97e6fb02185c0a2f1e44896718c4ac6c.\n\nChange-Id: Iafe63667512ffc60514538dad21bd3980faf453e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/e4e690ae531a394a6a273127f7516b6315c17fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/e4e690ae531a394a6a273127f7516b6315c17fd1"}]},"branch":"refs/heads/master"},"faa8fccbfee20162a72dddb28cdad0c6104ba8b2":{"kind":"REWORK","_number":3,"created":"2025-08-20 13:42:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/54/955454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/54/955454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/54/955454/3"}}},"commit":{"parents":[{"commit":"88560a1937c3ef2527fa93b972b063a435a59792","subject":"CI: Remove failing CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/88560a1937c3ef2527fa93b972b063a435a59792"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-20 13:41:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-20 13:41:35.000000000","tz":120},"subject":"Synchronise configuration with master","message":"Synchronise configuration with master\n\nSynchronised with kayobe @ 4d24fe7e97e6fb02185c0a2f1e44896718c4ac6c.\n\nAlso synchronise CI jobs with most recent kayobe.\n\nChange-Id: Iafe63667512ffc60514538dad21bd3980faf453e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/faa8fccbfee20162a72dddb28cdad0c6104ba8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/faa8fccbfee20162a72dddb28cdad0c6104ba8b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
