)]}'
{"id":"openstack%2Fkayobe-config-dev~967206","triplet_id":"openstack%2Fkayobe-config-dev~master~I5f24d73248bbf21f3e856aea6dd1f6c6d3ce8735","project":"openstack/kayobe-config-dev","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-14 13:57:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I5f24d73248bbf21f3e856aea6dd1f6c6d3ce8735","subject":"[release] Synchronise with kayobe for 2025.2","status":"MERGED","created":"2025-11-14 12:18:07.000000000","updated":"2025-11-14 16:37:59.000000000","submitted":"2025-11-14 16:37:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967206","meta_rev_id":"4ba8c78d56d644e9259a195075e77919762bc5da","_number":967206,"virtual_id_number":967206,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-14 16:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-14 12:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-14 12:37:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-14 12:28:13.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-14 12:21:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 12:21:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-14 12:28:13.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-11-14 12:37:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-14 13:57:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fdd59ea6c740bea5e191843fe6b7f115181cdb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-14 12:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca4b88475a12225c1d3ee0b4b7081a1d1a8e47c7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-14 12:21:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71b7ed014672f1f0614c068e5c09d375f920460d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-11-14 12:28:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"777027c6b2bed9b1016103f17f67067b1527d620","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-14 12:37:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"060509c8a6964f21ccff2e7b1b6b840c94e6680d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 13:57:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da6ae6f604ac4d7fbca86c481da95753\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c1d535b81f4be0a3b01f0dcf3a45aa : SUCCESS in 2m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d9f995104bfa493d8600a3a72d52d9d4 : SUCCESS in 12m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b789eea1dc7473fbf26f6ecf8717ba9 : SUCCESS in 12m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/34d602e822ad48e7a61ecf62c4efe10e : SUCCESS in 24m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcf98b5c7c9f4ba59fdd287c29e6d495 : SUCCESS in 14m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2310e60d35de4ae19e4f87867398490d : SUCCESS in 47m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fcdfa52a4dc44003be1323f736db7501 : FAILURE in 29m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8b4e491f975d4038aa1fd0019ff084ba : SUCCESS in 37m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa4af9af53e943118bda9044d4b99f08 : SUCCESS in 55m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d061c1de2f894f379eac90192bf7abdc : SUCCESS in 43m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddc9210782724ae88f57d0b171ec0661 : SUCCESS in 1h 34m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/72163d9d57b744f9b4e0222895793ec6 : SUCCESS in 8m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e046bbf571a4df5aac0310bceeac964 : SUCCESS in 8m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e529e22e10294366bad2f333984155f7 : SUCCESS in 15m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bfd73559bcdd47f6bb806e41c43a982c : SUCCESS in 8m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c5b5c2045794f5baceb77867350ecba : SUCCESS in 25m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/549fc6a776da4b279b313d3993f14b5e : SUCCESS in 10m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f527967a0bbf4e6db8a65042b472cd66 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3063e43a46fcc5c89feb74957dbf5099937b637f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-14 13:57:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2f37f2ef4933ded6bd84ceeced3eb8778b000df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 15:24:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/773e0a0fa9f54102b63baf330eaccfef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6941ea37a7844bf8206e33bca70e8b6 : SUCCESS in 2m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6babbc7032564e9bbe3e8e7a4f720eb8 : SUCCESS in 9m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/beea4170ca3544e9887e1d9eeea32e9b : SUCCESS in 12m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/81f9d21483a1405584ee8e7318de2495 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/436bfa4fea9a45b2a6821881ff39a386 : SUCCESS in 15m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/602a8c9fbeed4c128c64e5a0b64d37d7 : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/913fe90b842e48348f1e8189c738998a : SUCCESS in 57m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/23081f623deb47a3837eb8c55ef85cef : SUCCESS in 54m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92f0a0ca3ce14089b28a48ff890b863d : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b912cbc85273423796873a3a58b2f69e : SUCCESS in 1h 13m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aa5aa74c1f04b4c830a5f57820afcb6 : SUCCESS in 1h 25m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f5e0b56764274290b05dcbdc5257c224 : SUCCESS in 15m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bae1072f749a43a0a324ecf181427aa4 : SUCCESS in 21m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d89cf005826442ff9ee18fb5734854b9 : SUCCESS in 16m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de281b1da38c4aa09cb72c47047bc137 : SUCCESS in 12m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ea777564b2c490a95863643d1fc3a6a : SUCCESS in 31m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a0f71db15f74bd1bdf05caaba9a3a60 : SUCCESS in 10m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b8d46820bed422891c94e1d4d429027 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"68d3169426fcde44d1a9359b9c1516a39bf0e315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 15:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a2b236ffb2c0385b7630cd4cd88b2f7f395bc58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 16:37:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdedd589e2c341969cf72cc0de33e3db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c8b5b52967405da22845aceb6e08d7 : SUCCESS in 2m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9eb28ad45ed4253be5496a82762953e : SUCCESS in 18m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f2d8f549e50416f9d9d73cade30ba20 : SUCCESS in 14m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dabad086426b4267beba5d790650ce0b : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20827ca134d04a18adafef705d144eaa : SUCCESS in 15m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5de2ba25c4e849739b50a8c53185fda5 : SUCCESS in 1h 11m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f194c0ea575f41528f99b9a0f2dd878c : SUCCESS in 35m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4d1e6a0f47094c5697e5874c5c9d6a77 : SUCCESS in 56m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a025f1d1ccc42a589adcbaeb20fea0c : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bbe660dfd5564d9bac98b4a860cf8fe6 : SUCCESS in 48m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a18b5d337aaf4fe280168b458d70ca96 : SUCCESS in 53m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/855c19b651734b3fa9b86dfd579385c7 : SUCCESS in 9m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8f35d6ad6fdf49199ea9a1b2736c153d : SUCCESS in 13m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8b6255f7ecd4d2a91aed1bd7e981b66 : SUCCESS in 8m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5fa8f4dfbc4348cba945a42094894820 : SUCCESS in 7m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8697da1fed9c44e59107cb310b6532db : SUCCESS in 14m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3b7952986253425485fabfe666792a20 : SUCCESS in 17m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3767e4bc00843c49b20654726776ada : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4ba8c78d56d644e9259a195075e77919762bc5da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 16:37:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41bbc2119c9debf05f669a9c91cfd14468ecf525","revisions":{"41bbc2119c9debf05f669a9c91cfd14468ecf525":{"kind":"REWORK","_number":1,"created":"2025-11-14 12:18:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/967206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/06/967206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/06/967206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/06/967206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/06/967206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/06/967206/1"}}},"commit":{"parents":[{"commit":"c24a5195e63f75dce43905ae6607bebeeba4b0c0","subject":"CI: Use 2025.1 TinyIPA images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/c24a5195e63f75dce43905ae6607bebeeba4b0c0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-14 12:17:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-14 12:17:00.000000000","tz":60},"subject":"[release] Synchronise with kayobe for 2025.2","message":"[release] Synchronise with kayobe for 2025.2\n\nChange-Id: I5f24d73248bbf21f3e856aea6dd1f6c6d3ce8735\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/41bbc2119c9debf05f669a9c91cfd14468ecf525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/41bbc2119c9debf05f669a9c91cfd14468ecf525"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
