)]}'
{"id":"openstack%2Fkayobe-config-dev~986140","triplet_id":"openstack%2Fkayobe-config-dev~master~Idb827125dd1dc26f0b9f1287a131cf32f5e8beec","project":"openstack/kayobe-config-dev","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-20 15:39:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb827125dd1dc26f0b9f1287a131cf32f5e8beec","subject":"Adds aio-network-connectivity check environment","status":"MERGED","created":"2026-04-24 16:43:05.000000000","updated":"2026-05-20 15:39:31.000000000","submitted":"2026-05-20 15:39:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986140","meta_rev_id":"c200a7b7f74748a0af90abe6c626683fa9eaf97d","_number":986140,"virtual_id_number":986140,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 15:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-20 07:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-20 07:36:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 07:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-05-20 07:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-20 07:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 18:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 08:28:34.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-05-15 22:07:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-20 07:37:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"d917e4d60e6ca07c651a77f4dd1a56ea7774bde2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-24 16:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2766b3f0ba758862b277acebfe9a9d245f06c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 18:24:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9efb0235f4e044e486810f7244200cd8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f12102d76914667a9e819dec38a5a8c : SUCCESS in 3m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9a7516de64e445ada50e268fe0513270 : SUCCESS in 24m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0883d1f1f7964bcda68e8c9d11395261 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/689c0867a82a4b66908cce9f9622a69f : SUCCESS in 24m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb042875741c423c8d7a9a82e6848641 : SUCCESS in 18m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cd8c651ca6d94e4ab7aaf406152dbad5 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e459e73513448de895639703621f94d : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/04669b36ab6c4628bdd29d9cbd27bd08 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/137d7e5d1a964bd99b3c3fc443821291 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b17f6f631b1404b8e8ac9c295ac6cac : SUCCESS in 1h 10m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a59288dab4ab4b72977cb021a6bde4a7 : SUCCESS in 1h 36m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6925a1fdce948af93704103641fab67 : SUCCESS in 1h 40m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/492419b37b6047739666c371ec28e9c5 : SUCCESS in 29m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/be9165af22024a36a248ef3d73b5a053 : SUCCESS in 27m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c5c656c51da4c0aa803929e2f107858 : SUCCESS in 24m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2c0dc4318ca64f349971d11fea892258 : SUCCESS in 20m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/faa9a9666f2546b890fcacd82db14041 : SUCCESS in 34m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c94c4b408da45a98fab3bd284537151 : SUCCESS in 29m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b28dd0b6cfc049efaad5dccd3a49e0bb : SUCCESS in 27m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28de701f21cd4822b645f6a311496935 : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b0ec022256fb1806099367b1852d4e75169903ba","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 08:28:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9c07d3d75eebff22f30b5d98e39db456fdeae33","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-15 22:07:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ad8f451ec08ee211fd25cf726805cba0d204166","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 10:22:10.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"196f792f8f247406eb1db9e3c9fc66bf91762e96","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 13:14:23.000000000","message":"Patch Set 1: -Backport-Candidate Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abee6ccc8eef9a10cec2f04e542f364f9af3107b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 13:14:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03e8c2fcb14206caf010fec6aeaa45de3c685b3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 14:50:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72dab72864744ace859d19134c493c1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a4604cb2a74419ca0caec8dcf8a50f3 : SUCCESS in 1m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4974fd2998f8454cbb80e0bd442557fc : SUCCESS in 17m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/407380b93b9647cd9cb1d1a8985c70b1 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b63877c2480e473084e160d6db8f71f9 : SUCCESS in 11m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec2d17f047574400a39f5b10233e539e : SUCCESS in 21m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/01aa5e5ac72046cdbbf6f1470566d491 : FAILURE in 27m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/60e3c2136786418ca255996ab128a6bc : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3d9bab72430748c0897f84618dc75ff6 : FAILURE in 38m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b8505ad79994c02991755e256d39a82 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34fd2e167bc64129a3e7d6ae51cfeb7f : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/43b97abc919d45f4bfc1e399218b470c : FAILURE in 53m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ecf07a53079483485d60b94e80ae02e : SUCCESS in 1h 34m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5f78735b6c144ab0b28e992642e5cb9a : FAILURE in 20m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/27fe93f6d5054f48bb25a8526b1547ce : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/996ed858650f44b68236c468ed8b6554 : SUCCESS in 21m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59228fbeb04c48b19c9512bab8d77bb5 : SUCCESS in 23m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cbec9a2e6201472a80c9a02efbe28fdf : FAILURE in 15m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/615f91bc13db49798cdc53fbaedfff6f : SUCCESS in 34m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a64050adaca84fc894f5b363356716a6 : SUCCESS in 27m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57be6c8ff60746279e7368cfc67383cd : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"947944f7949bdae34e11665ccd8f542d547b9828","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-19 21:30:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e193c09bed7d1007fbf11a24cbac7df9d820a931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 23:08:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b737fd3f4ad943ea8e9b6dc1bee8d0c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5725f53963a47a09f2637e90ade06e7 : SUCCESS in 3m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8680cc9ce06342a7a2266ea640baf807 : SUCCESS in 13m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56af72f497d246618d0010027a6fe4ae : SUCCESS in 23m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/194ac643aa914fc38dcd242089ee99f0 : SUCCESS in 24m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52926b2e1b314f938ba43cde74a33b00 : SUCCESS in 23m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2904f847eed54ff5abf4fb47bb9b50bb : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a5d3d7139ef4bd794ea7c6d8e515bf5 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/96541909879f4d229ac351738b04abe8 : FAILURE in 1h 04m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e69d0e937b424054b91930f55a79fa8a : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0d534a34da84d788d42b69bf1c696c3 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e5a6cc14f53346048e9616bcf6294fb3 : SUCCESS in 1h 36m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6d1179b74334fab885b431cb5577e45 : SUCCESS in 1h 35m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/57cb4e94901941b882974c069ad82457 : SUCCESS in 19m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/81ff6388f13b47cf9655528b42045329 : SUCCESS in 28m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04c4355a339643edaced27f225e95199 : SUCCESS in 20m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6490a35112ca42f29782762eff26d25c : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1002379fc0a94837a1ffed8bf3e6a5c5 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e8574982018488ba880df398b45b7e4 : SUCCESS in 15m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/14191b88ac344b7eb8ca37b3ffbb6022 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06118fdfac134f97882f4a5e167962a9 : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"839bcef1fde04b7f21bf072073d3dd329f5b9a87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 07:36:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":2},{"id":"a0f4a69156147b22e32f80482ff1789e1ebfc850","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 07:37:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1779993d5387752a393b7892f51c7549e91ea775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:13:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb7c47c569a04d1295c2b34a2618cc10\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875a51628c154e7db5d383332fd2f9c2 : SUCCESS in 2m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/73c7307c68df4776a85c03eb77ccb1a1 : SUCCESS in 14m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/696baa10f27d4599876355a03c905850 : SUCCESS in 18m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/403578ee638a42148082362b9b9ba5f3 : SUCCESS in 11m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/754931ca2b8f4df492e58650960008d7 : SUCCESS in 20m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d3a5985dd2234634b469426d67a011e7 : SUCCESS in 1h 04m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0cde5720351d4f6f99765baea942269b : SUCCESS in 59m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e34be5e4c00c4919b47f2e744d1a669a : FAILURE in 1h 04m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9443e82507e47798c54b995d200512a : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccb6d2af06e5463d832baf4ee5b079ca : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/28e9b6c159de4c17adc727355141ea36 : SUCCESS in 1h 35m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3326d56cb71d4d6f9e432a68cc11af1b : SUCCESS in 1h 32m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e7f4e9f8ef8249c58123b6886a987a52 : SUCCESS in 28m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/622be6770a8c432e9e6d81408ed2a9bd : SUCCESS in 14m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6eb96d084e5740589b4fd227fe8ff796 : SUCCESS in 22m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e325240011164eb696e664c1a32b66c5 : SUCCESS in 20m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6f77d17bc8884ce48616f42faefad0c5 : SUCCESS in 22m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc5140f272914d07a008a1805df77501 : SUCCESS in 31m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/da7680ce8c3344c99cedb00c809f9915 : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9df76ae1ed4a4a7b954b86bfaf29eee5 : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e259dd9be071eee7c6fcaa36b02d2f19bafb807c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 09:27:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c1da88a4477e7e1bf8eb6d11600adb854cc7aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 11:11:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fff0abeb091944dc9bf644759ec28353\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af670f364ae84e5f8723acd46a7e9731 : SUCCESS in 1m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1247974316db4ad2874a586b6f466b3c : SUCCESS in 20m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/147ae7d1a3244c82a6addb127506c3b4 : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1c9f8245c75145ca9ad79c43a90d0648 : SUCCESS in 12m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c556b22aa3214020aa660d060dbd117f : SUCCESS in 18m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/27aed0625b4242e4b8beab871fce10e0 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/85c5c7dffde74cb7a4bcfd89144876c2 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/30731aaa814d43e18beaf9cdb836aab5 : FAILURE in 1h 18m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/602ddaf439af4fb58163da14c53261e4 : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d89fe85b179744bea03794c176bacf0b : SUCCESS in 1h 08m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0ae18e1c606a4786b21577486b2b72f1 : SUCCESS in 1h 42m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dc9052771e940adb5e60bd2aa83e281 : SUCCESS in 1h 42m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0128655bf97343bd8becc13a02efda9b : SUCCESS in 15m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d113192607c4284829f75e0cd2920a1 : SUCCESS in 24m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fcb4a4624ba438ba164b08bb6e1e0ad : SUCCESS in 22m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b1d16450a0749e48c47bd4cc2436071 : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/65ac798e4a1540e4928befd1c408ca2d : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b83c6057274a4ab3956cafabe96ac7d6 : SUCCESS in 15m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/354f8fb772394a62b0e4d1a822e3fa6a : SUCCESS in 15m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8df9b21a604a4650bc8c7c812e52a685 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4523b6a70d57d5d8b6a8aa003a32d155229dff39","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 12:20:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4d5f0c2fe015dadd7317f81eec948746500c5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 13:59:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbb2bbc734384568bcf032c156d8703f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42732667022f4bdb8995b4522290be79 : SUCCESS in 2m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4b675d0f19ba47db840aa2abe3116d46 : SUCCESS in 14m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c10a8a8470646b3bf8510529e4abb5f : SUCCESS in 20m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da53b60ec19d47e8aeee75547623a012 : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/336df61932f545d6bdf29cb4f1547a72 : SUCCESS in 16m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dd0123c284a8428fa3606bf03b908bf4 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0ef09b15be9c40f696dbdf04b5a67379 : SUCCESS in 1h 04m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a86fd229f7a341b48b79802cd002a75f : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a9399bfa1af4a058da1881b8534acd5 : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b13639f44b0f41e78205f9f05e916e62 : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/53284e8b415d4908aae437e0f7b367f7 : SUCCESS in 1h 38m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd3469ca9edd471f92890efd93151147 : SUCCESS in 1h 34m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/380704b76eee4f669828010e1699971a : SUCCESS in 14m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/05902ae35e8045338854cad6ae56e38f : SUCCESS in 15m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52f41325d3a14353beac642c4ba25582 : SUCCESS in 13m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f9754bba40a491aadf1d6719512a974 : SUCCESS in 12m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a8e8a8daaf324370905888edb0e5ea0d : SUCCESS in 16m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c517711ebce34609a39a20862a9f6df5 : SUCCESS in 31m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a46072c375e74bca93e5151d992944e7 : SUCCESS in 13m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecf2a3764a074541a5cc780bacbeeae4 : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7fb9563ce5cdeae29ab59dc62ac0c24f28f7ee40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 14:00:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b1aad0f03207864f49f88b736d82a142399f2af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:39:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc9576cc2c1a4f07a7c5359ad40715a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c0bf8e40f2442eb6b17b9c4c2fa298 : SUCCESS in 2m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/27ac981d975644d8b61483ad2a505b9b : SUCCESS in 20m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0671df4a6706441a8e80df8e3bc8838c : SUCCESS in 15m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1dcfdf3758c14b1abd2b5562a01355a1 : SUCCESS in 13m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cf4ca42927447a18ee6b44534b9fb1d : SUCCESS in 10m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ef05037b31f34659ae8184eaa58a0ab9 : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2041ffa644444d129609e607f4d79521 : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/94983b5a3f444e8383c10c49ea0f3a62 : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c532b719289b48fca691262002ac7a9a : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43f06422f1784de4a5863646513f2b8b : SUCCESS in 1h 04m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dab78a6c9e714bf4acdb29e47c49ed17 : SUCCESS in 1h 37m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/394efe32cc6846609114e22212696b25 : SUCCESS in 1h 32m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/19b80161179242adb1cbda8cf3cc3c2f : SUCCESS in 18m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/841b06510215427e8fbd39d2b900389c : SUCCESS in 18m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e090cc5896214c188481ca38ae09542d : SUCCESS in 14m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4b6dabc52074187aa1c213e68cf1f52 : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/73b9b9f3d0a447e6b86390c298223b53 : SUCCESS in 37m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45d04e51ac314208b9b71aed4d955ba7 : SUCCESS in 17m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cf15e1a10bd24c1c9be964d18a64876e : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b6a0394a4044d1b8e42c15ab87ace8 : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c200a7b7f74748a0af90abe6c626683fa9eaf97d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:39:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbf45b2ba6db1436056176f4f2497babaaf1744f","revisions":{"955c17f90a98af2c4e532577fe75b3e64041fdf1":{"kind":"REWORK","_number":1,"created":"2026-04-24 16:43:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/986140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/40/986140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/1"}}},"commit":{"parents":[{"commit":"b20eba7f25d483ccdd355735fe8af3604d6952f9","subject":"Synchronise configuration with master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/b20eba7f25d483ccdd355735fe8af3604d6952f9"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:39:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:42:45.000000000","tz":0},"subject":"Adds aio-network-connectivity check environment","message":"Adds aio-network-connectivity check environment\n\nAllows testing network-connectivity with an AIO.\n\nChange-Id: Idb827125dd1dc26f0b9f1287a131cf32f5e8beec\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/955c17f90a98af2c4e532577fe75b3e64041fdf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/955c17f90a98af2c4e532577fe75b3e64041fdf1"}]},"branch":"refs/heads/master"},"bbf45b2ba6db1436056176f4f2497babaaf1744f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-20 07:36:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/986140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe-config-dev","ref":"refs/changes/40/986140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe-config-dev refs/changes/40/986140/2"}}},"commit":{"parents":[{"commit":"5ab35589fcfc22671d22969c3fa2bdaa4e025776","subject":"Merge \"Remove outdated comment in setup.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/5ab35589fcfc22671d22969c3fa2bdaa4e025776"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:39:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-20 07:36:37.000000000","tz":0},"subject":"Adds aio-network-connectivity check environment","message":"Adds aio-network-connectivity check environment\n\nAllows testing network-connectivity with an AIO.\n\nChange-Id: Idb827125dd1dc26f0b9f1287a131cf32f5e8beec\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/bbf45b2ba6db1436056176f4f2497babaaf1744f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe-config-dev/commit/bbf45b2ba6db1436056176f4f2497babaaf1744f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"955c17f90a98af2c4e532577fe75b3e64041fdf1","theirs":"5ab35589fcfc22671d22969c3fa2bdaa4e025776","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
