)]}'
{"id":"openstack%2Fkayobe~1001650","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I05d3f386157b887d5f550104fef7c5e7ffeed923","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-21 09:07:45.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I05d3f386157b887d5f550104fef7c5e7ffeed923","subject":"Set MTU on nmstate bridge ports","status":"MERGED","created":"2026-08-20 12:06:18.000000000","updated":"2026-08-21 11:09:55.000000000","submitted":"2026-08-21 11:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001650","cherry_pick_of_change":1001480,"cherry_pick_of_patch_set":1,"meta_rev_id":"f2d5e25ad2383f7885b0b9b91128c1f467456713","_number":1001650,"virtual_id_number":1001650,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-21 11:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-08-20 15:18:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-08-21 09:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-08-21 09:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-20 14:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-20 15:18:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"956440cad462d3654c41b34ac838954268d451b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-20 12:06:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0b7c020db362aa8a192e212ea38751f11c12cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:09:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72f9110478f84246b33af4738cd24f17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199e72857cf94803ac13fbf81dfebad7 : SUCCESS in 9m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c8312e33174f48fabb28291192d03f5e : SUCCESS in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5348425d4bff4e0a9a7fb9a5c7d85ea3 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf2c56f2fe54b5e9dfc52bd7a40bab5 : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eb4c3533dc84f01b7e758f49e4fa69d : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78b94515d8b04653a33d82e99b8d4675 : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a722d1f6979c44099130828fd81c66ba : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dd3336502ee472fbe2167634e17defc : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9a382ffda644468847525e82a726d2d : SUCCESS in 15m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be5440c23f62456f8b7359e243008e29 : SUCCESS in 7m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a67088427884226a88544c9e59defc2 : SUCCESS in 16m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25bf6e79fd9d48cd9329c346312f7ce2 : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eed70d4040e0452ea539f3784f37a6a1 : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fabf692c3554da4b6c3c6e33fd2d423 : SUCCESS in 21m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8fa8e4d561604a9cbd90cb15bd39819f : SUCCESS in 1h 09m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/08d4a461972849a3b8646e701780e946 : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1b4dcc785434419084bee71e31569280 : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47cf480b858e4468b55d6cf7296e1c05 : SUCCESS in 1h 26m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c59899a4c1434160bff1af8820c76809 : SUCCESS in 1h 21m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de458bebbf9d4fa9831e19be9e8e9136 : SUCCESS in 1h 36m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a062c6eee4304e22be33e687ddb63c47 : SUCCESS in 1h 42m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4b21d84035004e0b932a4bcba0eaa4b9 : SUCCESS in 20m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fedfec856ed141a58b3b4a1f98bc1d15 : SUCCESS in 19m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6885b4e8043490b975d5538f0c154a5 : SUCCESS in 21m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bec3721f85cf4014a76148fcd7481d41 : SUCCESS in 19m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f15786140cb64c1daf9e01e9e9f88788 : SUCCESS in 22m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1dd584dd32943c58c08e148e49839cb : SUCCESS in 33m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/510fa955c4624a0e842bed7e62a78e6b : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62415ef299624ddd92b4892e8208837a : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"94432d2f548c02bd543ed0cb038997aeadbe6332","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-20 15:18:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa8973e978212873a03e66311df760c7408e3ec8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-21 09:07:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a7748cb35b4f8869412304225a87bd012461361","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 09:08:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6211684b2f2258c4e8115b804daa1a5ac04e76fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 11:08:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d7fe7215c44a50a76fbfad199a4ae0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d90e53c4014ba0aedb763012c2b927 : SUCCESS in 9m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5d3678e178e843c6af5179d9a4e84375 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c50d564ee27411da7d09b1e0b7bf47c : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a89cda3145f4494840642ec1529ae72 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf0565f7c45c4894b184fc88ca4f055e : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d7106d3e43f41fbbea53d182f6233aa : SUCCESS in 7m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4dfc329326ac447da12c00d654b766ed : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e184c1249b7a48f49fa6d857b4fb14b3 : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bfba1374f4234beb9fa517f9df29ab42 : SUCCESS in 20m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2e04cfe39cd4e0297a4b202e616f60c : SUCCESS in 14m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e6b82a603b2c4c99a9d95b081bbe0008 : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47b20f2cbc3a40799380e765262c9535 : SUCCESS in 18m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/99aaa7ea3609461fb38f5b7d25bef520 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b742a48dccf4e0b9822b36ef17267b7 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/605fd205d95e4a22b7982614748b9d90 : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a844f978ac6451e966b9393e72b253f : SUCCESS in 1h 15m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e84996e6562644cbb484c3d81813d465 : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/794819ac9729477285940c0c7b9acc2a : SUCCESS in 1h 30m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90f322e91f674e1082954cfa703e38ea : SUCCESS in 1h 38m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d01d5b43e4544d4782175c8b162316f3 : SUCCESS in 19m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f942737f6c34c2c8c55d61b51abfb75 : SUCCESS in 23m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91eca4b1129b43388e78ccb933e9d857 : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ad9af57dda44b0295c0caf9e9f562a4 : SUCCESS in 19m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/28cfb22e8fe34567a0ad3e14f1ded6d1 : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0df1303872447949e8cc33951038142 : SUCCESS in 31m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ba8eafa04bba42ca9524871901a23111 : SUCCESS in 20m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2acd0030a734c84a30d73faa718b700 : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":1},{"id":"06ce45ee9c71a9e7ab481b7746b56a0b6070b2db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 11:08:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2d5e25ad2383f7885b0b9b91128c1f467456713","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 11:09:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51ba29c7f95149c7bd6dc6e7f509c858\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e9108909b84134b6b8b635ddefd1c8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a27b5a9e1c5d6f6a97422a8280a015260a5ca3bb","revisions":{"a27b5a9e1c5d6f6a97422a8280a015260a5ca3bb":{"kind":"REWORK","_number":1,"created":"2026-08-20 12:06:18.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/50/1001650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/1001650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/1001650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/1001650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/1001650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/1001650/1"}}},"commit":{"parents":[{"commit":"0043396f26a51f28fe9b18ea1aa81997cf844835","subject":"Merge \"arista-switch: Fix switch interface configuration\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0043396f26a51f28fe9b18ea1aa81997cf844835"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-19 11:02:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-08-20 12:06:18.000000000","tz":0},"subject":"Set MTU on nmstate bridge ports","message":"Set MTU on nmstate bridge ports\n\nPropagate the bridge MTU to its physical ports when using the\nnmstate network engine.\n\nCloses-Bug: #2164513\n\nChange-Id: I05d3f386157b887d5f550104fef7c5e7ffeed923\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 4ac2b42dda806375091404fcd1df8e8cc7e7aa26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a27b5a9e1c5d6f6a97422a8280a015260a5ca3bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a27b5a9e1c5d6f6a97422a8280a015260a5ca3bb"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"cf2affd778016ea32b7ac7361f704cc30ac03df0","ours":"0043396f26a51f28fe9b18ea1aa81997cf844835","theirs":"4ac2b42dda806375091404fcd1df8e8cc7e7aa26","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
