)]}'
{"id":"openstack%2Fkayobe~1003045","triplet_id":"openstack%2Fkayobe~master~I2b933b7574fb7c19bbb8f310e7c83a6145b65b11","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-05 14:17:57.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-05 10:54:16.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I2b933b7574fb7c19bbb8f310e7c83a6145b65b11","subject":"Fix duplicate word","status":"MERGED","created":"2026-08-31 12:35:33.000000000","updated":"2026-09-05 14:19:05.000000000","submitted":"2026-09-05 14:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003045","meta_rev_id":"20e9662bddb1f1562d9df65aaa6a2f26f7b64e11","_number":1003045,"virtual_id_number":1003045,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-05 14:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-05 10:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-05 10:46:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-05 12:20:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 12:36:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-31 16:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 09:54:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"dabdce87be27968947d8fee1f37c43b020455d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-31 12:35:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36617b0a972afbabfc964a0d8656d516b5b23898","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-31 12:36:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6b65133ef4f29b2889637dfe174f607365c28830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:42:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0cbc91c0474b76ac4d871ba9a571f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c6505611c840409adb248e6ec74583 : SUCCESS in 10m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db7523b633764f9aadb339970ca8bf7a : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a03bc20d6a84487ba6435a45f9a61b3 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad4d236de73452fa0fe0786ed689f43 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/181c51d6dfb2427da56ae015cf025255 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f90467bc16c94a43964e55bd70a0e15c : SUCCESS in 6m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a976479d44a7490daf98ea4413d89f5d : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6597c4fcebbe46f380143b28d5a3ff22 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30f83be2962d44029aefbe5d47921d2a : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/421efacc49df4ab4975fd67bcc463f4f : SUCCESS in 7m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/83d365d9ce0943e680380fc8c7af2fb0 : SUCCESS in 13m 27s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/0b7e40e0175d499cb3f76c6eb0e15cc2 : SUCCESS in 21m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf7db5a705f744b6b715fc8d5a2a2fe4 : SUCCESS in 18m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/075939ba76eb40a2adc91d266d48327c : SUCCESS in 16m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/599e5bcc602c4d388ce28ace5a4c8703 : SUCCESS in 26m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc7a61def02147a496c3556f571b54eb : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/79ef2f02ea2c4e0ea70f21f7b2d302e8 : SUCCESS in 12m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4884b0853a504f8bacde3e7741d4c06f : SUCCESS in 18m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2f25d98a944b4804885359d6fa62e5f4 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/878cb6dee01f4313bcd5ed325c3feef0 : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/49441d2ac23d4630b3aacadff3488b93 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2876d55569e146fc9118ceb435557115 : SUCCESS in 1h 34m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/633a77bb03614ff584c612ed45e5aa88 : SUCCESS in 1h 35m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c36ba6ea606b4021aea2aece28789a71 : SUCCESS in 1h 52m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84f40ea07f1d49a784479a6014bd15fe : SUCCESS in 2h 40m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fdeabf42957f404393f5e6e1a213384c : SUCCESS in 17m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c6cac2b87daa4f5e8d47cb5fb24893cb : SUCCESS in 16m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0345cc391bf845debf9329de202b647b : SUCCESS in 25m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/adb21702700c460d9a507681da11148e : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a727b5c651024108a086cdda7d0e3091 : SUCCESS in 41m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30489886510c447290d25ee94af009a6 : SUCCESS in 37m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2b76d6f24aa94865a553b94e049a2840 : SUCCESS in 21m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90daf3798e24478b96ee2228031b7736 : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1d55d37b7e18a6fa7f43569d4cf5efd398e700d4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-01 09:54:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"427f73b42737d16169333df645bfcba8c4ac6f84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee80c6e87f9d7d64ce922f7dd8dd28a43cc00d69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:43:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a40395aea5544195a5a2212494f417c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24ecd04402f4b4a99b19fe175282227 : SUCCESS in 5m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c891790172214dd3be1650b8cdf6978d : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd42325b59340f8958fcc44c6a05036 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a65a8926bd294358b97dee04aa772ef7 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36f91c83e7b54669bd768cca5ba7d84a : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fecf474f3f244b88bea5c6bcaae9aab9 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e3e2a698a4841ffa7261c9a003a7f42 : SUCCESS in 17m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f742a5c89f414a46ac211fafa517488d : SUCCESS in 13m 18s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/efce13b332f846acbead56122a6d8c0f : SUCCESS in 13m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5342b8240656480b851268f6faef29cd : SUCCESS in 9m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7aaed978081840df93604b43321158d1 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cad6d08ac4884d91ac0ad3e463c095b6 : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4d214f55c2fb43f6bee11be51c79342e : SUCCESS in 26m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c997c18fce34da9b008a980b65ad65d : SUCCESS in 18m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/799c916fbb80400eb50cf08c781954fb : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2679110dc2f440f481b0ecf7b413edf2 : FAILURE in 1h 19m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/950115833011438999de453f656655ae : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e18163c51d04b6a8a4982c943131619 : SUCCESS in 1h 35m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/667b389c9ff044709aebcdfbd32b27fe : SUCCESS in 1h 52m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d21a2e5142ba4030a1657423ce942bc1 : SUCCESS in 1h 50m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/667a30604cd44e85bc2ec8acf43d9fdc : SUCCESS in 2h 48m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7c755d0ffa6345d3871d82a162b2824a : SUCCESS in 17m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0f1ad589145447959ed3b9f61b4bf111 : FAILURE in 24m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8733049849c444a902974d03a4b0e2a : SUCCESS in 19m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c63c2ddee1b146ff9625c55d8954b117 : SUCCESS in 26m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fa6daedbeacd4746aba5e064bb1b4faf : FAILURE in 31m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d3a549c4fa8414987638359155f1f8c : SUCCESS in 28m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/482b96cf1547421a9ee7cd2e4a8bde68 : SUCCESS in 29m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8c4bf722bdc4ae8844ddabfa9372935 : SUCCESS in 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5ff456bf9f881f1f70f7e1e9dc47a71037b81acc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-03 10:46:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dade4d470235ff00de203e4e51bdd5956fe7bd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:31:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec819225f4ec4bc593a6e8dd220c68e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e62dd41d57149fa93a7d421f27ad2a3 : SUCCESS in 10m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2c0d2a1c2c54015ae1c2083c64b8c8a : SUCCESS in 8m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/681f3d7215cc4aa78cf78963e82d3855 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27668bd6bd164c3d87c9b876e627e327 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d0a925de6e144778b926ba8f25e05b8 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb5a53667e8b495d91f0dca3a3de3ffb : SUCCESS in 8m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f92e70530074a13831f8cd58bee9336 : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c609b378cd04155bbda66f6fa94e033 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b048de036954f9dadf3669fa641ccf3 : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6090e0f5a10402c938a6bd756cbfaad : SUCCESS in 13m 00s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a5e89dd208b14da995f1de8bb682a5d9 : SUCCESS in 19m 47s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/895ec5caf24644ffa0783967a5f39621 : SUCCESS in 16m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4070902e0554e60904465d998214fae : SUCCESS in 17m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/da4a91a2ecf24e7181e83cdef29f6160 : SUCCESS in 13m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/42b59e68676343098a1c69bc7081e75e : SUCCESS in 20m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41aada45e61343b79280e7a34a90d5b3 : SUCCESS in 26m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2086945769f94dc5b06e39f81319cf4e : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/155f4712c50e49efa142cef7a71d17c8 : SUCCESS in 22m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f6b360a08cbc4aab86efb4f4db7e40f2 : SUCCESS in 1h 14m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f675d20edaa848feb8b79c553624ec0f : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/50c5c97bbdb24f7987ff41a68add302b : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1960fc8e9eca4800a8e97d9e099a58dc : FAILURE in 38m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/350bfeeedb5847cd99c8aafa5d6c0db6 : SUCCESS in 1h 51m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1935f7ce998542c4b82d81829034ba9e : SUCCESS in 1h 58m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d9106e8a01e4394b2c117dde8cd578e : TIMED_OUT in 3h 03m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b386a0e68bea44bba4e0082d9eca63bf : SUCCESS in 20m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0bd1b3582e314218b5915b72036f5028 : SUCCESS in 20m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b51f424447c4738b007a3da00115c9c : SUCCESS in 27m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4cef80e2f59940cf85ebceae3a2e5c43 : SUCCESS in 25m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/98e79f5bacee48318a57d923ecbc8fa3 : SUCCESS in 32m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17c63831e3924faead990d595983769e : SUCCESS in 36m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1cc2fbae889344b48ad591afead3679d : SUCCESS in 19m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4de53a1ce8d421d964f129895a1cc78 : SUCCESS in 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cfda48d0a855ed5aa2df204a8e1332849ee4210a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-05 10:46:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"352f5e9dbd47c29f43a2015d1b46f9175fd3f407","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-05 10:54:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68326503740d3691072c278731c7de31a1efe223","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 12:13:07.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e956edac66b64e08a8185af1fe3c7862\n\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/66cd38399f164c5299332af11512a366 : SUCCESS in 13m 32s (non-voting)\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/51c3dc586b35468487696457df20b389 : SUCCESS in 14m 32s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/3a4ffb33d17a431b9d02e66db17f74ef : SUCCESS in 20m 40s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/f657feae33cc4aed879556b64e48a438 : SUCCESS in 21m 31s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/ee36f32fdafc42d7a4e8753bc511c1cb : SUCCESS in 26m 17s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/4e04dcdf3d0d4c4aa876089025c12c3b : FAILURE in 19m 46s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/8a12aa0896884c9cb5a153455b948060 : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/7f735732ef43404f87d2ed2245b64366 : FAILURE in 23m 45s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/e9ae6021c51a4e51b644f5a86e17ca8e : SUCCESS in 13m 50s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/f9989e7d07af4d4eaf2c5bd1f36beab5 : FAILURE in 33m 29s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/c76be0d712324122aa0eeb6a5a7b8619 : SUCCESS in 32m 45s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/d0b6d33dc0ab49de80912431bd502f04 : SUCCESS in 58m 32s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dd7499a731047719af25a7c8de19529 : FAILURE in 55m 37s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29896773606c4cc6bd7b26571af6df42 : SUCCESS in 1h 13m 45s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/b0a4f972a39e45b19fe546766a8461bf : SUCCESS in 20m 45s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/e6d7db96cd2a4da784923716041bd184 : SUCCESS in 24m 39s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/07bfd2ee79a04f0cb7c9bd86fb441a28 : SUCCESS in 19m 59s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/48cff473eb194553aa71fbd1ea6836d3 : SUCCESS in 16m 56s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/6eb97123f8dd4489850110e1061ff3c7 : SUCCESS in 17m 32s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/e6b9557621d74ebaa7a9294812799da0 : SUCCESS in 29m 53s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/73950330db5545df8c4ff548eb24d12a : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":2},{"id":"819bdc6724d3681b4bcbb15f3ec805d497525476","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-05 12:20:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"febd82b351ece6ed3a1efde5f46164dae5bd1397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 12:29:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0484fa71bce4c79861a8508ab2fc26f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4d69c2cf944066ba2c03bbf14f409c : SUCCESS in 7m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ec3d97b41bb4b35a6bb28ca7a76bf2b : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b08595745cc14f42acc71e799ac8675e : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1bbbc6ad02d4977b8746afd5197795f : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae693dddb3bd41db9bd43f5cd61cfac1 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c0ca41a45c54ac09e2ead45016619d0 : SUCCESS in 5m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7508fe330a78452fad45f9ea31c0a0db : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07c193fe713f47e9882be5eedeefe0c0 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28ed13521e0e44b29b80561c38bc7ec0 : SUCCESS in 6m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6b808947f5740a99531587636273ee9 : SUCCESS in 10m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1acdeeef6d514153b0125e9e70de71cd : SUCCESS in 13m 53s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d2d636705124f978115794652cabfd3 : SUCCESS in 11m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f2a1fc22b5e4d21b29886c9b498274a : SUCCESS in 15m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f65a6bf8dcb84d4d96ee1317de4caf75 : SUCCESS in 15m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/955f684a7ca446c5ad612bd4176adb23 : SUCCESS in 26m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59609715e14c49f8b721d2a594358019 : SUCCESS in 18m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/978dbfb9c3fa417892e8722dd0237171 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26d87c7dc6a04b0789d9c11c8fe58221 : SUCCESS in 18m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8250f5d55fe145e99655c0236f9b4dfd : SUCCESS in 56m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/04fcd70dec0c42728671359518e2d5cc : SUCCESS in 56m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c96de7f2cb344287b778bb4fedce7efb : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7a822f38c6d4a89853bf6f044901477 : SUCCESS in 1h 26m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba71fa4ccd3a48f98dcf9f8fb1207add : SUCCESS in 1h 24m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/386f9e28810b49879abd3be3cc11e87d : SUCCESS in 1h 41m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3949f9d5c3f44a8ab174dbf1ff7527d1 : SUCCESS in 27m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f3ea7de2a6464be9bcee17646e8167d3 : SUCCESS in 15m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9047e6b6852342caadfc4bd4920d90fb : SUCCESS in 21m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d60178976c7448efab6fe0c98087afcf : SUCCESS in 15m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/64dc5a71cc12411391a49879f3be9770 : SUCCESS in 40m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef37dc6529aa424eb10610001ce54864 : SUCCESS in 35m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e7d14420262248348aafd8e3f485dcc2 : SUCCESS in 20m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46d93bec273945f1a4a6cacc40387b14 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2a0c0b328998b8640c30434577e608a1f438f25b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 12:29:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d5330ebf99d3820d73fd6a9ff8aacdfc4f95717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:17:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30072a0870a543dd97f8c21f77aba93d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4070aac33cae44cfb95bc91693bdf106 : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/712070ce0482477880aec08d48297a46 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8658dd5abc0f416fab732d0d5a628f8e : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/151988bcea6f4c3d8e50dbc85fdd6305 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ef54e3ea3224064aafd67ea463b9888 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e34b471d34949c994aee7398479b279 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d75581dbfdef4b3799f151c208cd1633 : SUCCESS in 5m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/982b5bee4cc24644b2dcd91a53337de6 : SUCCESS in 7m 12s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/176b2cb6d85a4e5b84471e7895e02be5 : SUCCESS in 10m 44s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc2bbde9acab4ea69f2bc25a4a699856 : SUCCESS in 9m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1fcefc9d13c841f4860dfc65d93cd737 : SUCCESS in 13m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69ad9b2549c14f2ea5537186f8d109a1 : SUCCESS in 12m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5a2ab841f54d4192bf0d9775f2f4e469 : SUCCESS in 21m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03b335c7e20c4de09fa7200b424ef761 : SUCCESS in 10m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/06c41f507a0c45898839c28792cb35ae : SUCCESS in 42m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/201249bf60834def813239b07a5a953a : SUCCESS in 51m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c8300212377d481f903c237f311dbd7a : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/035983feef344e218fa37ca5c5eb3b49 : SUCCESS in 1h 24m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/074fe1ae7790418cbb3568f7db23fda0 : SUCCESS in 1h 36m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bcc38afbb80e4c3fae8a9c56aa299575 : SUCCESS in 1h 47m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e07cf66e321c4fd3b36373500bd1d3cc : SUCCESS in 26m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/91f5da8423c3435db414d0cfdb683d8c : SUCCESS in 18m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf496bbc48bc47c29d6faf657f4b7b42 : SUCCESS in 22m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df6d3485fb8746ddaf79a35eabffb3a3 : SUCCESS in 13m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3cc8d30063164d428a9f411c3e8c2db4 : SUCCESS in 21m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6a06b2ed86848e8853a25f4b10c253e : SUCCESS in 40m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7bfab534acdf460e8404b9c0d5de8498 : SUCCESS in 14m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fc381ddbee34ca08f11d95e43e245e3 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0db269e19ef8f599acfe448edcf29f4d841a35a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:17:57.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"20e9662bddb1f1562d9df65aaa6a2f26f7b64e11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:19:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e6bbf20b854290b03ba81a6680a124\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d216321cf9504250b558e1c2e5610b5b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31062ce91962cd563fd3bf90eb7e5e7b3a64237a","revisions":{"9bc77b2782755befd986204586cad0c2661b7f0d":{"kind":"REWORK","_number":1,"created":"2026-08-31 12:35:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/1003045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/1003045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/1"}}},"commit":{"parents":[{"commit":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","subject":"Add NGS support for Cisco NX-OS switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-31 12:34:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-31 12:34:47.000000000","tz":120},"subject":"Fix duplicate word","message":"Fix duplicate word\n\nChange-Id: I2b933b7574fb7c19bbb8f310e7c83a6145b65b11\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bc77b2782755befd986204586cad0c2661b7f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bc77b2782755befd986204586cad0c2661b7f0d"}]},"branch":"refs/heads/master"},"31062ce91962cd563fd3bf90eb7e5e7b3a64237a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-05 10:46:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/1003045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/1003045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/1003045/2"}}},"commit":{"parents":[{"commit":"372e505a474c8c137796e1f936b3f6431d319bcd","subject":"CI: Disable inspection testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/372e505a474c8c137796e1f936b3f6431d319bcd"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-31 12:34:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-05 10:46:38.000000000","tz":0},"subject":"Fix duplicate word","message":"Fix duplicate word\n\nChange-Id: I2b933b7574fb7c19bbb8f310e7c83a6145b65b11\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31062ce91962cd563fd3bf90eb7e5e7b3a64237a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31062ce91962cd563fd3bf90eb7e5e7b3a64237a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","ours":"9bc77b2782755befd986204586cad0c2661b7f0d","theirs":"372e505a474c8c137796e1f936b3f6431d319bcd","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
