)]}'
{"id":"openstack%2Fkayobe~1003283","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I33189b5be3e4622fd37ee854410258c1f8d41781","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-02 11:41:37.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-07 13:30:57.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-07 13:24:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I33189b5be3e4622fd37ee854410258c1f8d41781","subject":"[CI] Improve network connectivity check","status":"NEW","created":"2026-09-01 14:05:40.000000000","updated":"2026-09-07 16:25:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4b8f2b461ebfe969800924e9f3da992534b491ed","_number":1003283,"virtual_id_number":1003283,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2026-09-07 16:25:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":2,"date":"2026-09-07 13:30:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-07 13:24:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-07 13:30:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 14:05:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-09-01 14:49:04.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-01 17:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 23:36:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-02 20:58:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-09-07 13:30:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"f0c73c61e6972b77b29c0ac3aab8f6a4031d2f17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-01 14:05:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84590b5d319a8929a3b818b16ccf3c7abcb77a4f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-01 14:49:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fdd9315efe31716b6baf5998ffec6fe887695e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 17:04:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91aef636b5fc44fb9fb38bf0993ce3e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ea9ed8b4894993a1debec7abeda92f : SUCCESS in 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0fa44faa0ec34a32a24e5b4d28086153 : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4e5a434c0e643e5ac1d599e18a44e9c : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06dbcc4ec5704e70949a6b16e9fdeab9 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6861f01153654761b8048ea7fd73aca8 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88c436b7bef444e88ee2be5b7273c70d : SUCCESS in 6m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7707e4d59f7e4982b24e37232fc2fbcc : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0127212007142d3a7a42b7aac6e6743 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e77104c79c045a68d8a558a6f4f3577 : SUCCESS in 13m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5bf4e993c7cc406985b258486846a890 : SUCCESS in 12m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ba709a7f6f894d28b856a7c4030abefd : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a79a743a4db94cc18a0ee1901ca7ce87 : SUCCESS in 14m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7013baa066ef4f7185704faf55da2146 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3466d3b41b984d46b65da63de3e8f9fa : SUCCESS in 19m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/967f67c395624977845ce0945670fc9d : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6f774b5d546745f8a7d0c0a3a6050612 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0aee36082b484b80982ffcc81e3447e2 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1d40fe4673d48d2aa852c707fe51c12 : SUCCESS in 1h 41m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/14a41470fb6745af9e7e7037eb68c1bc : SUCCESS in 1h 34m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e4dfd374b92a4718864af992d1e4a97e : SUCCESS in 1h 34m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd6e6a0fd56c43febc744652ea2ed8cf : SUCCESS in 2h 57m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/16e04ba0f4aa45e6b2c88f55f88e3013 : SUCCESS in 20m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/603860ed66c54f6a834aaaf19036f88d : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43ff8929564f4e1c8ce2b74522eb7089 : SUCCESS in 20m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eeda845182254f4694124a060be469b6 : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/04fa0eb8c7704838869fc8d2da6deb50 : SUCCESS in 22m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40ec39ced788444ca78aad75a2be3f6d : SUCCESS in 33m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/965606c9972b4c12bac84b339df55f17 : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/697efc785a834e138b15cfa44f802ee0 : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5d3d85cb5e2bf3a4223235687c9a617bed264a81","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-01 23:36:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b672a2f3bbfab694f1797e86493505012bcdca9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-02 08:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"758d709c7d6501d20b6ef9b1418574db2e95cbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:41:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b33ec923bc494585804b309c692e670a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29a658f90ed4204942630439e9b2679 : SUCCESS in 8m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3006fa688ec40308317e66f3e5a30a0 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/234695eef6054a6c8ffd207bd4c4d6ef : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea8a07e0e8ad4a1497136b5aca71ccd2 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/334fec2a2abf41e6955885bf0e38c920 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f30f2a97905842808a8b46a30ade13ab : SUCCESS in 4m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97dcc98e13dd4c5497b1820f965b9771 : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f98c08e11f104d9d8acc8bed95440d89 : SUCCESS in 3m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c292c3a6efe9455f90cede847f1d2bde : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a6b6f92bebd46aea70cac19ab37669b : SUCCESS in 15m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1c7bcc59c4b74d82aa43e5acb1b07881 : SUCCESS in 12m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b7d7b4a0d6e476ea23fa811980c69a7 : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9401ec1eef98463aa199d11fbe80c35b : SUCCESS in 15m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a6142b6154944e3ab1807cba4dd0a35 : SUCCESS in 26m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b56dcf8514754ae18a8eaba79db2e4ad : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2499f7887c6d4f5db4c4e9f20f9d918e : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/32463c73847241b785cc3b0b56632bea : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c84e259ce27d4ebcbbe1e73407a6ef6a : FAILURE in 1h 49m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9a0092e9775411fa5fb339619542cce : FAILURE in 1h 40m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92a87939a4ed4752873ccf7fdb499aef : SUCCESS in 1h 40m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e14b413d9d524d36bb9e7423d44b0f30 : SUCCESS in 2h 42m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cb82f1245ede4da5a257551e0e0d752c : SUCCESS in 25m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/782e2345af96499abf8f59a35f1bf7d7 : SUCCESS in 12m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fc5ee77c32a45f1bbe72410ffa0ae1e : SUCCESS in 13m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8c0299b4c8a47bda11de635eaab4be5 : SUCCESS in 21m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0f85c84becd340e8a91fd4c5b404a5ae : SUCCESS in 25m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe89c3ad5cc7487db6c087cdf98feec9 : SUCCESS in 37m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/da746d9c53d0425f9f528dbaac709b6b : SUCCESS in 25m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63615f32b75548f4ab2e1a420beff1c6 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"35e59467f4e90402a229f5591077f20be82ac49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-02 20:58:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8e3fb58f628440925cf0fb4baa110412a0fb8dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 23:49:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72b2b9af92d9463aae88ae4040e4f3e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cced9d4e0a54fc8a7a9725d3cab2572 : SUCCESS in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a9d9a16308742f9b8a0f4ef5cc75e38 : SUCCESS in 7m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8559dba0881748e798f1abb9bcbb9a1d : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9155c8691192418e8d839707db975924 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/890006e59b3047e79dcd02b96970e624 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46be85ffe9ff4dbcaeb7bf13714af90a : SUCCESS in 6m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2edf4e284d65495eb3c402dda0e2ccdf : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13e7681b034943589c0d1da768082581 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d43a0b8b756b44ef80716d0b5cfc5f14 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01834709318c4853b5fcf41956e53c4c : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9622abecf61d4ca588af393de0aab93a : SUCCESS in 13m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8f115d608754e9a9494644aea4ac1a4 : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/134fcc38700e43e3bc178ab111fe43da : SUCCESS in 17m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8805e6d9d9d344829c0cf836fdd9a590 : SUCCESS in 13m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8e8e4430f0fe46e4b95545f119c6196e : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/31da0c646de74c2e9204502a72038e63 : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/383deee44bbe4ab7811358f22fdc88ea : SUCCESS in 1h 06m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98b5e79131af4e11a2496ceb114dc493 : FAILURE in 33m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf511ea49f944fa28206596e595f1119 : SUCCESS in 1h 41m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c33e5829bf3f45ce879ff094825428ae : SUCCESS in 1h 35m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f155bea70f554d639fdc1e83b0d90231 : SUCCESS in 2h 46m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f1733cd7a4de4a898718be3a963406e5 : SUCCESS in 19m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a78cb293534043b3a04d8cd2f09631f9 : SUCCESS in 14m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b51110541e0042d1835b040240e6a8dd : SUCCESS in 15m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/447aaee64c284a098e53cc02cbb93f0e : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/df75e1706b5b4928848263963d1c667f : SUCCESS in 31m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9359c5ee56642eb9642f457d62c15c3 : SUCCESS in 24m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b33ff5cf62aa4977b1c46d0cd94237d7 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f155ec04cc8844ebb329fb2e4fee90bd : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f650893d6987860e1c9287695e88db717e1918a4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-03 15:52:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd0b28719b12bee246b0cde05dd2ccfd37f21075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:50:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a3e530f00f14b60a0b5784efcec6a56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2387dfd76c3d412b88da80537c0ef7dc : SUCCESS in 5m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d6706149e9749448854854eab8ad7ba : SUCCESS in 11m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84523be1d2bf4773a8553f8fb900c63b : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc7afa4c9cc4e32856505c1fc51aff0 : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7022d3a0b5b42f2825f3e643812e7ae : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/596710e1f5d8411e9733577e5a4c91da : SUCCESS in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ecffb96261e4a92a6ee62e5f0a2fcc3 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83a027ce46554ecdb5f1075f54950374 : SUCCESS in 8m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7aa557454fa542c49bd2937b97f4ca90 : FAILURE in 16m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28cd1111427c4667982d1b53d2973f8c : SUCCESS in 14m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/564dc1f7af404dddabcf451249055dae : SUCCESS in 21m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98287ea7003e40fcaa048db82bc9a6cb : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ec2fe56d55654039b2f8fc22de215042 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de4b106aa51c4983838784a2de237195 : SUCCESS in 24m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e113b42e3911411c95d50c500f612037 : SUCCESS in 1h 08m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/97d81b90b16a424f9a8f4491f1c5faaf : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/34a4fd53aeac4d1fa376c4d1d64b6818 : SUCCESS in 1h 07m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1be5e55573c64eaead5548a9a97984f4 : SUCCESS in 1h 38m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b4943591b1034f64b0a3918aeb6502b5 : SUCCESS in 1h 50m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/abfa8dc241b3469997552d79f8bca700 : SUCCESS in 1h 40m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a9c6a32fa9e48a1a181f596495a315f : FAILURE in 34m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/33723b6ff6cb493fb31426daf54bb3e3 : SUCCESS in 17m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/799fd511861949bdab3f37def7b9757b : SUCCESS in 21m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d993829c9c6d4c258256a29098104745 : SUCCESS in 25m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d824462a9e34f2e99856e054ac95014 : SUCCESS in 25m 13s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/83efcd799f25468a8bc8ff0154b121af : SUCCESS in 26m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e32c45a2f5464d1f8779a5cc9e3792d1 : SUCCESS in 35m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db9506b536124a99915c6475a38f370a : SUCCESS in 20m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08fdb015d9c8411da8ed2d41eefe7901 : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2039e0db51e859267b9ce4c5a60400983c33a8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-07 11:50:14.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"91b020d7f3cf44803f8650a5d7ac1d62f505b309","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-07 13:24:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"82733304adbc9f60cec5fa84cd5214915b8234b2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-07 13:30:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fc9abcbec2a3f21537938e97686f7788e5d2668d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:43:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc03be4f43b41eba06f958c847dacc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b383cfc10e3249d9949b04ecff2af211 : SUCCESS in 10m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0089224d8a3b4dd1bd44905d7cd81c5d : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5128688016f04757b6ab6ada389375bd : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c1739a45dd48f0b497da23e44e5cd4 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a71c0016b914a33b65c44c4f16911fb : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9daf0df7e3cd445c869ced8da105ba83 : SUCCESS in 7m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/befba0a471854c17b727bc2bb3e1a09f : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67732ab70bab41f987817498079c03b9 : SUCCESS in 8m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fc05661d35d4d009a5e95af33f9dc09 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1db5201ada5f40ed9da46bec037da45f : SUCCESS in 13m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8d6a9fc341004538b84cb8e93381a640 : SUCCESS in 21m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bc20f1d25ac4ad08664b4beb03c5fb8 : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d76e90d4da8b429c82e31b0ef88d9bc5 : SUCCESS in 24m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741748064530464eaeb633f00e2e34e2 : SUCCESS in 22m 09s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d02abf887c7142609da3b02e872bc324 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e042d2807d7a492388dd92d3e9137606 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d420aee565324b1ba315def9c92fe1c3 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ecf243f51149fd86dd06fdef66e745 : SUCCESS in 1h 35m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d74535700f074d06abbe2f35584210ed : SUCCESS in 1h 47m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ed89152b3b944ea7abcfc374e656af5d : SUCCESS in 1h 36m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd5a2409e0147bd83d8417850d45766 : SUCCESS in 2h 52m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bd546e4511274ef3bc14ee1e0df3a521 : SUCCESS in 26m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5bfde797e7de4222b6658cea4afaa3fb : SUCCESS in 22m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d68ef42bfaba45e4b4f8f40417bd9a83 : SUCCESS in 18m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ed561ef137b4501a5769aff7d1e2d47 : SUCCESS in 21m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad6ed3ce35c744c3934dd6e7fdab8302 : SUCCESS in 42m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d954f20745441e49ab8b9d27213844e : SUCCESS in 36m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/98b21eebe7c2409386262b97f4fcdbf9 : SUCCESS in 21m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0b2cc5fc0974f88a90538794d93af9f : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4bc0068f805edcf3fb59d1b48973d2f295aec05a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:44:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b8f2b461ebfe969800924e9f3da992534b491ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:25:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3fc7f6b5cab4cb6ba70b988789a3380\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c4a27d01a046689ad576d5aee55b7b : SUCCESS in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c388d7a899f343afba9cc4ef9522efdb : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ba774a6ec44bcf8ea78fe230b290a5 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6592e46f79b4eedbbf314919e1bc561 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be41b232f76c4953a21c73c7909f0672 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9355c1adc2154069815db59fd31bfae5 : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07a6414841b348d49f298eb73576e133 : SUCCESS in 7m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/047c6ea1a9be4dd8834783cd9ea3e15f : SUCCESS in 13m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e3979f625ea6498082d30e6b4c29d1ee : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90421981a1fe46e68e922031906d93fd : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/43f6ad96c4234826905815026922099b : SUCCESS in 22m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ccf4a4ad0db406e8185ae9642d69b23 : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4aec29047f9547c4bd979df51b3cf669 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a935254579024458bf210e67f2f22169 : SUCCESS in 1h 09m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a31196ca3f904f2f9e522c53ab530186 : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f05ab8b008441e7bff0a4be44578251 : SUCCESS in 1h 40m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/179901a26eb046cbbe56192654099026 : SUCCESS in 1h 29m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/957278b88ed84d9fb423ea6dbda549cb : SUCCESS in 1h 39m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4301890f4c6d4c4699a25506e5b540ec : FAILURE in 48m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d26796a021a04b058bfb697b86f76707 : SUCCESS in 13m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8e263ab9d8da487e806bd32112b4ae97 : SUCCESS in 19m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf5a3534fe764b7caaf3354d32e3ae5a : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efa26e58f29740b8ab07568631028c8e : SUCCESS in 17m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d506aadeecd3403ea27d9f6c24803ec7 : SUCCESS in 28m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5911672b5e884d7e8de3ae42639b4e19 : SUCCESS in 17m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/36ef60d04fa74358a4af6bce5425d3ab : SUCCESS in 21m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0c8bc4678a34266ae6658a5b72562e8 : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ca86ca71f0a5d6019d2b0cfdf225bc03175be035","revisions":{"047befd51e6e70dc4ccafd79aca431dd768b29d3":{"kind":"REWORK","_number":1,"created":"2026-09-01 14:05:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/1003283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/1003283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/1"}}},"commit":{"parents":[{"commit":"48bfdf213f6afe8e1ec846c5878f3f1811b35a2f","subject":"Merge \"dell-switch: Fix invocation of dellemc.os* modules\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48bfdf213f6afe8e1ec846c5878f3f1811b35a2f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:44:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-01 14:05:25.000000000","tz":120},"subject":"[CI] Improve network connectivity check","message":"[CI] Improve network connectivity check\n\nIt will now ping between \"two\" hosts. Since this is an AIO, the same\nhost is added to the ansible inventory twice. This effectively makes the\nhost ping itself. It does mean the task is not skipped.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/1003282\nChange-Id: I33189b5be3e4622fd37ee854410258c1f8d41781\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit c85f3ecbf7da92d4b9b1c855131d725480ce9a74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047befd51e6e70dc4ccafd79aca431dd768b29d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047befd51e6e70dc4ccafd79aca431dd768b29d3"}]},"branch":"refs/heads/stable/2025.2"},"7fa0f9041831073bd9c68ebf7d9b034f6e576c3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-02 20:58:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/1003283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/1003283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/2"}}},"commit":{"parents":[{"commit":"1800251239ad460d98d63dbd458a63a9e9c8ee90","subject":"Skip external connectivity check when behind a proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1800251239ad460d98d63dbd458a63a9e9c8ee90"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:44:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-02 20:55:49.000000000","tz":120},"subject":"[CI] Improve network connectivity check","message":"[CI] Improve network connectivity check\n\nIt will now ping between \"two\" hosts. Since this is an AIO, the same\nhost is added to the ansible inventory twice. This effectively makes the\nhost ping itself. It does mean the task is not skipped.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/1003282\nChange-Id: I33189b5be3e4622fd37ee854410258c1f8d41781\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit c85f3ecbf7da92d4b9b1c855131d725480ce9a74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fa0f9041831073bd9c68ebf7d9b034f6e576c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fa0f9041831073bd9c68ebf7d9b034f6e576c3c"}]},"branch":"refs/heads/stable/2025.2"},"ca86ca71f0a5d6019d2b0cfdf225bc03175be035":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2026-09-07 11:50:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/1003283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/1003283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/1003283/3"}}},"commit":{"parents":[{"commit":"343e82cbf5517601eab5e1d73dc025c84ed061ae","subject":"Skip external connectivity check when behind a proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/343e82cbf5517601eab5e1d73dc025c84ed061ae"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:44:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-07 08:40:51.000000000","tz":120},"subject":"[CI] Improve network connectivity check","message":"[CI] Improve network connectivity check\n\nIt will now ping between \"two\" hosts. Since this is an AIO, the same\nhost is added to the ansible inventory twice. This effectively makes the\nhost ping itself. It does mean the task is not skipped.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/1003282\nChange-Id: I33189b5be3e4622fd37ee854410258c1f8d41781\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit c85f3ecbf7da92d4b9b1c855131d725480ce9a74)\n(cherry picked from commit 7fa0f9041831073bd9c68ebf7d9b034f6e576c3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca86ca71f0a5d6019d2b0cfdf225bc03175be035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca86ca71f0a5d6019d2b0cfdf225bc03175be035"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
