)]}'
{"id":"openstack%2Fkayobe~1003521","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ib6f815c319a7e92e675382cfe9d4011598e72aba","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-07 13:20:02.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-07 16:08:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6f815c319a7e92e675382cfe9d4011598e72aba","subject":"Skip external connectivity check when behind a proxy","status":"MERGED","created":"2026-09-02 12:42:00.000000000","updated":"2026-09-07 16:09:33.000000000","submitted":"2026-09-07 16:08:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003521","cherry_pick_of_change":956871,"cherry_pick_of_patch_set":11,"meta_rev_id":"3392624bcc1d603d334f5a9472b5a201b49da219","_number":1003521,"virtual_id_number":1003521,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 16:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-07 13:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-07 13:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-07 13:30:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2026-09-07 13:31:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 12:42:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"CC"},{"updated":"2026-09-02 12:56:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-02 19:48:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 16:33:28.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-09-07 13:30:46.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"7db1871dcff580e834adbc42f3fd75da247b3b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-02 12:42:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c2c954de93ccd50253ae7e279fb49711f9552d2f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-02 12:56:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03c106cf5e8c7940121664a7d80779fcfb4e36d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-02 13:55:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a4b79c1185f358b25642b480361e2ee774b1eb2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:48:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87169ecc4fed46819ac3b87b4fd76477\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c890ab12f6d44f2eaa336fbf58e9ca92 : SUCCESS in 8m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/44d40b01748b4b6792a0d4d1d5f08bca : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd8e05b8d0749eea24ec5dfb5996113 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0890c41102b34bdab738a0d50b313744 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca1dcdbe4b524698aa94b0e6a987b800 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3eb5d9cc8bc74f2fa539526fe3e19308 : SUCCESS in 7m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/494588f193384465a4a7fcd5fc2beabd : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ef2571d6f9e4eef9f1d7c785d64eec2 : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/598f935fdc6a41afaf65bcff4f13aa68 : SUCCESS in 14m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e99e3db1ef32448e9cb3d1f0f2131a87 : SUCCESS in 6m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ae944407324d4f66acb2c1ac14f5746b : SUCCESS in 13m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f38d638afec04acf9a71692d15156a38 : SUCCESS in 15m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f38f24db28b44c4b9ddda901eb4783a9 : SUCCESS in 13m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62c37c4eb3e64b26b6cb46063a7fe5f7 : SUCCESS in 20m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b64fc5f851884f0382df2bf301131399 : SUCCESS in 1h 11m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/92f287b81101447aa577b671e5ccb6f0 : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0d2a89b3c33e48c69f15344cdf7b7dfb : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77555bf4f4db49018a292a02a24c7081 : FAILURE in 1h 39m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3da5574b8a7436bb106e7e6a343fb9e : SUCCESS in 1h 38m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b9aeaf7a1eb14b89af9bcc026c078d6a : SUCCESS in 1h 41m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0b7ea7d79c24c3e89681842101d1933 : SUCCESS in 2h 47m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/50856a5159d04183a67d78dd9d6955fd : SUCCESS in 24m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ed8c7040dbe34c479f25634713281c24 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53520d98c4254167bc8946fca9ac856a : SUCCESS in 23m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cef7a3007807452783e1c797ab33751c : SUCCESS in 20m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/21d814af976b45bbbc5f09eabbfd64e4 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dd082b7a7404b2988c89b4d774f1ec5 : SUCCESS in 34m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/17cc0ba051e0493d9a35e04002054a71 : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f605f4fa32ca4cf5a72455fb21914862 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ae475e0ab01bfb197880f20f587eaf9fd5eb336c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-02 20:58:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cb6297d43f212a4dfa3dab3c3c9f366622624c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 23:40:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38311fdddf704e04ae42076d89456f6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6d022b88e04ed399f187f768fdfbe5 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2f58d88f78404ecf8ba01c7783b18f73 : SUCCESS in 9m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39f41d8a1ed64467a0ab57c467f35014 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde9866c35504198bafcd5c44e23b2d9 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64e1a333a1ad4c3b8aea17cf6d39d76a : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/554a793f0fd34f68a97e5187b265ac0f : SUCCESS in 5m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be193a48cc09469faf05703e85620897 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/933fbe38d4b343088e6058a7068a7e33 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76bd26a9945f4f12a4461aa6927af8c5 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16d67dd6249e400d91ea435da5233328 : SUCCESS in 8m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a4e13b587e284c90a44a24aee4077804 : SUCCESS in 18m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af21f144551e439f8aa52c6580db47e6 : FAILURE in 8m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0434381e32ac43e4b51c83582a1ba640 : SUCCESS in 14m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa7bd7e840c3474f84516e6aa6365145 : SUCCESS in 19m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9106b6e3675b474591cfc93b5607aeac : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2bb173c177a4e60b33c390eaa7c635f : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9595c7a9b62448f39a64d65683bbdb40 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3050bbae283d4b70be0a61f2928f8c92 : SUCCESS in 1h 45m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ae4b88ece114f5d9b12c3a3235ce76c : SUCCESS in 1h 42m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c576962402b4bfc85b337032d67aa6d : SUCCESS in 1h 40m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22bb819d73ce4cd4bbdead2b140a50ad : SUCCESS in 2h 39m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/88239c38f82846d397228d9c377ca08b : SUCCESS in 12m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/424d6bd2481d4dc28219e12eb8d4a374 : SUCCESS in 15m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f4acd5542ee4fcb9ec555288a437889 : SUCCESS in 20m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38f5230eaf99449fbde143f7124dbac4 : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3e3b02674a63443aacdc92cac68981ac : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9857596a57e42ecb34a73ed6612af97 : FAILURE in 26m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/933e87a53bcf4b3089b787485ab69d2e : SUCCESS in 19m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b16d46d18bd45ee943ef9ae5771a543 : SUCCESS in 20m 28s","accounts_in_message":[],"_revision_number":3},{"id":"18277f5dd6fe71dc9b19b4e6d2e2a8b3e028373a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-03 06:56:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c6b844b4ce756f8902e7162eaf386ba51423a481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 09:52:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35bce796888f47999ce35dc0cc24f115\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705e0ffa77914e8a981ecb6a5f201ae2 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2f0ccd9093447a9be7874aec9d58204 : FAILURE in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40b726d5052c4d148c27bc255095cfe5 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d1d793c24f4c96ac44588a82f25849 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29b6f388a2084da3a443275e7b0c2084 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2dc26aec3bdb4b57ae0405f740018931 : SUCCESS in 4m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e480cb900e34c759cbd86c63d06f943 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b10a7c688b9345d3a73ab036fb968519 : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39ebee98e2404c2da22e5b5cb3df0165 : SUCCESS in 15m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b619ec4cd514e91bd9f743b34281458 : SUCCESS in 11m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f2d59bd984ed452e8eb4d1a33b21aeb4 : SUCCESS in 22m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87dd5d98a5b34c69a688bbff3fab3cb3 : FAILURE in 7m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/15d16191c4a0455888e1d0226999c50b : SUCCESS in 13m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9743084f5694befa13d47fc227b7f54 : SUCCESS in 20m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a86cced49b494567b2f6f4f0f989a999 : SUCCESS in 1h 10m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d626426be1ca4236bc58f098cdfc9f43 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/97689b2a392e4d7284e1df4d7ea37b29 : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55d42c59ceb243e79f1c6e959e324bde : SUCCESS in 1h 38m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28e518b58b954d3f85cb21ffcd1be7de : SUCCESS in 1h 40m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/97e43322d9504a3cbd992af802238fa1 : SUCCESS in 1h 38m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3061f80384ca4df89f00d541a875ec5c : SUCCESS in 2h 55m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bc889cbe1e39432aa2461014f623eae9 : SUCCESS in 26m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd849a8707fa489683c506c807559062 : SUCCESS in 28m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/705b0eded22a4191b84074f3150fc6fc : SUCCESS in 23m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5661aa1171314e8aa5273a423c4aa736 : SUCCESS in 16m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/18280c34d6c54735b9cb8fc69726d8c4 : SUCCESS in 36m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf23941fb65447b78fc71c44685d8d13 : SUCCESS in 36m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c8400c8b162e4712b0168e36ad93972c : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab4940480390435688a436fa4b7cf5d5 : SUCCESS in 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ea58f9cb99b522420f2314ad8d6504231e18ece5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-03 10:16:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95c4102aca94646aa50c90a1e9d83a53776868a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:33:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eeb563e50b4d48c5b9bcb025c2e59a69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72bedef872de4d8cbd53b8c0fea19607 : SUCCESS in 5m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c84f9883483f4d3ca2badaa9ab5965cc : SUCCESS in 7m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044b5dde25a14c26bacd25646892d479 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c84dc8697f040d6aef935d29656fcc1 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c8e3a06b51b47c3b2f4b14f94fbe37e : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6c4789801b34065b4bb1c6ed025e86d : SUCCESS in 7m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5afe7f3171043a2a08864330dc95fd2 : SUCCESS in 7m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25556b20d3194d7981981164e031aa0d : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87e5d417623b48389cb3ed0e5c93227a : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/550e7acd13954abfa8bb048262b7ef43 : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9145d15fd942475db07d77331745f5ea : SUCCESS in 10m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c52a6993d0b94ab9a9ebde3e9125cec3 : SUCCESS in 8m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d96fd67a790e4df596824a73f75c9f73 : SUCCESS in 10m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ffd6a762ef412bab6496bba81f4cf2 : SUCCESS in 10m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9cb316dac6a04335a6fef5dbcd162cba : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bd85971f661b4cadaf0fe569f8bb52e5 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/704206025d7541259302d159d585efe2 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04eed513e5fe4b6caaa1d431cee37b48 : SUCCESS in 2h 02m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59ebfacf94fa41beb069e7c4c43d6a87 : SUCCESS in 1h 57m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de7620f127ec4dfc9af80085a13948f9 : SUCCESS in 1h 35m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ae61155cad548efa7c2f276e56f4f77 : TIMED_OUT in 3h 05m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/90102e5bde5b4b2085020424d2e815af : SUCCESS in 19m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4447ddb98cc54ebda3f775067f5bd365 : SUCCESS in 11m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/058704859d7a41a099f3398d6b02b3f4 : SUCCESS in 14m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/873d78f7eaa3432099f829b7f6305e60 : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/941994c32c454a6ab5dbdf00cbbae063 : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a65c7115a178460fa341dad7bfa7d758 : SUCCESS in 19m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d5ab9f0025cf4f34977eba497e5161f1 : SUCCESS in 10m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0883441255774ff2b95d7945fa1909a0 : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"890e4539635ffba6925a3a9a6595418416fecdf9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-04 16:33:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e53d3ee7f7c8b9e6bf8612f90e76bf7742a0763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:32:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97a9766f837e4fcf9309314110ee0066\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aadaa6d535834b3a9aeb97336ba749fc : SUCCESS in 11m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e290fac3320d49f79bc709b0b1266382 : SUCCESS in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56aec73674b840908ce48acf1a4e53e8 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eba2637b3cd493880445d4cdd9663e2 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cae1bcf2d424a50b10475ab6a8c4174 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fae25b2c97d14281b343aef172e9c3ef : SUCCESS in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d46955ae2b384b798e0012d70941961c : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea13e5fc933449ed955694a03f19b931 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ee5dd0e7b8c4071b5b418a5f4193d66 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a39d8058bcfd4ade8a139cd6834607d1 : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/70134ab34b5242c98a26430e68690304 : SUCCESS in 20m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e69e3bddd8eb458883c7f1ca690b5ee0 : SUCCESS in 8m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/578a823deafa455ebfd9e733e5b208af : SUCCESS in 9m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e18dbc5c915e406e8be6bd3bd4f98daf : SUCCESS in 16m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fe3ab1fe6542416ba4c23fc14f752b30 : SUCCESS in 1h 03m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c425fbd39e9144a8b58c83a9be0be3e5 : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4575a504299a4c0da7b25fbe0656e14c : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ac6b566bb274c1985154fda1940c95a : SUCCESS in 1h 45m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f0b306c8cccb4abcb6853f4a3588cd00 : SUCCESS in 1h 25m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/18f71b1167144a14a01d02ce332586a4 : SUCCESS in 1h 31m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9131a199e104376a5d3d84bf38d44cb : SUCCESS in 2h 54m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b8af933bfb384027a3ef85a758fce20a : FAILURE in 10m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ae4b6ae0e974434b5c05d5205583ebb : FAILURE in 12m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eacf2db8b86424aa1bbb4f73d3e410c : SUCCESS in 21m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b44d94ab25b9429aa2144fb547c469d5 : SUCCESS in 12m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c0dbe20951c84f9e80e21b148078a518 : SUCCESS in 26m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0897d4bb6f040f6860f324d51776c14 : SUCCESS in 28m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5041fcf32be74b7dbcee8556195d849b : SUCCESS in 22m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/734f4e9fbf7943e1be132bd97590187f : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":3},{"id":"443e04a48e6fb8600e758eaee06c1ab13e8a30fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-07 07:29:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"32d7f0f512b4f7267d047c9d7af361b2a966a9a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 10:03:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f24f03e300344609af09d01b40134bbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89fd9111857416e88ce861663f36b13 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6b33e9f498a4f819f9d8e06378b482b : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f7e2e722aee4c0db7ab0701e02c0bd7 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83704984b7e14c5c89ba0eb5ec9ac77a : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e4cec053ca04769a639c08ad79c29c7 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a4409bf3ca94dae9d567703ab0a8c54 : SUCCESS in 7m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fca07d4010449ddaa8bcda358f7a386 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a28869551c5045ea95ed8a10a03a7c18 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/301f54c74a7347c497e69a74e02ab200 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52025c2581b94ca494c0119658c38809 : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7b6151807b2a4e919696d8aef1fd81fe : SUCCESS in 16m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4544c77c914c4184a116bd8ba8bcb5da : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8aae758d2aa948d28c74f97e05ecc26c : SUCCESS in 11m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08607d30cb1740f8b7b3b852b2969820 : SUCCESS in 18m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/93944df019be43da878175533d2ea2b1 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ba34dfab35334b5ab1f51a60c68c0aed : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1799f759a70943ee835d412a8bae1237 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc53e9c31dfe4f2d88add07af2c73cf8 : SUCCESS in 1h 38m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/95c3c7cf368346bbbc0d572b7eb30bd3 : SUCCESS in 1h 39m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/519e5ac4c3194ed9b327d508633c5816 : SUCCESS in 1h 40m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d62a5cd48aa8474d9175f3121712820f : SUCCESS in 2h 32m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4552f3a1dd7247ffa9137c45fd8858cf : SUCCESS in 19m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/445f90ec3efa40848eca851346a1b5d9 : SUCCESS in 12m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aed3e38bb7846ca8c8259aa97d48873 : SUCCESS in 12m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b5a13b080071416499d2ef313714ebe2 : SUCCESS in 12m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/82177b7927144c3f9c416c8894b4901b : SUCCESS in 40m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2582818527af4321837dc1671ab2f6f1 : SUCCESS in 27m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ddbb704ebc7e42bc826254fdb61af5f4 : SUCCESS in 12m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1151e37a6dd40648008c2a288c888c9 : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":4},{"id":"1c318daaff3f74c594ddf2ae978ee73736be20cd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-07 13:20:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8de1c0ca92657f3cf540c830891d943b0e7de5ab","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-07 13:30:46.000000000","message":"Patch Set 4: Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"940f35ec804c4ccedaf9ac1bfb9ecb3e69886b10","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-07 13:31:05.000000000","message":"Patch Set 4: -Review-Priority Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2cf6ad988d535414319780ea50cac2d640cce6da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:32:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"81b0269af4ea8154e51dbe606dc9b3fca7cb5fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:08:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa9a1358a024cc1bedf63d7d9824627\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747cb668e621448cbe1e0759ce5cc42e : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/443f98f06e1841749299e67e87bdefdc : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c046095d7654ecbaf37af8acb5892d9 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e93da21d01e84dbdaf7c0e0a58e444d4 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/075586108e424c7a9d7affb27f540c42 : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bf2cea3c7234d03a7de7d2f73d1b8b9 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71717fc001dd4efb874f31d0bc70dcab : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cb33413224c4c0f80ae89a9cdeb9e35 : SUCCESS in 11m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/05ffed48d0204c47a2636bb24d7b75de : SUCCESS in 15m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7e857caf5fb427b975f5ada20b49c07 : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d07f4df316884db28aece7ff9cf7962b : SUCCESS in 10m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db3c595c5534f7d8f1548e81af1926b : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b5d93858bae44b2e9fe627c99477ff13 : SUCCESS in 1h 05m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8d29fa43c533417c8610c3fff6f50399 : SUCCESS in 1h 03m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c4704760c144cde93a927a429b464d5 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a47e38e15877470b8e02a247e26f9487 : SUCCESS in 1h 29m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c50f3ef155e940159d47ea1d7ae1a0d9 : SUCCESS in 1h 38m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7ba89defdc664d6da6db04269169a82b : SUCCESS in 1h 35m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcb59a2a69ea44b58f16a0c9d4762d0a : SUCCESS in 2h 35m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4ac949deb3894b9a8c3ac5e2690f2445 : SUCCESS in 10m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2328cee05c3d47e39ffe98a2ca926c53 : SUCCESS in 10m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc12cdac3d164ff3890ba0278d96eef5 : SUCCESS in 21m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ffbb08d93c3448f8bde73494470ea0e1 : SUCCESS in 12m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6408d5f43de540318ecf1a97f214f524 : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c29a843abd04bb3bde22c918ef79e07 : SUCCESS in 28m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3934d0cbc86c4ec5af3a37ab4a1d9b5c : SUCCESS in 21m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67d699370cef4c29b23b275a8d2d5ae8 : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":4},{"id":"49fec9bb040393e581ff2717ee8dcb32a4654cef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:08:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3392624bcc1d603d334f5a9472b5a201b49da219","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:09:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6ed1685b124814bf632eb2bcba4b07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9301e333e695446ea82097fcf8358646 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd75cfc9886f406889c1bcd7a9195589 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"343e82cbf5517601eab5e1d73dc025c84ed061ae","revisions":{"3a6a0ba328c00d97ba6f27e467ff87f7cad0d171":{"kind":"REWORK","_number":1,"created":"2026-09-02 12:42:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/1003521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/1003521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/1"}}},"commit":{"parents":[{"commit":"48bfdf213f6afe8e1ec846c5878f3f1811b35a2f","subject":"Merge \"dell-switch: Fix invocation of dellemc.os* modules\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48bfdf213f6afe8e1ec846c5878f3f1811b35a2f"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-02 12:42:00.000000000","tz":0},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails for hosts that have\nno external network, so this check is now skipped if\n``http_proxy`` is defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n(cherry picked from commit 8767fcc69281277139f7132f36a2a65f8a06f5e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a6a0ba328c00d97ba6f27e467ff87f7cad0d171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a6a0ba328c00d97ba6f27e467ff87f7cad0d171"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"8b36acddab1fc16c35ef86852c6a816da3617546","ours":"48bfdf213f6afe8e1ec846c5878f3f1811b35a2f","theirs":"8767fcc69281277139f7132f36a2a65f8a06f5e8","merge_strategy":"recursive","contains_conflicts":false}},"c73e803fb13945b7df2463fdd281b5c26501cdf1":{"kind":"REWORK","_number":2,"created":"2026-09-02 13:55:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/1003521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/1003521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/2"}}},"commit":{"parents":[{"commit":"48bfdf213f6afe8e1ec846c5878f3f1811b35a2f","subject":"Merge \"dell-switch: Fix invocation of dellemc.os* modules\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48bfdf213f6afe8e1ec846c5878f3f1811b35a2f"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-02 13:54:47.000000000","tz":120},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails for hosts that have\nno external network, so this check is now skipped if\n``http_proxy`` is defined.\n\nBackport note: replaced unsupported truthy filter.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n(cherry picked from commit 8767fcc69281277139f7132f36a2a65f8a06f5e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c73e803fb13945b7df2463fdd281b5c26501cdf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c73e803fb13945b7df2463fdd281b5c26501cdf1"}]},"branch":"refs/heads/stable/2025.2"},"1800251239ad460d98d63dbd458a63a9e9c8ee90":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-02 20:58:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/1003521/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/1003521/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/3"}}},"commit":{"parents":[{"commit":"da5b2ee6975b41469fdfa4325c33403f70ab80b1","subject":"CI: Disable network connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da5b2ee6975b41469fdfa4325c33403f70ab80b1"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-02 20:55:37.000000000","tz":120},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails for hosts that have\nno external network, so this check is now skipped if\n``http_proxy`` is defined.\n\nBackport note: replaced unsupported truthy filter.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n(cherry picked from commit 8767fcc69281277139f7132f36a2a65f8a06f5e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1800251239ad460d98d63dbd458a63a9e9c8ee90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1800251239ad460d98d63dbd458a63a9e9c8ee90"}]},"branch":"refs/heads/stable/2025.2"},"343e82cbf5517601eab5e1d73dc025c84ed061ae":{"kind":"REWORK","_number":4,"created":"2026-09-07 07:29:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/1003521/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/1003521/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/1003521/4"}}},"commit":{"parents":[{"commit":"da5b2ee6975b41469fdfa4325c33403f70ab80b1","subject":"CI: Disable network connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da5b2ee6975b41469fdfa4325c33403f70ab80b1"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-28 11:36:11.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-07 07:28:40.000000000","tz":120},"subject":"Skip external connectivity check when behind a proxy","message":"Skip external connectivity check when behind a proxy\n\nNetwork connectivity check fails for hosts that have\nno external network, so this check is now skipped if\n``http_proxy`` is defined.\n\nChange-Id: Ib6f815c319a7e92e675382cfe9d4011598e72aba\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n(cherry picked from commit 8767fcc69281277139f7132f36a2a65f8a06f5e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/343e82cbf5517601eab5e1d73dc025c84ed061ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/343e82cbf5517601eab5e1d73dc025c84ed061ae"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
