)]}'
{"id":"openstack%2Fkayobe~1004574","triplet_id":"openstack%2Fkayobe~master~I1545baeb7e437b4b0928fdb054b4ad34045357dd","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-09-20 23:10:19.000000000","reason":"Change was submitted"},"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-17 16:24:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-16 23:40:41.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-19 05:33:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-09-20 23:10:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1545baeb7e437b4b0928fdb054b4ad34045357dd","subject":"Add support for configuring Dell SONiC switches","status":"MERGED","created":"2026-09-08 09:14:41.000000000","updated":"2026-09-20 23:11:22.000000000","submitted":"2026-09-20 23:10:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004574","meta_rev_id":"ab8a8bb3775f84f1012861a20b824b26f90d14ea","_number":1004574,"virtual_id_number":1004574,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-20 23:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-17 16:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":1,"date":"2026-09-16 23:40:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-19 05:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-19 05:33:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-09-16 12:05:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 11:34:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 13:19:30.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"REVIEWER"},{"updated":"2026-09-14 13:40:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-15 09:23:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-16 11:05:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-09-19 05:33:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"ff3d39a18eef5b87b95fe490f022b2bb5b89063d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-08 09:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e0c8cf68084c2a73986c12a1b6b27f8faca8ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-08 09:41:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"678caa8ba980f4a81e051430109f5bb7380d87b7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-08 09:42:40.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8c480777e79be3eb8d5d152110140b45e96a080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 11:34:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0c65515e29e4684bce7f9dc590d6fbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37bc2a025c5248b4aa8482289f424939 : SUCCESS in 4m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ab9a38eab9194b449c1c83696a756a04 : SUCCESS in 8m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/338b1a6f0c384cbab1946590922c36f2 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447e7cc9ed5e4937a4d90f9354cae3b0 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a29aecfb6824382afb969bfeeb5da29 : SUCCESS in 5m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbb6eee6cea14b5e9e1b8a39767f2073 : SUCCESS in 4m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d61ffb5025f48a089cb2f3289694177 : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5689cea43ed64b11836419e77260e11b : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d893f8d752c2477ca17c1d50fe704e27 : SUCCESS in 18m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f189e2a71474083ac6d41d247298bd2 : SUCCESS in 12m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dceda062354c40b584bbc1682df055e3 : SUCCESS in 14m 53s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c0739a3583a647358e4b8162db1aa5c3 : SUCCESS in 18m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfdbda0f376440798f1c8957dc4c8320 : SUCCESS in 17m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1e21443c31244fc68ecd2ba3a4861306 : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/56a64d7554664f119a195b8394241dc4 : SUCCESS in 21m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/957dc8dd21fd47d8b95274faea1ec6e2 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c67a88643e74a15a6c5f75b4b997b79 : SUCCESS in 28m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2cc0939db9942738be21b118b352f6f : SUCCESS in 13m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1f81c2619a704fa9ad4a5a90388acf92 : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e269dad4c6d542aaafca150cee171543 : SUCCESS in 57m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ed45ad289a0545d6a4d8eea2f5327ca6 : SUCCESS in 1h 09m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6fa8a9fe873461896ca15f192255588 : SUCCESS in 1h 51m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8f0de1bce18405eb04b55ef887a2fc7 : SUCCESS in 1h 27m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c6996d2372c5451498924fd8b27c9ca5 : SUCCESS in 1h 52m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ea9689bd36cb4d5c9f53b0c3c54033c3 : SUCCESS in 19m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/12650f355f8b4baea739b03e25a6563c : SUCCESS in 23m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/235acdc5f18341ba94e9dda8d997f1d0 : SUCCESS in 26m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2d708eb1f3b439e87ba8f47ab10a6c8 : SUCCESS in 27m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/75b160fe6aed477db1a27129066857b7 : SUCCESS in 34m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34755a2a51684679aa73b0c01c80a70e : SUCCESS in 23m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/97b493dca920457888458c0368a8acc6 : SUCCESS in 23m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40ba671962eb496d935c80484c91f493 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1eab35b341401b8bfa70b5d8776e117fe34c4aad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-08 12:30:03.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3060b20b276542e56042a16af8166245df2800f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:30:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49df54762c674470a9f479aaaeeea902\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9f1bad3cad49c7b095ed3d69a6dd5c : SUCCESS in 4m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3846ff316a194be7bb47e028e8019cda : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb9b3b2cf747488daf5b7eb9cd02ec41 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db228c2c6dce4e279983611714921a05 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bc6aba940e645f79ed22973d47779bd : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05a52315107f4bc2b1789f6a29ead1b3 : SUCCESS in 5m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b7f424588494acf9b9b4ea695297357 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c9e43a6a122443780300f41c7ce5c34 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/24cf4f361ddb47fca900b487490b86fd : SUCCESS in 15m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77759807abb442a99d653f8b2e3815e2 : SUCCESS in 12m 33s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/240cdb4d5f074c3da5ea2994fddd7980 : SUCCESS in 16m 02s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/cdacccd4eb994821a4dfb82387d31ea1 : SUCCESS in 46m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6373072311354dcea97099fe1c475cad : SUCCESS in 18m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d3add82b6a8491ab8c181b2fad70dfa : SUCCESS in 19m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cea8702c2e174cc6b4d6877a82224f49 : FAILURE in 17m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35624f09fe3941208d18e54f5286c521 : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1b76a7429aad4b239b6c1cfc1298b538 : SUCCESS in 21m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cc6fe4350a14357bf5c60c6a696cb15 : SUCCESS in 20m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3074707c42dd4dbe9d1dc25b5335b15e : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/16bd7bafac934221becd783d3a6b3fac : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ea36e519a915447883da3652e335e4cf : SUCCESS in 1h 11m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd2745e0dfb47d0af4b1df84b8e82af : SUCCESS in 1h 40m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b20e7080228452aa5775db6551f5ad2 : SUCCESS in 1h 27m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a2a8a5a6b24d49099c1ee4bec560891f : SUCCESS in 1h 59m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/06ab4f881d244331a8470680a48ae983 : SUCCESS in 19m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/baa759aab4f146bfa66670ab3358cdb1 : SUCCESS in 46m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/555c58a9d6474f2db76aee6ae75c9799 : SUCCESS in 25m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f396ae04eb1c448ea23e351e96a9cd21 : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/09d65b328ce54e1ca25f0cf8f1505702 : SUCCESS in 26m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b62be9dac57548c0ae742422edaa1a0a : SUCCESS in 24m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a477830e68a4529b4f4b1f1ec10dc01 : FAILURE in 17m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/325b3c7754b548b78c2cac6f009929e3 : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":3},{"id":"3e244b2a9c594536783076cede885f08be9a06f8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-08 15:38:11.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a5bd96fef3fcd7435290ca14b508a795318f17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:35:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fef42e006ebc442c8a9483477a6838fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66079a1ce48a4e6ca95c3963eee2431c : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5b50df30d0d84ca9a41f2df373b3ea5a : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29ea8c1eafeb4a0398e82e4e1c2a7ba5 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/801b00e7209c424c828436e9e8bb1fdb : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/938165ffd85c4b96937d214f93af0960 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc54eca497a2485088bef71115358793 : SUCCESS in 6m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ff32b248ccd44bd980d8c1127a6e133 : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6622f0069264c49afc681ca9722bcdb : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/edf06c1a7afa4b23b447944d22050e7f : SUCCESS in 14m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b801077f3b34756b721415d38dd2f27 : SUCCESS in 12m 01s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/90c8b10253334d4e8da7d1be90fc2098 : SUCCESS in 20m 51s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9e18b02ee54a4311951d65ed2ace3bba : SUCCESS in 25m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/171096417d1d4c5fa2aff288383e7d4f : SUCCESS in 19m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/549ac4a50c7a4962ad45072913de581d : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5007531613e146dc89c629a7f671d345 : FAILURE in 18m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36b92bde234841d2a42c371eba140bdb : SUCCESS in 22m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7286cc59942f4279ad57b286e2bfe12b : SUCCESS in 14m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e22a327af9c84530b4010c5bb53a13c5 : SUCCESS in 20m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/469a6c4cc6a24aaa844ce03c29906685 : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cca80c719abe4ffaaa5b8f20763b7481 : SUCCESS in 1h 01m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/806db93e535648f49aabd2021c6ba9f0 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e0dd9ce0b6b48bca07c5df6d161b4c2 : FAILURE in 56m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c8b875ab01f44da385727821b60c0d61 : SUCCESS in 1h 39m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f768fc92b23743c39c271f163d0ed347 : SUCCESS in 1h 54m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/22c842a7f37e4766a497f38d13c853d9 : FAILURE in 53m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e22324c913e497b949f1cac2ad5c803 : FAILURE in 39m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a9ca77c4c5245e6bb13cd2e9b4f1473 : SUCCESS in 20m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e1d0dd46950148d493a003b59a463c88 : SUCCESS in 23m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/35107e6d31944ee383decbea1a8ff667 : SUCCESS in 28m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b527288ac04466b80b8b435d500e97e : SUCCESS in 37m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/52e9b2effa904b4ebe1123c23c415507 : SUCCESS in 19m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6bb297587da4755a3ca18b070834cd3 : FAILURE in 17m 09s","accounts_in_message":[],"_revision_number":3},{"id":"983e7bdaaa526c8ad558568dc45059d35dacdb2b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-10 09:56:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b9e1d562fb21070da981e9c03ae143794a63399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 12:07:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c3b1af7980e49a29fade56ac6a23647\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da638bcef55a4561922b0f0561cc90f1 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa7496a8939548ef9da481df01561696 : SUCCESS in 12m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6445fda86b024c4cb66c1e9ee59c16d3 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ef6e54f4904916a3dc5e7e19715980 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6903b88b84934589bfc0bbafc341446c : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/016a77d6954c4208832c5f11758189d2 : SUCCESS in 8m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5f540120b46467397717d6d2e8db51a : SUCCESS in 6m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee0a7f12f3db4f90b5cb1d3d6d2e764b : SUCCESS in 10m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9b94d166a3b45099c6d3159e9082abd : SUCCESS in 22m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa903c86b69c40059c706ff1cd6fdec2 : SUCCESS in 15m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3052916495834fa4a0bd710df53ce56f : SUCCESS in 24m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8e9e2b8f352f4f7080d1eab4362077bf : SUCCESS in 26m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac15e343d8874cc3976cd2389e89e60e : SUCCESS in 22m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d4fcc868e854374b833e2cb43dd23df : SUCCESS in 27m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8dcd8c7d0cb54c67858b02efc2316c59 : SUCCESS in 33m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02b12cab92e745e7a9f1f22d0342c57e : SUCCESS in 28m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4a945bbafc9743f695bf39e565400f91 : SUCCESS in 19m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/589858290eab4c8b8bdac1a4ffeae3c8 : SUCCESS in 32m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/17df24e2c6bd4b29840b4eb1d44faffc : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bb2a268a1e544923a76e2f1de7c85189 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/00df0236cb8b41748974f7681786a822 : FAILURE in 1h 40m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5386f22fa04c4a138d01e73c77af4257 : SUCCESS in 1h 33m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d5a128c985d146e98d8aebfaebe39e3b : SUCCESS in 1h 22m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8d85702575eb47aaba868699c2658269 : SUCCESS in 2h 09m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/453936eac3864b9cb667bd7bcd0de3f9 : SUCCESS in 30m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/67dd88e8f4f84d9da577db57f6728af4 : SUCCESS in 28m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b493a73823a46c7a5445b2525555f2f : SUCCESS in 34m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32a9d6f291fa46e5af42372b0dd2f48b : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8066f6370aab4e2b919c5b593fa4f9c6 : SUCCESS in 35m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4906a03ad76a495daf8dbc09e4430f98 : SUCCESS in 35m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d782b372c8f1490ebbdd95af8acdc878 : SUCCESS in 28m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41560b755fc0495aac98e9b3e1a18603 : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4e129dbdbf4a1510d8bfa3944631806fa5c02fe8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-14 12:51:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"291dedaf8d60a28d9f661ff74bb9d806e5489a91","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-09-14 13:19:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c6cd5ab753c5a64e731757dbe2f3c9aa2969061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-14 13:25:59.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5d50868bfc022273bf19ecc99f5d92f0cd30b7c6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-14 13:26:22.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6d47f240516e9f6f647c6e283cef62f90338495","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-09-14 13:26:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7a849c6a80ba0ba2edace3226ee9098d22fff397","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 13:40:08.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7a37ddfcae7ed6d361771197e54419abd53eb36","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 13:40:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"328b378746d86e6fbb7f47a3ebf72ee9a33c4464","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 13:45:01.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2ad82ed7f72440f05b63a200f984499ae6dfb4a3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 13:50:19.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d0370baa1dbf83c72fb28b77a5345a92c3fe15bd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 13:52:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7bdaa38a89412e67e8f489e259ae902696edd3f3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-14 13:58:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cea12626d3f86679084adf8d6549c4ed5c3a7376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-14 13:58:56.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"569aabdacab79067c7c624bbc44efb20a6ab170c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-14 13:59:03.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"09ebd77c6837b99592ca8e00745fc3d6ebabb844","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-09-14 14:20:30.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"dd8fa5e88c3bd9e33493f90d5dfb6ca41135a3dc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 14:21:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"96229d70e4b1d3fe38210b87985fbb0f76b0a1be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:43:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e731fd65f74c472ea27b2f845c973c40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6880433327a64be9886e4fab80d1e80d : SUCCESS in 10m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0be0cbec8f0480caf07be511362e89b : SUCCESS in 24m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8ea84602044958aa9b05c1af43a99f : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d3cd66ed82a470a962353f5f471ce4b : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5a3b8beb95748d8994363d566f2eaf6 : SUCCESS in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4de34867f0f843369438a72cb66043f3 : SUCCESS in 7m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4748b3136dd47729b1eeb9c2260e537 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d3177bee4d64c76a61627186e7cf594 : SUCCESS in 19m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a26e50825954694bcf33d86a7bccdbc : TIMED_OUT in 31m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7925e41b49cb49cca78b3284b0ef6a3b : SUCCESS in 18m 05s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/919bcff518db46a98717d483ecaadd3f : SUCCESS in 44m 24s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c364f7f4c2314ac8b6eddd2711056597 : SUCCESS in 47m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718629757d664b4f8c5f119554a8dc42 : SUCCESS in 46m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/09f6eb64e3184948b66dc1ed011af629 : SUCCESS in 45m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db3795d4faaf48d98fe864336812c50b : SUCCESS in 56m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edd249f3bbc04d819b675657c2f4aa37 : SUCCESS in 50m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/305a0234f63f4a0797fc9e1a2c691f14 : SUCCESS in 43m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d532d051f88b408cb3e7f25afb46ad87 : SUCCESS in 46m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b9342d1acf124d99aaa0dfe9326462bd : SUCCESS in 1h 32m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7d5a84147f8b442493fdf5b0f8a28e4d : SUCCESS in 1h 29m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/94e67b1a26a8432ab91e2423ae75cf26 : SUCCESS in 1h 42m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d934c1a0b004f4ebae684da6d0c8bc8 : SUCCESS in 2h 01m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3d6e041f8214be8bd3e9cf0d0f57ecb : SUCCESS in 2h 14m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/eb35de12e7384b8db54a55fdbff48474 : SUCCESS in 2h 41m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7b1824289f3a430799382b8f0273a47f : SUCCESS in 46m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a4dc1f5bb4a740d49e9345108d6ca007 : SUCCESS in 53m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cd2f94d8077408c8855c9c290b64406 : SUCCESS in 47m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38cd41a2c8f849df96ada079a17d7401 : SUCCESS in 48m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a27973d51b944d8bad21d724ad99ed02 : FAILURE in 56m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/845c5f78ea5f4829a3bf118ed737d81d : SUCCESS in 1h 19m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b34657ed3936461fb21cddde33078b3b : SUCCESS in 46m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc091865ffc5428182a10f64c50eef23 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":5},{"id":"3b2f381afa60db381a853e37c789177bd3f9666e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-15 09:23:12.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"82bf7614542cc1dff14875c6fba7d2d0f3224983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 12:06:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f96a89112624589bdd75baa5b5e53ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e4edecf06c476893b12089ada6988c : SUCCESS in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a0893cc6a134ca9bdb5019497fa75a6 : SUCCESS in 21m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6870472c049845b68bd567e0eb894243 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97ae2c90aea4c00929e175972562d30 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/262f00def5c14cb1bcf7f3eb4ed2c97c : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/796f1672d01e42d49c2e26a40c9145c4 : SUCCESS in 6m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94cb46518a3440a6b4097e83076e602d : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1df1a1cc6ac14c999d3c7b672d9ac575 : SUCCESS in 17m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2fe3058ce35d49dca681f945bbb6d8bc : TIMED_OUT in 31m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3a4ebebc10d42d19806fdfb366782ba : SUCCESS in 17m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/52ae2801afe84579bf4532ec99d3243d : SUCCESS in 45m 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a17b063da12746c783e541c8ade445fa : SUCCESS in 38m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73ce7c168a884dabbaa22febf9197987 : SUCCESS in 44m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ef64075af514e91beeb42dc29ae0c31 : SUCCESS in 41m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f66d1e89a8947f1b12002b76d0f667e : SUCCESS in 49m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ed1d0a5e36e4a2f8014ceeb7ec332d7 : SUCCESS in 45m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9601f99a0864435d9e7b5ab701abf205 : SUCCESS in 40m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88a8d0d63fa54cd18f2bdb147b1cd909 : SUCCESS in 48m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ccb366b71f1f4212a7eda8b455f0c4c9 : SUCCESS in 1h 32m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/def5512329f349f18a7c365dccf9f596 : SUCCESS in 1h 25m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7ebf35ecf2b54ebea1d170099fc9c706 : SUCCESS in 1h 32m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50e5a96505b54185a078de17f021ff96 : SUCCESS in 2h 01m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7178e05a47ff4b43af82727df5f2e95c : SUCCESS in 2h 03m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/94e05a4b0d1745b6b7131bd8adece6ca : SUCCESS in 2h 37m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/774229686fc74f6eb196339417a02d1c : SUCCESS in 43m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/522021003f4949229f2f21c9dce298f0 : SUCCESS in 50m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d55fb907a8046dda2064126d6002e75 : SUCCESS in 58m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3900a1d6552f4139a99e859c5e058754 : SUCCESS in 43m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/06c2733e22ab4473ba12c8bf76f10d4c : SUCCESS in 1h 18m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa7f80502ec04f06a47337cd588cd3fd : SUCCESS in 1h 06m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ec77e023fbf418e958f5155ea3fd9f7 : SUCCESS in 45m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f6e13e94fc4544af3ddf3abab46c21 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5d7f0a7702c05c9c82acbd439244f6931c8ccea7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-15 12:08:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7c9abfee44e07b0a19a49cafefa6a7f6bed7e7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:59:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a26bf30f8e8414689121aa8514464e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247f3b82de054bb79a9fa577c864663d : SUCCESS in 8m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dc00f6b9d8d24b9d971b1f3a358cbd34 : SUCCESS in 24m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74aad8dd14244bdfbce8854546f2e221 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4d5dd4db9e41309f492b7bcf6f0b69 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00a8123883814aff9baea9ff05edc8e3 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa73d356ec6d4c87b21a10a30585cdfa : SUCCESS in 6m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79b3d77bcd704d8f884d3b97af6dd356 : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a622e9588d3e4b8884a0751bdcae408b : SUCCESS in 18m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b751206b5554d2ebf93b3a460fcda4c : TIMED_OUT in 31m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/267605e878004dbdbaa0754912bac0bf : SUCCESS in 18m 45s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/27c1149d3940438ba3ac5e65691c2609 : SUCCESS in 42m 26s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a13767d85a0645b59ebbce6193c714d0 : SUCCESS in 43m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a477cb2937042b2997394be87b3cc5b : SUCCESS in 43m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6836bc4fd34d4c549619649a4cee35c4 : SUCCESS in 40m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6526869eb3ec47c3bc19a2753eb924eb : SUCCESS in 50m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8960bfe079924f4285ed376fed320a8d : SUCCESS in 53m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5d89292b356b47c28356496f59ccac02 : SUCCESS in 53m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e036fe373f584255b7e9d231867baca5 : SUCCESS in 51m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/38909199d8d44046972369ae6f78a402 : SUCCESS in 1h 32m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f03b6e54d70348a090a6f295a17493ec : SUCCESS in 1h 28m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/36af62b45819415eb373795341eb025a : SUCCESS in 1h 46m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/554af36a10e44a5b90dd6e568839aeea : SUCCESS in 2h 22m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0d25a27336b64505843510584915d901 : SUCCESS in 2h 18m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c8f547d7321467f92c27ac24bf785d6 : SUCCESS in 2h 49m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e425c1c3796240728f257700a9bad267 : SUCCESS in 43m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bf94825c73bb4fa38ab3dc48cbe502b9 : SUCCESS in 46m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70ca58ada90c4600bf64f3c346e0db1c : SUCCESS in 50m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8073f48cc053421fafb70a2a9a93ed9b : SUCCESS in 48m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4684940a9c5649efa4dc4eaed88e46aa : SUCCESS in 1h 08m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4efd13dd377e4086a418df91019e3aef : SUCCESS in 1h 28m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/40071aeca19b4f7eaadc882a462b18e2 : SUCCESS in 51m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bb4ef778da249c98c4bd8fcd6ce1ae8 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":5},{"id":"c39d2b6e773d04ac1057c204b9bbb9f8b90efcf3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-16 08:25:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc11da55b391f32456b1da83291c2ad4b3fb4c09","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 11:05:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c5541ae687cb713690242201be674c8766ea7123","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 11:07:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"42ddcee8fce19287fde71faca153518293ddef52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 11:21:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e052dcf50b54f87be40e6edbd40d1d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4573eb0c810248e987232bcafb86dcb0 : SUCCESS in 9m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/875512c8600d42f5a16fb98e3f5647be : SUCCESS in 22m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02fd64476b064fe8be89cb4f60569630 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80afea75c17d48c79f901e9ca731b759 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd9c1fda944247ce8b91c81d98eb4430 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e14801cab14d444db5061e7cd07bebce : SUCCESS in 7m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/521d24e061b34c9dbd031bd25dd3c545 : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c96545d2cf8d49c2ba0fb76d1ea79b7f : SUCCESS in 23m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b81be4347d9045c086313fe915bad787 : SUCCESS in 42m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5588b78a333f40e4a21f6fd3e564eeb8 : SUCCESS in 16m 47s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5ee6229f2bcd4750b19376e41c4b55af : SUCCESS in 56m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b9d9a902a4214391a3ef33f5bdeb1f66 : SUCCESS in 47m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0f94ef0d92e49aaa34cde5c1088c435 : SUCCESS in 49m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a86934cc7beb47649e5542890d87fb15 : SUCCESS in 50m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bcf92a070d1a4604b64aa0506ec9f660 : SUCCESS in 49m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b37f832015604aa8903c1ea8f49e3ceb : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6ffb3e924d7e471e89e76bf3e136db68 : SUCCESS in 56m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f68718d1540416a82a8b2fc3d7335b7 : SUCCESS in 52m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d9c8ae5068704b51999136a8ba0588f4 : SUCCESS in 1h 31m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d0f4dba0dae4feb94fc567d7605c0d1 : SUCCESS in 1h 34m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1f43ba6d937544f0a88edd66f6576020 : SUCCESS in 1h 37m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7e3e4dcce03497a9b827cd794338094 : SUCCESS in 2h 28m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee80877fc7594d33b276fd5fa87bb37f : SUCCESS in 2h 05m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de18f0dc14eb4d5a803b087f91d1a8c5 : SUCCESS in 2h 54m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b7226f8a054249c898f62750f7d4414d : SUCCESS in 52m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c7fec7ee993f4d11bc8e386c2800da34 : SUCCESS in 59m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/756600211acb4c2ea6daf9ed745a9e7e : SUCCESS in 58m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/721db8c3c27c4aa1a271634df7c367b2 : SUCCESS in 51m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c9b4549822b646e2b5b291eb160cad08 : SUCCESS in 1h 17m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c781415c101a4d1ba9d8fdf750f48655 : SUCCESS in 1h 22m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d4d24212af8348c8adb5a218386c9262 : SUCCESS in 1h 01m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a6f5d0a03db48f188677022cdaf5947 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":5},{"id":"71c77bd39fe32241eada9bf6c0f219e7ee45dda7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-16 12:05:32.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"cf8cdfe4dd82792983cf371f14d78eb943f2dfda","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-16 12:05:37.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4a9bc2c0d387f7b946f7ae99f9b8f63b6b336f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:12:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ea5c029f12b451ebf42a366bf814731\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400e9005f7a04868916d3952c1c45844 : SUCCESS in 5m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23d58cda89ad44cb8a87d19ecb95e1e9 : SUCCESS in 26m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bea1c0c115d41cda206bed495b270b6 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659a3fb493954a558cc762cc818340a5 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1bc04fdc82574c0eb8fb0555c6de0079 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f650da733576434c87c4cad33daeb486 : SUCCESS in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74a37aa77e424a048e99254265d4f698 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d9fec2ad94f4db2b13ba99b39586be1 : SUCCESS in 24m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/760e0af7aa464bb4b5615401165ad824 : SUCCESS in 41m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b343e1d8b580433c9eb0c3331916aa74 : SUCCESS in 20m 13s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7896b8306b4d4617a75e498609e66b5e : SUCCESS in 48m 56s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/0bfd5e9c5fcd477faafd00c5688f2865 : SUCCESS in 52m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cec15d87d1974f5d8642144eee0e8e7a : SUCCESS in 57m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e826be1b019145cab174a3efcd589fd0 : SUCCESS in 55m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1158a046a6df4922a22ed3a2955fb30f : SUCCESS in 55m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/927eb89cd5b546a7952473d1cacc37d4 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1447f576d79a437bbdd1ca48620cd389 : SUCCESS in 48m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c3db51964ad45d3849c4e82a9bf99d4 : SUCCESS in 56m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/744f2805e864413e8cfb78d1023cc650 : SUCCESS in 1h 43m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d040f7d4c7f047b3bd0c4af14156efc6 : SUCCESS in 1h 40m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fc7d01414265412499b191ac5d472284 : SUCCESS in 1h 47m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60b733b05a144efe9407577802192459 : SUCCESS in 2h 19m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ab6528dac534151bbb0e54a9d46fbef : SUCCESS in 2h 18m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ab29b40b11744e258994328e689a67f3 : TIMED_OUT in 3h 03m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f1f53a7c6a014e3bad4429911134267f : SUCCESS in 56m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2333fbf25c594afb96ac1f2dd9d0adc4 : SUCCESS in 58m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f52bf74fb4fe445d8b8e50d50f01c297 : SUCCESS in 1h 01m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/082bf17553834c80b2a820bf8092f7c9 : SUCCESS in 59m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/153b31350df04cfd8b24c7194b438e56 : SUCCESS in 1h 23m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1203eb56726241ce8f16878d5b926e6c : SUCCESS in 1h 32m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c38c9a07a4a24895b93de2ead14cf4c1 : SUCCESS in 54m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a37b832ae2a84a5ca46d066f2bfef7a8 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":6},{"id":"d07c1aabc726b415e68238d3f112afb924b20bc5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-16 23:40:41.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"64f311d44be99656842a31e1eec8ebfbd5d5bd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 01:27:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa597a9aed74fbba930c7e0031bc77f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10a3892b3ff48ebb25b2c60a1313db9 : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d668b3a85a83425c931cc7bca84892e2 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2643cc2c0f3452b8ef35f14e8cbafad : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded14a553c9440da8daa2abdbfdb31a9 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d12dd3cc681643de9175c5c72557b19e : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d31ceefb5c3b4536ba9d855419c6870a : SUCCESS in 4m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b60013e2624451b90f5aadf3736964 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fde94da7eb88454d9de8313456f7c84f : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a1e936456c14432a36538212b3f9488 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9968bb3e1ec4056b8c5bd30e68cf6e0 : SUCCESS in 11m 11s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/11c4aa8c1b38429b8893ac05f8510ce3 : SUCCESS in 17m 10s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/19064bc706224b128d8cf766a8fe8d24 : SUCCESS in 11m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4bcd8d307484c1d914b699a9873cb25 : SUCCESS in 11m 07s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63c741959ac040eba7737b859f63e021 : SUCCESS in 13m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b63c56f2aa2d464cb5a4cc310d7c2a34 : SUCCESS in 13m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02ee6ab68ede430fa785137e5a5f2744 : SUCCESS in 16m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f521c6ddc40048e4a8d604ef97066bd8 : SUCCESS in 18m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aabbcc71ebaa47f8b52757670a77ecbc : SUCCESS in 19m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/32e32063d415475f8aa1f78e5c0d463c : SUCCESS in 59m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f2f2478135d94805b852084aa6f28f79 : SUCCESS in 53m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b44962e45857482a846102d3273903d9 : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69daf8edee3d408886cc0b5d0ac43694 : SUCCESS in 1h 31m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03e3a14612c14fd2bad0a33b95f019ce : SUCCESS in 1h 40m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c6c25c42073e4d259a3b3f47f9e2195a : SUCCESS in 1h 45m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ecdbc6f3f3e64e0b8b57fe6b72a1dddf : SUCCESS in 15m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4017d47a69b2408e88e58f82051b798a : SUCCESS in 14m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dbc01b9af1748d286eb04d275eb47c1 : SUCCESS in 14m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bd8abe6f56a34c0daefff70b9db390e7 : SUCCESS in 13m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0e7f3e5f7a1a44e688d2bfc8a6ae2071 : SUCCESS in 31m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e1b909bbc0e49c3a2e0014d02ad0552 : SUCCESS in 34m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e92627dd77594728a269d47269b1ee39 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aef01bd2afe477ea1256e61c01df30b : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":6},{"id":"8f432c43ee11d2e843e4eb4fc668920209d9984a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-17 16:24:35.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"73b14dae5db8771c6c5a1497244406d1bd565183","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-19 05:33:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"bdd71dff96420b9860602bc69c5d3aafb83d1fea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 05:33:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f18673ee83a8528ea060224c16d53cae2e0e7561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 07:23:49.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b1345ea81d04d4e9b328c1112f93172\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3a512d2c614a8ead18485938b182cf : RETRY_LIMIT\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54a8dd209dc1410da42709c7eb75c012 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9aa71aa5efa4fb9912ea6e64fda3874 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1e511054cb945d8bbdcac6b098571dc : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa3973a3f2764b1499bbb713a88e6514 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eea88cad4eea48f283eab85a43bcb5f4 : RETRY_LIMIT\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fb20d87cdb044aca64c0e62222dffef : RETRY_LIMIT\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/540e1053d4824af7b98eb28338bd98fc : RETRY_LIMIT\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/38a8ca3c608d4972b5ff6648f004e626 : SUCCESS in 17m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a80d4f8855864bd3850332c76cf1a20b : SUCCESS in 15m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45e24bc466164647aa8501c511d8f9d8 : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb70de8baae44a639a41234b2fc46b09 : SUCCESS in 18m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a494d292fd54425694e938171a52c223 : RETRY_LIMIT\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/108a1b56b6584f468ab6c4553ec5baa4 : SUCCESS in 19m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/00b6de495a31466f948e41969233d072 : RETRY_LIMIT\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/05947147bd0444b0a6ed0d9cea7d3413 : RETRY_LIMIT\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f47eec6e1e33477baf355a18d24e7109 : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d719f9c6cf2447e58cd7878491bd62f0 : RETRY_LIMIT\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/999b8b8035b643b89e8dd6e824e471b8 : SUCCESS in 1h 31m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/41249d84827447bd89308c7631640914 : SUCCESS in 1h 49m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7df15449198942389e5c98f058ebd2f6 : RETRY_LIMIT\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4bb62e42be384d42955269e807dafa25 : SUCCESS in 16m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7158e647a9049048a644f2a32f80655 : SUCCESS in 21m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8b1e530eeadf4d459bc2aede331f6141 : SUCCESS in 22m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d0d6babdd7554654971d30d4601a9c82 : RETRY_LIMIT\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60702c0ea7044916aa770d9fab147fb9 : SUCCESS in 29m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd5f6e88703b47d5b00ccc210921ff11 : RETRY_LIMIT\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb36b947cdc84fd38a153370cb2c742a : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":6},{"id":"a3e55b47463b71f62ac8d721fe9c9dba596ab30f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-20 08:27:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9d074af5b5304542963952fa2038cc0e344d29ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 10:19:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ae69821cd6d4c6fa95090091e9bb5b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1665bc31824d4286bd31c4e9f6f287d3 : SUCCESS in 6m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/78d5fe3c81694ec7bd63d19bbd14e83e : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/004015f080a04342948fbb83c94e1672 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a3f08f42d84bef812974fd9e6617b6 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8db12ef356ac40c5813d0c36741268b8 : SUCCESS in 2m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a42bfbed0ce04124bc44384a7209ab06 : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b74e4f11e4794186a8d355cea5de62dc : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cdffb7a0a256424eb7ec5622345f6d58 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b192d04e013a4d6b8748ae7948604baa : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc59feed73de4b0f86d11ceee5883f98 : SUCCESS in 6m 24s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f0000adaed094f72b5c459abaad8d9a7 : SUCCESS in 16m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e5ac354b6b354f94876d1f0c10fe8d0c : SUCCESS in 12m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d1bc7172590463c88a66a0917db64cf : SUCCESS in 14m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83301bb12ae34791a1a0b3ed477ff165 : SUCCESS in 14m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8636895638574f5697cbdbf6211ee98b : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9468fa8a69b64785ab56e69c3dbdbde4 : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a8e0664401c04ed49bd65f3d42307c8e : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb26015179104804bcbef33e40bb9951 : SUCCESS in 17m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/76c10ee455f44f0596dd8eef5db7c430 : SUCCESS in 58m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/926e6bef292c4e3fa48895d5c5bedeb1 : SUCCESS in 56m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b4620b0807094875bb8cf3853af250fc : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01bf939386f1480cbfa4c0d8fa3a6d80 : SUCCESS in 1h 23m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb4fb0336b6d484ebb1b0df5a62d80d7 : SUCCESS in 1h 31m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a1ae188eaca74c3e9457ab35fab34b56 : SUCCESS in 1h 50m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b59cfbde6e46466cadb9dac40fc45269 : SUCCESS in 20m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c65cfa76c4a3438ba8dcc74a100f3576 : SUCCESS in 15m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/944aa1180b8e45d8b1defbe9406a19c9 : SUCCESS in 21m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3686817db0542a9a316b7aea4a7ba04 : SUCCESS in 20m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/52c7e69244d647d586ef46e4a13c24a9 : SUCCESS in 21m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d70b12a87e144b31b2f374ef53eac54e : SUCCESS in 32m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a98e02ca662349f98a5d7ffd1c5d64bd : SUCCESS in 24m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c881d0e29c0432d9479f6a6e0ae6645 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":6},{"id":"6e9e7c6c6ba6d1bdbe3e9c1cef33900786e48800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 10:19:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4996384a1e33c3f7f3190971123062a27eba95fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 12:12:36.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90dd887e85e44d8686bf1a7526e1abc4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b15ef57d17b46cab3659250f44045a4 : SUCCESS in 4m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5c9a011f687249909f6589762ca0d015 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b0891b4c5147fb9e40cf0e6c79d1cc : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bc960529ff243b8802fbd3156e879dd : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2438cc6d481c4b50955114d637eda0cf : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/824196208d164b53bf7cf92a6c1a516b : SUCCESS in 3m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f27649b4bfc47299db3476720fe53b5 : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac2f2b3d05b14b3f813c147eb078871e : SUCCESS in 15m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e6fc4bb58bbc4408bd19c0f8e270009e : SUCCESS in 21m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16d189bd309e432a9cf6ab50212353e2 : SUCCESS in 10m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e2172fa235834ef88b6433fc50a4571c : SUCCESS in 14m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f552eced89b4097bb7fb1bf951a83bb : SUCCESS in 16m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a2bd3db10e77455d8f8430d881f0c8b7 : SUCCESS in 11m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea8963d4e1e8417b8c96cb489d602f9f : SUCCESS in 19m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/568892d877b54886bde0bc9f6f543f69 : SUCCESS in 42m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/670b29fff7064afda12020ae4cd01dad : SUCCESS in 59m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/560688e97a4248d7892717df5b0da3fd : FAILURE in 1h 29m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53c2d2cbb3bc4afa822bca796df39ba2 : SUCCESS in 1h 52m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/332e6e7aad414886a4b56b72ffec747b : SUCCESS in 37m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ee4046d3f8764065bc3174921e600fc4 : SUCCESS in 1h 50m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ba188ea97e024f55b7d36edc2e157be4 : SUCCESS in 16m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/723b5a1da20b4358a7c6e7fdcb385ab9 : SUCCESS in 15m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3457d984f83647c99d76453ad504b6df : SUCCESS in 19m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcc9d525652545b4939c4190d8c633e9 : SUCCESS in 22m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/290166fddc014abab6fb6a802e874734 : SUCCESS in 24m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e253297231c48ffac8452965b9d93bd : SUCCESS in 34m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bfebb4f9351a4cc49570693fe997efa5 : SUCCESS in 13m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ed64cdf07814d48b371a8604da1c0e9 : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":6},{"id":"62b3b46290729232fd85481e16ba929d3b753a44","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-20 13:26:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e84f7551cea32453ad54be909671f82ba337c59e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 15:19:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41bd95eaec1e4a7583a2eaac083e387b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6999a10173f49c89e166a2bdce4330f : SUCCESS in 6m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c767df06184499498be4a1bf839d8b4 : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74658fa2028649b58de041a27bedad6a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5702fa7796b849faabbc6c7775bc48a9 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a98bec2ae73b4bc3924beab37b14a5f4 : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93559cfc4f1c4be2a3477b0f6ee2ed0f : SUCCESS in 10m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e252fd9ca314e3ca13a9777d45d52b1 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e00d5c8e315e4fde906b0bf0850b2f7e : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cab9c0df7ab540d2a2bdebe5d90f3e78 : SUCCESS in 15m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f8a34e720a7460791107d56d0c13575 : SUCCESS in 8m 55s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c15245d147394bef9d147bb902d2d576 : SUCCESS in 14m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7af0b9763f254f0fa06312f8dbeb79ac : SUCCESS in 14m 44s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e697e35537d3419785f13079d2342ac3 : SUCCESS in 13m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a63d9a8b3fed4324b4d9492cd381b702 : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f3321821a0334422becbc6c55f1b34e4 : SUCCESS in 15m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ff8c0020bf495b919e97d03f75ce60 : SUCCESS in 14m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d7333f0a8b2d48afaace2e6d0b272c16 : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d88430fd33344bf9a9efc69291829f20 : SUCCESS in 15m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ffd2060cfc5b480bb19c8f7932732c62 : SUCCESS in 40m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bfa5d7b430d3441690b2e52584f5979f : SUCCESS in 41m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9d70198a9a3c402390cb20cf6e1771df : SUCCESS in 1h 06m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1d3b534eff34526ac3be3fde012b102 : SUCCESS in 39m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/380f6a1e0aef4409b880fbfa10f42b07 : SUCCESS in 1h 29m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f81f7a07d7354d1d9f46211251496035 : SUCCESS in 1h 51m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b2bbfe681fcb4e5eac576a94d242b004 : SUCCESS in 15m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1b791d66226942908a2bb6ca0525d053 : SUCCESS in 15m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae49d031978d4cff8de8ee2354c431b6 : SUCCESS in 24m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45599d185d524a5e899de9b92d931063 : SUCCESS in 19m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9f83e38349d744378a05fb9fb3a19d07 : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd2023906953479999a70ee59cbf354a : SUCCESS in 33m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f9fa73a0883a41559f4e4f584b625fb2 : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cee883150c694fc7a0336c636bdefff5 : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":6},{"id":"618b63272009ac6526e09a11c3ca6ecf2f5a3e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 15:19:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f5974cfafb5bcb10970d16afb329c724077306d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 17:13:48.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/374c30521e454ae198ed99af2d1e6b9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fbc797466364a6a8cc9aa661dfed8ce : SUCCESS in 5m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23dca8c3b8764a6bbfb56c044a19c917 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/638c8acb8e2547a8af8346c6d0e1be1f : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e60f6a719204d9e94b889c0156c9ecc : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ebea0ca37a6414a8edf4a51fdc39bca : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/98b51de782e64cf4ad622d46ee01f980 : SUCCESS in 3m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1e27cba1e6d44f892f9eb16ef3ea3d7 : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa4e8b0a624e4629a58a7187d6fe4c7d : SUCCESS in 10m 47s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/97b4abfe3fea446a80c211d4180190fc : SUCCESS in 23m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7075e3e888444592bbdc699123a9bb49 : SUCCESS in 12m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/888bba9144304a7992403f3f3138aa57 : SUCCESS in 18m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32589d40ad6c42ada119f8d6c51002c7 : SUCCESS in 12m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d5f1b5faa4a24d6fadbceebf59d45aa3 : SUCCESS in 11m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58269b48b588474f87d6712fde61ff37 : SUCCESS in 15m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5314bc24b970458ebc45372a708773d5 : SUCCESS in 57m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a019ec87dc9649b4bb68e4a90dd64ac6 : SUCCESS in 52m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/dd8c0e402ae54e448bcdb1d885802960 : SUCCESS in 1h 03m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b069ac806dff4c04bdbf699a99e11570 : FAILURE in 39m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61bd91f7629c4019908a4a2bebad3a2a : SUCCESS in 1h 23m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/17acaf8155f548e8b6c70ee6d37ba6a3 : SUCCESS in 1h 52m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/36b78d635c3649b8aafd304cfc9d4367 : SUCCESS in 21m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6e1faf6e3a6a44aaa4465f2012af5ce1 : SUCCESS in 15m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bc333f7e58a46c4a2174845fec14929 : SUCCESS in 13m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de11c41540d74dae817291e506ac7ec6 : SUCCESS in 17m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/98c0d954118c491ab72cdcb6d9863a1b : SUCCESS in 29m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dfdafd30bf94541b6f4623b76ce2647 : SUCCESS in 19m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4ce7caf6772f46828d3d136f9ba66a07 : SUCCESS in 13m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c29787918d1f4ca4acd642d20d196f29 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":6},{"id":"789212efac37497d14f1d90ca1606b3628a13930","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-20 17:20:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f163bb17bff7d979a65272dffd402c2dff331372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 19:11:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90c143f9e83842cf91e54421e282263a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1267c1cb414b099265111d942791a7 : SUCCESS in 8m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/24b2962c4e7846daa83be42a97a63421 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74888fb9410746ae9e3001085237ed9a : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577b919bde98496e893f848c4ed734af : SUCCESS in 7m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9389fbc1d2484161aad593c5e8fe937c : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8043f31fe024418e99a9eba4d9d788bd : SUCCESS in 6m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b4e378e5c1f42e89b65e79f4c3a132b : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89629b573bb44f08a203de4fc16794d2 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cff10eeb966493fb2c0b0e5d9e01c2d : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c24853406d443d796cdc057d9747da1 : SUCCESS in 14m 58s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/468da402809a41b4b6ab5bb4a63f0d5e : SUCCESS in 9m 27s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/250b5802bf4b4349b63cfab911731a75 : SUCCESS in 11m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46a6df996be24ab3a30623052ce542b8 : SUCCESS in 9m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/91e6ea8e93d8478eb54b5121629bae8a : SUCCESS in 16m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fdfd05e173e4481596eb7d4db7967100 : SUCCESS in 13m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93843b59ca9d4d719b02f9965eadd35b : SUCCESS in 15m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5d612152c3cd4e4785524a33023da80b : SUCCESS in 23m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75ff0ef7a4c94f63b556a4c894d4f7f2 : SUCCESS in 18m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3f5e8903bae148d9a4b0cf6a6035599a : SUCCESS in 56m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ab59930284b4dafae301fcd78a7beeb : SUCCESS in 58m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/dbf9bbac4d324a3c9f10080fb97edaf8 : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cccf7b40edb4a19b5d7b704853530cb : FAILURE in 1h 08m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a0535967aef84e93a3658bb7fe2c2649 : SUCCESS in 1h 42m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/abcce639864e4837bb74fe0cbba89456 : SUCCESS in 1h 50m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8d3798fba8754c40bf5a6ae4d1a752a1 : SUCCESS in 27m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/999dea44c9254e028856769675e17cb7 : SUCCESS in 17m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e815e76f7e934369a08a9abd43e2a761 : SUCCESS in 24m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e60efbf5d7ca49d4823809fa6c23a5d1 : SUCCESS in 19m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3f415e8ef31a48089205a636fc63acc4 : SUCCESS in 27m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39485a3338904388b2ad2d632d88069e : SUCCESS in 30m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f61e2f8a310483386be67db8a2fcbc3 : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e01459761c434f6d87b66fbd4151bb04 : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":6},{"id":"51e57bdd9cd44210c7684b36b3139cb312da43c6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-20 19:32:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d6c1c23b52d628269a4583d0a74bd8f69d641c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 21:23:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3961ae088f849878e571f7b90f08fa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb079d14c3e430dbaad6d1f5295358d : SUCCESS in 9m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/15ef30015dcd4e958c08398f030426e3 : SUCCESS in 9m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9e31fcc99de4f96ae8a168aed1b5092 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34edbe0a08654db19543027b94ad5542 : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0613f714e64a4c8eacc5c07af7372df1 : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/344826cb5ecb42f1a42708656817d533 : SUCCESS in 3m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d75f944a37c54620820bf1e59175db82 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d26d29c464645d18cb1fcf3e58d11cb : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0dec79ccd8647d7959bad789414bddc : SUCCESS in 7m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cfeb449bd134c52ada56b7689fa23f1 : SUCCESS in 7m 35s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e1e2865fd2b34e7e88863d8686a30210 : SUCCESS in 22m 00s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a9141a5673254725b639041c96f3a01a : SUCCESS in 23m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d09117f5e3b64431982c6b2868aaa6e3 : SUCCESS in 10m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aff85eacffa147b9b86521f7f54ff833 : SUCCESS in 15m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cf3ad559bb914cdca6e9772a9673dad4 : SUCCESS in 13m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f6b513396414b9eb5a5812a43b90805 : SUCCESS in 16m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f8fa1e54e3bd46dab79f3ca35523e20d : SUCCESS in 15m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a35e0fb0be64754bd7a6bd122fc0cdc : SUCCESS in 18m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e76ff466807d4ed08c23fea489e5854a : SUCCESS in 57m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/da476d071a57418583296a72abf4eacc : SUCCESS in 58m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/35b3cb61c29b44139abefa0c108aff01 : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf413c735e704433848ca5abfcab0a68 : SUCCESS in 1h 46m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd633f9ee452416fa6c23d976e34ced4 : SUCCESS in 1h 49m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f4906ccbb2a2448c95f84db378c46003 : SUCCESS in 1h 44m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8732c5a955d54bb498b06c4e970902af : SUCCESS in 15m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/10cefe73fe994b51a1a87642ca3a2eef : SUCCESS in 18m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4811b2ba092e4d9f914253e03221ceb0 : SUCCESS in 22m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2a69e0c7062c441099410a58e604ac87 : SUCCESS in 20m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/56b7d21b1c9949d9ba086d2abd7dcf46 : SUCCESS in 26m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4664da22c7dc4ddaa3e0ebd191544aa7 : SUCCESS in 35m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e18221677a3e4e358c2ec04ba9a3fab4 : SUCCESS in 24m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d10d4169c53b4a51bdcfa953a549c7ac : SUCCESS in 19m 33s","accounts_in_message":[],"_revision_number":6},{"id":"82b34580ec0fad6423bed8cefe4ef874b7237fce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 21:23:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"34c88ea794acf6f22660fb94c0876d53447e0c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 23:10:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b9d3a776c54feb8b97cce6dcae0754\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa14df8ac75b43339eb79c6ba00831ef : SUCCESS in 6m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/decff0f66e394c999d27cd11376377d2 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea2e895ad1d46a48154cb9d084105bc : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c3f2ac3798e4b27bcdcbf1b5783e5dd : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b12d17ef6280450caa55352f6f9036f7 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b28ed612b37f4792b594baadd27f1af6 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6fdf8765a3ec47038635f661b1fb1f90 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f90e9a15bc344eb0847711935019cc1f : SUCCESS in 11m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3cc0042f7c9f426bbeeb1c5587366c23 : SUCCESS in 14m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3fb34af8f8948878f817c458505e9c0 : SUCCESS in 9m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/61a654751fb2472584c2c6db21e06e8f : SUCCESS in 13m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4988cef3bd54b37ae28186d5306acfd : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/514ea90a3bd34ca0945025794347212b : SUCCESS in 12m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23d823a5e5f247f5961ec025834d371a : SUCCESS in 18m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/138f712f51de45ddb02973e6419f4256 : SUCCESS in 57m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/94b2ef562767481caaa8db08902994d5 : SUCCESS in 57m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8559996692934eeea81a2886e3f34266 : SUCCESS in 1h 03m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cc6ab5ee348480a93642ffb0a64aafa : SUCCESS in 1h 26m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa0090a211244c5bb2bfab6ad1b1b2bf : SUCCESS in 1h 38m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f0707e357784daa873b0a7338798601 : SUCCESS in 1h 45m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/11b106560bb24c01b1737c97726332cc : SUCCESS in 23m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/69cb5fd6e4ae4a678cba0686af5d480a : SUCCESS in 14m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95acc9dc773f4dd7b6383b0c58939930 : SUCCESS in 22m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08a288f2dc7d42c2a61dea8233c7a4cd : SUCCESS in 20m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4a9b2483802949feb0b7dc3d57a15a43 : SUCCESS in 25m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11c52bcf729f4510a69d703b64d517ed : SUCCESS in 21m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/291b3e75266c4d01b94cdc22b9962398 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c2cd295f1aa4a27a6858c5558ca6339 : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":6},{"id":"57e24106d67d9ba14231a6007f503e6aa9660b6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 23:10:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"ab8a8bb3775f84f1012861a20b824b26f90d14ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 23:11:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7121e94368304364960c1869c16d9708\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f30ed66a654253a5d1457eb55b73e0 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54b2f2b6362d423aa57ba5515a8e1577 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc","revisions":{"270fd1a3337c590df788b3b780a6ba47a4977856":{"kind":"REWORK","_number":1,"created":"2026-09-08 09:14:41.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/1004574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/1004574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/1"}}},"commit":{"parents":[{"commit":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","subject":"Merge \"Fix nmstate zone sync for cloud-init profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9cbf9277581b0eb2b9bd241844d75a1301f7a27"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:14:08.000000000","tz":60},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for congiuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/270fd1a3337c590df788b3b780a6ba47a4977856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/270fd1a3337c590df788b3b780a6ba47a4977856"}]},"branch":"refs/heads/master"},"ef74bc2275fa078409d019713413813bd034bd85":{"kind":"REWORK","_number":2,"created":"2026-09-08 09:41:01.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/1004574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/1004574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/2"}}},"commit":{"parents":[{"commit":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","subject":"Merge \"Fix nmstate zone sync for cloud-init profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9cbf9277581b0eb2b9bd241844d75a1301f7a27"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:40:55.000000000","tz":60},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for congiuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef74bc2275fa078409d019713413813bd034bd85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef74bc2275fa078409d019713413813bd034bd85"}]},"branch":"refs/heads/master"},"7c03f7d899a05ceef5d60e7eca8aaaecd55db64f":{"kind":"REWORK","_number":3,"created":"2026-09-08 12:30:03.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/1004574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/1004574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/3"}}},"commit":{"parents":[{"commit":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","subject":"Merge \"Fix nmstate zone sync for cloud-init profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9cbf9277581b0eb2b9bd241844d75a1301f7a27"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 12:29:56.000000000","tz":60},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for congiuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c03f7d899a05ceef5d60e7eca8aaaecd55db64f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c03f7d899a05ceef5d60e7eca8aaaecd55db64f"}]},"branch":"refs/heads/master"},"e438c00268e7903d5701bcae0ef757da6fb5f58c":{"kind":"REWORK","_number":4,"created":"2026-09-14 13:25:59.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/1004574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/1004574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/4"}}},"commit":{"parents":[{"commit":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","subject":"Merge \"Fix nmstate zone sync for cloud-init profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9cbf9277581b0eb2b9bd241844d75a1301f7a27"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-14 13:25:52.000000000","tz":60},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for congiuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e438c00268e7903d5701bcae0ef757da6fb5f58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e438c00268e7903d5701bcae0ef757da6fb5f58c"}]},"branch":"refs/heads/master"},"c3ddcff1693a558c959c15255a1ec82605294790":{"kind":"REWORK","_number":5,"created":"2026-09-14 13:58:56.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/1004574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/1004574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/5"}}},"commit":{"parents":[{"commit":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","subject":"Merge \"Fix nmstate zone sync for cloud-init profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9cbf9277581b0eb2b9bd241844d75a1301f7a27"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-14 13:58:52.000000000","tz":0},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for congiuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3ddcff1693a558c959c15255a1ec82605294790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3ddcff1693a558c959c15255a1ec82605294790"}]},"branch":"refs/heads/master"},"a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc":{"kind":"REWORK","_number":6,"created":"2026-09-16 12:05:32.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/1004574/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/1004574/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/1004574/6"}}},"commit":{"parents":[{"commit":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","subject":"Merge \"Fix nmstate zone sync for cloud-init profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9cbf9277581b0eb2b9bd241844d75a1301f7a27"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-16 12:05:15.000000000","tz":60},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for configuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
