)]}'
{"id":"openstack%2Fkayobe~1004582","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I2e1985a76b7d532f15b38f257db05fb386b5dc8d","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-09-25 11:12:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e1985a76b7d532f15b38f257db05fb386b5dc8d","subject":"Fix nmstate zone sync for cloud-init profiles","status":"MERGED","created":"2026-09-08 11:38:50.000000000","updated":"2026-09-25 11:14:03.000000000","submitted":"2026-09-25 11:12:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004582","cherry_pick_of_change":1003932,"cherry_pick_of_patch_set":2,"meta_rev_id":"0d62b17824a39cb6b0337df32387bc7329472421","_number":1004582,"virtual_id_number":1004582,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 11:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"all":[{"value":2,"date":"2026-09-11 07:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":1,"date":"2026-09-08 12:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-09-11 07:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2026-09-11 09:38:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-09-11 07:33:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"all":[{"value":2,"date":"2026-09-11 07:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 12:28:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-08 14:31:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 07:33:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-09-11 09:38:07.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"REVIEWER"},{"updated":"2026-09-25 07:07:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"46683116d454bedaee28d151076f69216504193e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-08 11:38:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b6fb36b31f2c36e94e5e7dd0a0165cfde3b19ee3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-08 12:28:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab6b47caf1be4214c450347435f2e227b2ab7f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:31:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dbb84255e154c9699f452f4446604b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0525361db2204d4ba44639a55cfd4657 : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/96c0c6bcd0d048c3bcd836ad355c4661 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0a85c83fc4949c7a6e8a1a292b53494 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b481b7fe9bb440c2a07663c9202bfa97 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e955a8eebd646a88a8a06977e97e1bb : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd02c5b2eb0147d2bc5ec197f6af0426 : SUCCESS in 6m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c7e25b0d7a44410b298c1a07bfb66c1 : SUCCESS in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5940f30e8644b99915c9cc1e1e83b2a : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ba5570dfdc1941ff9705c20e4d0aa77e : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d42d0620215949258ff6e071293286fc : SUCCESS in 12m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c7622a467fbb47e49b24ba287cda5ae6 : SUCCESS in 22m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b51f1b495cde46088ae0052ea9dc7bf7 : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/031bbce05b6941508ddcea9bbd9b83be : SUCCESS in 13m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81f299035c974837b169951b9fd04f0e : SUCCESS in 18m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c9b81bb5ffa245149c1500ceef90252d : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/36fcbe4de4ba471996a3a450395d56c8 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a59861758d8e4bd098c0977041d529c8 : SUCCESS in 1h 06m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f40ec7dbde54476a7782988a27b4fdc : SUCCESS in 1h 48m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94474b7662ee407c90673ad03d9efb58 : SUCCESS in 1h 44m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a43b8e5b7d8c497a9ab428ee27c92eae : SUCCESS in 1h 41m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c7965c2178c4e7f96ddb68a74f018c4 : SUCCESS in 2h 50m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7b8e64a97cb64a3491e8f2fd9aa1913c : SUCCESS in 21m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4cd48125f7a9474c951d051f70562090 : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbd438c241cb4a34a9ecd30409e85d56 : SUCCESS in 20m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c20f73672ec4c068d3033c054c5553c : SUCCESS in 14m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/deb882ecd7c1464cb2f94e33bb8cb5cc : SUCCESS in 42m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4df3213259de4c529bc2c800d3e2e89b : SUCCESS in 36m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c2b89f794747416983daa6cb66e9fbc7 : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/805712b63c40417393ae0d87a394ed44 : SUCCESS in 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8d0c50c39b9e017091382a773a1159ef8775fa43","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 07:33:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab07f193a7d9398687e8ef643ec78a81375063ab","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 07:33:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9353391b85294ff5fbcef9ac74d27f567890d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 07:33:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3bf0729b7fa6586aea0e627dc91a977476ddaab","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-11 07:41:56.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"71d72764f5036a0376fc21c2a5d0a1f1e284d927","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-11 09:38:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e5f72e59d1d5a559559a6cda0aae8da35262f7a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:58:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10691a9055dd4dcda3a0e3894273bfd5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7997ec7f744cf38503fd1aacf95534 : SUCCESS in 8m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bad899c0e1ff4b7ab6e224c09aa9315c : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c06d54f1c2451ca59bf251513864f4 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d839721870a2445094c8379ad0e1c522 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e87837001ef455dab8943f384559035 : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a16fbfe819946ceadaa558916b04022 : SUCCESS in 11m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd66c6e541bd4cd2b4af86fe01e34efb : SUCCESS in 20m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c76bf2b3e6484fd899c190db011fb3a1 : SUCCESS in 15m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c87fc864a4db4eb183e25edc04e708c0 : SUCCESS in 20m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b7e8c609b15492fa9a2900e2af8da73 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/247aca1340ac477b86351afe323b8216 : SUCCESS in 18m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2441848f3ad4f3bbea23368d227c3be : SUCCESS in 24m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d5b02f031536452aa6f0a9d515a3e5a5 : SUCCESS in 1h 18m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2cc913462f8f4a89ab4946ded31db46e : SUCCESS in 1h 14m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/22ac58a73add446ca13848467917c18e : SUCCESS in 1h 13m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e50d83ec5033409396780c933d3c411b : SUCCESS in 1h 49m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c0471a1ecd24fc5906c9750c2bd36f4 : SUCCESS in 1h 57m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ca224e523ebe434697f58186de71032c : SUCCESS in 1h 48m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c812eb514864c6c83dfe721ceb956d6 : TIMED_OUT in 3h 03m 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2abecbc8b9734652bcf2f6c5f0f9c1df : SUCCESS in 22m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/413f4af09b264f27a79f8a7eca7cf7cd : SUCCESS in 26m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da1f2594b08642978d7cad1c516e2457 : SUCCESS in 32m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bdf19e54b1e745f4971e6551e7616311 : SUCCESS in 28m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e16b46bc60649e88767c26b059959b8 : SUCCESS in 47m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07244b13a68549fbaa613eed57123b74 : SUCCESS in 32m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/239e8f79da5d4891bda103c618cbbc65 : SUCCESS in 21m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09d6dde728274fcd8e55d78144dcba43 : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7ef570b0a8f3b4a5223bd2428a5cf3c03f0d2e0c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 13:54:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58a2ca2c496dd540f7b7890a7676e5a0376f8462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 17:30:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a2c86df85e9429ab3a8fc66a7476b77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c06582283a247cfb54803452ce15b4b : SUCCESS in 5m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4a9b86833a645a29d9c8e216cc3775a : SUCCESS in 13m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54166cf91d46450d98f7d2a937128d36 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20388ea93d04d4692209e81d7d645c4 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b355c6e8117b49e09505298c5196c339 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d40e71d471504910bb353ca501be647c : SUCCESS in 5m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/212eff086cd246dbafe67a7ae18cf43d : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14fd9e36ffff4d34a254fd7c1dfa6b0b : SUCCESS in 13m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6348d4293e0745f1a362a9c317cf650d : SUCCESS in 21m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/586aebfef3cf4d33a591bd5fdc26faca : SUCCESS in 13m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fc78358839341a7bc666a1605be43a0 : SUCCESS in 22m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0535a80e01ab403e8a9053ed21b79127 : SUCCESS in 28m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/556bbd72d0c54f4dbf13be6db2ca8766 : SUCCESS in 29m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd7e1d0dc8c24652b41c983a20e899e8 : SUCCESS in 27m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/90283561ddf74c098b09099090ba3ba9 : SUCCESS in 1h 18m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b4f772d17164f88ae0b7edf1d440ed3 : SUCCESS in 1h 23m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2a3f8ca1d7b844f68c0126b0e97b7162 : SUCCESS in 1h 15m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d6215e350c7418abee1186247182fb1 : TIMED_OUT in 2h 24m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c1aaff79a334bdfb0cb4c49e4234ee1 : SUCCESS in 1h 53m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b345bcd2e5e44ad493115edce41ed8b6 : TIMED_OUT in 3h 31m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/beb8d0fe65c749a3aaa7126d5ecd9512 : TIMED_OUT in 3h 04m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a0638484cde2477696812e5387573917 : SUCCESS in 30m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f6300858ac54b4c9e1449667f814c37 : SUCCESS in 26m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07b12859abc04428a04cc48291db6ea1 : SUCCESS in 19m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a3a2d837556e41409fb8f236cbbd01e1 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/99fdf20880b041908e82bd480de451da : SUCCESS in 50m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5dc20f0c220540a8a67d36aa173933eb : SUCCESS in 54m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/04a7682ccb1d4b3484cc6c3f86b7ec87 : SUCCESS in 28m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df57c89b53c74c269037b09c8139dfd1 : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a01234cef9802d5bb8baf86b60b58b17810bd22d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 17:40:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d8273b0f8714c21a7ea5c851fe8c9dcc62aa3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:01:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1452dc7324c04646b7ce835bc3c18326\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62df71ca11444ffab0b019714fc8cbf3 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb01f930dec442bdb90e9face98c24e5 : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b4a2160fe8a4d9e88f5e468403bf62f : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e17cc993f514f9298b1d854e68ae676 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45570328b4664cbb92764191e04083f8 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dbb0da47fdd449d93f919b00c7a5a17 : SUCCESS in 6m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ab4ed445ead47ea9d432fb8abe13267 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a88ae704df847e1b71088addf5a62c3 : SUCCESS in 7m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cd95623f94d48fa921a45cc4921f629 : SUCCESS in 14m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1c15c27171145edbd9732dc540f6100 : SUCCESS in 14m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1cb3f2ad501a47269b2f2ae954b7edd4 : SUCCESS in 29m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea246b9a008a435eab65d81755af595d : SUCCESS in 21m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/27d8b1fb7fef4f718610a8fd31a53d30 : SUCCESS in 35m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b174435143e64ddf99c17cccef9ab0ff : SUCCESS in 20m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/795bd1a156bb433f9a031b5ec2f8ab06 : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/21657163b77c498e83a6d7dab4385679 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a4f12522ae6f4a85a94beed793bc7932 : SUCCESS in 1h 14m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1377325899a444fb1942c1fefce94f2 : SUCCESS in 1h 50m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04f08a1edc8f43858f5c99d6b085b7db : SUCCESS in 1h 44m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/479cec3c32c34dc3b5c5c724aabe1f76 : SUCCESS in 1h 45m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d21a4d4ac024af391d5d840b6d9df7c : FAILURE in 2h 18m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/30ae841e890040a3af5e167a43d23b8b : SUCCESS in 19m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ae95eb5d5e904fcd85449d68c391ecfc : SUCCESS in 25m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2e5eb4a8b444e9d922bc005206e2b1e : SUCCESS in 22m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1e0c39cff6646b8bb61512b5cd38bb1 : SUCCESS in 13m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/487c3b4dcd3b43898db175ee430e676d : SUCCESS in 47m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25dd55844bc94d4eb72fcb0478d69db2 : SUCCESS in 37m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2daf154236434efe85e4ac861142bf5c : SUCCESS in 18m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bdf6712d2954f59b9912bea46151bef : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4d14c9fa6a742e3f768c81e48535af9ee8d8b4e8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-25 07:07:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"074087996a6fc2244ff3773ddac36c46427220f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:20:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91da41fde924cb4a10ca8d20937a70a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a789a1d8965495091ebc6bd15483f53 : SUCCESS in 11m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c10f1272215444bac80c31cb2571eec : SUCCESS in 9m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d816c17ed3ae48a0b47eb96b6cf89176 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f15a702802452fac553a157a2c28e2 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d0a7337e191481faa9cc92cdf984740 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58a1591ecf924858b1cb0099f5ff9431 : SUCCESS in 4m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d49c871ef754f319bfaf821e638cde2 : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e6410e313d04e0d8a3a6f0c9b01145b : SUCCESS in 9m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57d10a42ae7e4928b5e5c9902bca0bb6 : SUCCESS in 15m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08259a4301c443a6b2705fdedb8b518d : SUCCESS in 6m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6159154a248f4f92b23e7d684aa29bb0 : SUCCESS in 15m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86d2c8a8049f444da1c4120fc1c1d418 : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b58ec53d2ae41699c3182167de235e0 : SUCCESS in 24m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17a286fa11a444b7b9f6e7175442f28b : SUCCESS in 17m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/354a77d1cbf94b37ab7e17e84e94fe3f : SUCCESS in 1h 11m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/336a9dd2073748de97ef2ab5eb4e924a : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8274dbe67cf54333907fc2b45ad19885 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/717541262ca14fcf99347fbbfa98b705 : SUCCESS in 1h 21m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c256bc3d6dbd48518d2ac5cf38ca5f18 : SUCCESS in 1h 22m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cb276d452029471aa501fc7667b60d8e : SUCCESS in 1h 49m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e930e2e365bf4773882afc90ef4121cb : SUCCESS in 2h 11m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/51ccbcfbd36f4ecfbfba10a17cb7cfe4 : SUCCESS in 19m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/acc796f5184841f6889ef3483575c483 : SUCCESS in 21m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5618855e22a048b9a36f4ea00e97306f : SUCCESS in 24m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/953629f72510401e88f43c50adb25d87 : SUCCESS in 23m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3c1214a0f50649e58384c3e280e4eecb : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfe0094700394d768349de27873ff5df : SUCCESS in 37m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ff8172a96bb24488afa828076acca584 : SUCCESS in 24m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/267135922505480c9ccdf39e12cdeccc : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"174a3dea5e24a6711ae36ed1b8fc455b2a024823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:21:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25bed60b6a1d4e47b48a2778bd01dd94f84a1360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:12:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b922e5e407438484fe6d410fd04176\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c263205bf440d58f664253f440e052 : SUCCESS in 6m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d98f3057c98435ca3a9419f9ecb7bd0 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dcafe6da8864bd6b2e1405c71251223 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37ad5f24bcc34a68aa395c4f179437c2 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27cd2539b0914a44a61c7150408a35cd : SUCCESS in 4m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17827e23c8cf4aae864b9c6b36693d03 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74ca59dcc90143f49d45d15b2dfdac62 : SUCCESS in 17m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4eebe139bf5e46478402411f68e805bf : SUCCESS in 12m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d7d85517541b4542b03a610aaedc7481 : SUCCESS in 23m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6227a821658a49a8af28b897c3c85226 : SUCCESS in 23m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3e72b60719e34dc5b2208012d6664f87 : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c14d28f621d4995be9ac88ec022facc : SUCCESS in 21m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/60484c74a40042c2a31bc82c9b1952b5 : SUCCESS in 1h 12m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d2b298127c2b4354979ce5c01e8a255e : SUCCESS in 1h 11m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fa55af26aba44b03a0a60f8f99367466 : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45f074d566964d36b681e6774c8f678a : SUCCESS in 1h 24m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d68c60ef7eb4413a5508adb889f6dca : SUCCESS in 1h 22m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8e9080d37a6e4476aa263b9ccfe0bac3 : SUCCESS in 1h 45m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b584c1f93fa4cb2a69ba559d1a6d11c : SUCCESS in 1h 50m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e6552a741e304d95a93311df4111ef98 : SUCCESS in 18m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/960cb64905ce41fe8edffecbc1af3e75 : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7545423a84d4a53a8f45fb86626b24c : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b250736b761437faf30b5429de4d54d : SUCCESS in 26m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/755036ed5a6c482eaf757a4907100035 : SUCCESS in 31m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b027f157c604949a6f278d33d35139c : SUCCESS in 25m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6db0f30b382944dfab40546d0ca1e629 : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8018b74731e143a485dbc73c0d8a59bd : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"475ca0ee3f3c4366804361cb275f273fe64cea48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:12:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d62b17824a39cb6b0337df32387bc7329472421","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:14:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/359b30853dbb461ab825d7203a78cb3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf63f8d1e05e41718e5cc6dba263d032 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6afd88a8f90e427d8cbca1ba88af6de3 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fd551a38f4ef00e227976b5516310e09dcd64bb","revisions":{"9fd551a38f4ef00e227976b5516310e09dcd64bb":{"kind":"REWORK","_number":1,"created":"2026-09-08 11:38:50.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/82/1004582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/1004582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/1004582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/1004582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/1004582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/1004582/1"}}},"commit":{"parents":[{"commit":"343e82cbf5517601eab5e1d73dc025c84ed061ae","subject":"Skip external connectivity check when behind a proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/343e82cbf5517601eab5e1d73dc025c84ed061ae"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-09-03 20:41:19.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-09-08 11:38:50.000000000","tz":0},"subject":"Fix nmstate zone sync for cloud-init profiles","message":"Fix nmstate zone sync for cloud-init profiles\n\nCloud-init may create a NetworkManager profile whose ID differs\nfrom the interface name. Resolve the active profile before\nsynchronising its firewalld zone.\n\nCloses-Bug: #2165006\nGenerated-By: Qwen3.8-27B-UD-Q4_K_XL(initial review Opus5)\nChange-Id: I2e1985a76b7d532f15b38f257db05fb386b5dc8d\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n(cherry picked from commit c77e490253ba658838e8760098c7f8a948967d0b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fd551a38f4ef00e227976b5516310e09dcd64bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fd551a38f4ef00e227976b5516310e09dcd64bb"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"2489788216a1c90303e7fe11d0c802d9cd9b1253","ours":"343e82cbf5517601eab5e1d73dc025c84ed061ae","theirs":"c77e490253ba658838e8760098c7f8a948967d0b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
