)]}'
{"id":"openstack%2Fkayobe~1004584","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I2e1985a76b7d532f15b38f257db05fb386b5dc8d","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-09-11 07:41:39.000000000","reason":"\u003cGERRIT_ACCOUNT_35345\u003e replied on the change","reason_account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}}},"hashtags":[],"change_id":"I2e1985a76b7d532f15b38f257db05fb386b5dc8d","subject":"Fix nmstate zone sync for cloud-init profiles","status":"MERGED","created":"2026-09-08 11:39:33.000000000","updated":"2026-09-11 13:52:12.000000000","submitted":"2026-09-11 13:48:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004584","cherry_pick_of_change":1003932,"cherry_pick_of_patch_set":2,"meta_rev_id":"ebdc1021492e238300337e3d12a0e5284efa290b","_number":1004584,"virtual_id_number":1004584,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 13:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"all":[{"value":2,"date":"2026-09-11 07:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":1,"date":"2026-09-08 12:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-09-11 07:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-09-11 07:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"all":[{"value":2,"date":"2026-09-11 07:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 12:28:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-08 14:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 07:33:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"840390d7a1840eae65ee739168e29f448e2b52d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-08 11:39:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9e017136b0ce58f5edac4545886523f874d52157","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-08 12:28:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e7b1f492ba0fe417ea387708c614b354ada7085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ffb06bc92e46e4bc628ee7fa618101\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db5df508028c47ffb36b8fc6cef80611 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019f5899bf1e42a69d50538b0335dc09 : SUCCESS in 6m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c478f28cdaa4efcbe633656cbd3ed65 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a4776167f04b0088502adfa14012d2 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1878cea69e41179483e9d1a49cc3fe : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29eef007f7084e5eb8bd3c124d22677d : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c973488f8e9242909bcf352f46c732c6 : SUCCESS in 18m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6fbc5f150f0b4788b4c1f3d68fc6978f : SUCCESS in 6m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/85da310923504d9f99ed21e4c552d7e6 : SUCCESS in 12m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0d6700f51a564685983d9ae6efa047b6 : SUCCESS in 19m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/243f32fd51d04602858a0ed3e0016a0d : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ff8c858c6be34645a36fd54e2fd05376 : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a025060c7b744096a9007ce152d28c7e : SUCCESS in 15m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce1507b7389d4699b919f5d4d75989f7 : SUCCESS in 16m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/92fe93320ace4c6ab6942f72e69d9585 : SUCCESS in 44m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/afd7607a3c1a4152acf1c0314d41852f : SUCCESS in 39m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bb5d3b8602b547af8c094abefa008521 : SUCCESS in 1h 19m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/aff9c182dd8a4ebfb4e3a99ac63b45df : SUCCESS in 1h 09m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3462632ccee043a681c6641e817ed738 : SUCCESS in 50m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2657639e47b54e36969e14d438b0de4c : SUCCESS in 1h 23m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/315efbc6237e4232aa93caaf91a78d24 : SUCCESS in 1h 15m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e22f04c26e74604b8e9d369d66ae48f : SUCCESS in 1h 08m 35s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e56f338260e1428486bb782f82f0afcb : SUCCESS in 2h 26m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e578eb112d4a4625b316f259a164e5d8 : SUCCESS in 2h 27m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9214f53c762d423fb3852d1f6d886a56 : SUCCESS in 25m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b8f4c06f4a474bd18c825ffa1c93d277 : SUCCESS in 16m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a12d5c2b803c4bc2a8063fbc719a1884 : SUCCESS in 22m 56s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f0b9ecc48fdc498a84a3d2fd353b5236 : SUCCESS in 23m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/597863fd2eb14f2d8ecd568fc977b6a0 : SUCCESS in 19m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3da35da8d96048818956e0ca8dea0433 : SUCCESS in 18m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/74fac31831864c42a6818af11250422f : SUCCESS in 47m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d79ad6c735d3499287079fe469bc9651 : SUCCESS in 12m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d87b71afcf64f5dbf81c5bc1372b860 : SUCCESS in 20m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50c6faceea3445d9b4bc4093c8e1d3a2 : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":1},{"id":"37a01a0fc4dd71cff99fa7cae8bd86e9a555da1c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 07:33:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3068bc3b4edae89c109d2f678f80e886087e3df5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 07:34:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a33debeaa1af998de513bb66480fdfc8867e286b","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-11 07:41:39.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"152e6b9a1ec74a6da9ef8cd21c9fb40eaea2e50e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:48:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd245552927a4ae6b80c18c9ebd19fc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e750df36ec4d8f80fc7701790e5c4e : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dab832e64ce408392eaff0ff92b3afc : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477928a36bfd486ab4ad750abf7bc362 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/025c09a02e374de59fc62fb1cc2bfeaa : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c990340c0fb2412eaed2178e98b9bf68 : SUCCESS in 9m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26f60b5a873949a79155b499ba184a0c : SUCCESS in 18m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70077544157f4b319dd4f55c5df59fbf : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/02eec5fab63a46c7bc56973dd3bf6c76 : SUCCESS in 31m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a2bbe0346864dae96119dc02cd10ae5 : SUCCESS in 20m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df37be0066774f3cb90be2768224a101 : SUCCESS in 22m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bc70ca5e38954f158e58b0f414caaec2 : SUCCESS in 24m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c2ff7dceb1b244bca07cbf41f74e3538 : SUCCESS in 30m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70cd24db16ed4509a850f0d2dd3cc81c : SUCCESS in 23m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a26852797ef74764ac9ca7e264abdda7 : SUCCESS in 1h 23m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b524d4147a3947f6ade2bc3246e91ec7 : SUCCESS in 46m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/edde83e3340f42f1a2881af1b9345ecb : SUCCESS in 1h 17m 51s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7c6933f49ae94638b719cd88124577c4 : SUCCESS in 1h 22m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2714f87b3e0e4a7e9d9e14b998b974f3 : SUCCESS in 1h 15m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fb5715f7a1bb46bb8c818aa5d425054c : SUCCESS in 1h 35m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/774a54be10d7406c8634572f3c1f3654 : SUCCESS in 1h 29m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b92e63cb38064d549683d2983999ad5e : SUCCESS in 1h 17m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/17aa8019bba7490c8e570d0adb80c345 : SUCCESS in 2h 30m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c431228c47645b7a3c1623174a476bd : SUCCESS in 2h 37m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/adc5e32e86ac4aa8bce419f8ed3a1c7b : SUCCESS in 27m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e368efa120084b309fb0b12788298f60 : SUCCESS in 27m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/107d07348b354c0fb0ac3d34f1b4196c : SUCCESS in 27m 07s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f1c1a21c9fd408fa7a28df011fad747 : SUCCESS in 27m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2a7b50c23ea49dab3e7be5b3e60e8fb : SUCCESS in 29m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48556a9ef0774c66aefd3b2e422ebeaa : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3588e84c93ff4e69828d3225859d5078 : SUCCESS in 46m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a7be7833d514798b937a06373916cec : SUCCESS in 20m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8f87dad21c4d47b5af2d1d03fad4a2b7 : SUCCESS in 24m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab3ba702f01f47c092fc5a48b8918b23 : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cc14de88d09873ead7566aeb152406d27f0e116e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:48:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebdc1021492e238300337e3d12a0e5284efa290b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:52:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d103f32fcdeb4b99adbf3fafba0c3f85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae30fd6ad4f4e14a2878a3e99e03497 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33d62c81face479db5bad2cb64b38325 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb28325c16f197db5cb1ff86245998d306f42b5c","revisions":{"eb28325c16f197db5cb1ff86245998d306f42b5c":{"kind":"REWORK","_number":1,"created":"2026-09-08 11:39:33.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/84/1004584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/1004584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/1004584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/1004584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/1004584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/1004584/1"}}},"commit":{"parents":[{"commit":"7b31ae9dd595288730d7c008bba0948735ff8b9c","subject":"Add NGS support for Cisco NX-OS switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b31ae9dd595288730d7c008bba0948735ff8b9c"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-09-03 20:41:19.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-09-08 11:39:33.000000000","tz":0},"subject":"Fix nmstate zone sync for cloud-init profiles","message":"Fix nmstate zone sync for cloud-init profiles\n\nCloud-init may create a NetworkManager profile whose ID differs\nfrom the interface name. Resolve the active profile before\nsynchronising its firewalld zone.\n\nCloses-Bug: #2165006\nGenerated-By: Qwen3.8-27B-UD-Q4_K_XL(initial review Opus5)\nChange-Id: I2e1985a76b7d532f15b38f257db05fb386b5dc8d\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n(cherry picked from commit c77e490253ba658838e8760098c7f8a948967d0b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb28325c16f197db5cb1ff86245998d306f42b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb28325c16f197db5cb1ff86245998d306f42b5c"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"2489788216a1c90303e7fe11d0c802d9cd9b1253","ours":"7b31ae9dd595288730d7c008bba0948735ff8b9c","theirs":"c77e490253ba658838e8760098c7f8a948967d0b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
