)]}'
{"id":"openstack%2Fkayobe~1004586","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Id9d4d2643b0faf04d83553e289a40d2de8f00c53","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-09-26 16:28:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9d4d2643b0faf04d83553e289a40d2de8f00c53","subject":"Add check and diff mode support to nmstate engine","status":"MERGED","created":"2026-09-08 11:40:43.000000000","updated":"2026-09-26 16:29:53.000000000","submitted":"2026-09-26 16:28:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004586","cherry_pick_of_change":1003635,"cherry_pick_of_patch_set":2,"meta_rev_id":"af28d2ffc51155550a4f44950fd6a8fcdf06e183","_number":1004586,"virtual_id_number":1004586,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-26 16:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-25 16:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-08 12:29:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-26 11:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-25 16:29:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-26 11:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 12:29:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-08 14:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 16:29:59.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-26 11:43:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"7104e6168bdf5722bd3dc6662af3445fbaf6556d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-08 11:40:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fbb6d86ca3ffab0ac8656662b906363f5f2dd49b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-08 12:29:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf6babfed0d67209708f116a13efbeec2cd0752e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:31:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c342f508c45f4ae28cf0f537f006f1f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc46ab31e9cd4d34b4def675386225d8 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083d81909da74ba1bc9f8b5e2a7637b5 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fc5e632d59b41499604016d8cb7b7d0 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/802a41f5bd9c42f58d29115984abdd60 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b94da9f034924332b0925da23812f077 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91def2ad5bee405ea1fa103a7b859015 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/068e1e54f269405a91932e5a5bd5f677 : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcc811a720c0433b98b8b368549edfd8 : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0ca6a9793e2141598b2e43a341e3663e : SUCCESS in 18m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/108e729744be4d1c916316483b50c3db : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0b35d933219453580fa96b48fef0d70 : SUCCESS in 13m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d9a5a47e0251448f97687ff27c496015 : SUCCESS in 13m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5993cb3a4ee64cafb4cee536bdbbea1e : SUCCESS in 14m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf3ee435a4ad4a9f91ece76e71c3c345 : SUCCESS in 16m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9fc37a436db146fd97b1dde5de489561 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/918effefad1244cfa49f9283c0d49628 : SUCCESS in 40m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/baeaad1cd26a4ae9b455e717896a1d81 : SUCCESS in 43m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/adfe1df7904843a1a8450f7cdeff6057 : SUCCESS in 1h 15m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/51ac7af855924e11975561adfea8f938 : SUCCESS in 1h 46m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/60451bdd31884510ae553a571f40f0b0 : SUCCESS in 1h 48m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1693b6213d94432382c3b85dc2a1f387 : FAILURE in 1h 12m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ccfadfd0e5c4c059440cc0a626eb9d7 : SUCCESS in 1h 13m 48s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/26eb580f65a24922a52f4105af720541 : SUCCESS in 2h 43m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d5fde963389477eaea163fd12ef7d40 : SUCCESS in 2h 16m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6925525f14104065833107d3bebc8316 : SUCCESS in 15m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/997f011a7edd4acba2161ab797d5c4b8 : SUCCESS in 14m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/355af3b2fbd549b190e1356d83db38e9 : SUCCESS in 25m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/71d1bfc377614af1b1ee042a3b2d497c : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53c7db07e5964536819017a332141d8b : SUCCESS in 13m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c76c3ec6a64c4e51985c35fbd527e95c : SUCCESS in 17m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0cdeaca1ed634284b07cf4b2f609adfb : SUCCESS in 45m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab7a61bbb91344ddbf1dbb360f4a6fea : SUCCESS in 13m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/63ad7da7c860453e810c5fb7f3054b47 : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/033159c5aa0f40a7b0c3cd36e63bc7fe : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b0ab4cf5728d450b8cb8c32f36c48453d6871cec","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-10 09:53:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a6d0676e3c94112a4eb65f5ddcb592f11e86497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 12:40:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/088d7cec8e814e3b8165a8e2a0546f68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b18f885d2f342298aa889dde6a0cf31 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c147b59bc3241259938a0d442194389 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2b2a42d761b48bdad8373131cb7c576 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3bad93b79ab4d6b89103f33eca92ab8 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22debe280f224d669ec8ae6ea23e181a : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36e528a5a5d3487b97b4fbbdc8b48434 : SUCCESS in 10m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ca64ec705314edcb31a90b0dfce0df9 : SUCCESS in 20m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88c52a9c9b12439b9db38c7655b604b1 : SUCCESS in 7m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6c7ef6f513534295ac0f2a8bb4ab43fd : FAILURE in 25m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a1186466873048a38fcb334b049931d4 : SUCCESS in 29m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/204c1da7ed984c8ebb0ecdd505359433 : FAILURE in 26m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2e0fc7495fd0458f9b859c9143253960 : SUCCESS in 28m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e8c65fe288b540588bec0ec55370d0b7 : SUCCESS in 30m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfcc9fe33510485f9fde0550d88f27f0 : SUCCESS in 27m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f652f39206f74b17a991ec68b6f4b1fb : SUCCESS in 44m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5dffe98049f749e08d35d9a4d4a1e8be : SUCCESS in 1h 24m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eaeefc386af74abaa6ca6c711b13ea58 : SUCCESS in 1h 23m 41s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9323ec2c08304cafb79b50ac26497982 : FAILURE in 35m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c3c87519fe754392942dcf127930f521 : SUCCESS in 53m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/615a78ab0cfe45c0b198582d543cfc19 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f564595b63684b1fab8990aea9b6253c : SUCCESS in 1h 18m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dff32392fc78493987b7ad44ab99dbe9 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/39ecab6d296745abaca89407eb000e72 : SUCCESS in 2h 44m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77d822cc2ae048228ea8b236b7bfb1a5 : SUCCESS in 1h 15m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6b287dd8c7f94029934b5e5fa7f41021 : SUCCESS in 20m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fe9b31f01dca4e6e8e88e58c50f5995f : SUCCESS in 28m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bea54199d6bb4506a942239a4f82faf4 : SUCCESS in 30m 06s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/012b6642bfee493699e78ccfbb3809cb : SUCCESS in 31m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8013084fff0a4383a2ef3f083fe714b6 : SUCCESS in 28m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9db8914eb0c474dbcef9d9c41c0dcb3 : SUCCESS in 25m 43s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/578c30e9a3654364885c1b4a083eef63 : SUCCESS in 46m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c0b96d9abaa64be994958896e3a919ed : FAILURE in 27m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/db2c26a166ec43a4a355200bbc7a5994 : SUCCESS in 29m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09773a0c79ef4363825fbdd2f53317a4 : SUCCESS in 25m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bef7a1b103d3e1298620ee371cdb4c5437ff44e0","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-09-11 07:32:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec22b3d12920f8031e4f9b197dcf64e6a7755118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:19:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86e602d65c374607921c8c8eb17d4e6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5b5f79f158342b1b0dd8d0b7fa0677d : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f34bd93e4554f12a77ca17dd9833c4c : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf875e72873f473c9f7625bed0127972 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a7111ce48442e688a68c06e11796cf : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb72fe7cba84420ead503bc92ca91d1c : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ade7c3ebda6f4a8fb6a53cd4664c2d83 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dab23e76862842c2b431c7805de1ff42 : SUCCESS in 17m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e6ae0cc85fb4b78a3e21f7d84bc4b5e : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a0491eddf0f84f269dfe4ea1fa4a7ae6 : SUCCESS in 26m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b95c5b79b89e49b4b2bb6733c5865e32 : SUCCESS in 23m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fff9ae68754f42e1a29090cfe6b87188 : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d8c3993132e2495ca8c585f915eead56 : SUCCESS in 23m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c93c25741e84c36a36c510e46048e0d : SUCCESS in 23m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b895c76b06f4a3ca720f87f269f8ad4 : SUCCESS in 20m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/61e8e41b95164541aaa3ea6c7c20ce21 : FAILURE in 1h 18m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5965abafaee248c7927f2b422f061cc3 : SUCCESS in 1h 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d74f402dd0bd44ba917e972c835e3636 : SUCCESS in 1h 13m 39s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/6aee268573d84a5c8ab2d7d57a8c3a5d : SUCCESS in 1h 14m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c278bc8288b74115ac09fc08f0c9c771 : SUCCESS in 1h 13m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d3a30e9fa73d458d8f188e516ac6e398 : SUCCESS in 48m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ff7760c82614a22873fbba4040f9533 : SUCCESS in 40m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/290ad6d2913c4cf6813451b0757b0d08 : SUCCESS in 41m 44s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0cc09a7f729047959a7c0515803c4d88 : SUCCESS in 2h 41m 22s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df75b9c4bd14491da3b0f42283ce42b0 : SUCCESS in 2h 21m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/16001590ff1d4b1a931c047b5ee2251f : SUCCESS in 16m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54e6f1f8bfd04d25b9b144c2e64a98c4 : SUCCESS in 26m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5e3b2dac431b44faa06c3887482265cf : SUCCESS in 27m 09s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ee06f9aef184350b907550f3a64771e : SUCCESS in 24m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5a7e7d1e72041368e43259058752b26 : SUCCESS in 22m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f62917f103d42719c07d779431a16c9 : SUCCESS in 21m 34s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bc6d94eff3e04bc4b408f219c4c4c10e : SUCCESS in 50m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/874cfaf059f04b65b9457ea2aa05852b : SUCCESS in 27m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c45c9414083c4848b0fd4163d4805de0 : SUCCESS in 23m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b9b8dc9233549a4aa8833f474aca8c1 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"41d165c97b3907a9a23ca5f33a58915e0dd01baf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 16:29:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22ed0fff1c27a4ad6ac797084a3736a04fe7f094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 19:00:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23d49996d9244a818fc58148670353a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1feb8b4489c245fdb3a3510f72305df9 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea82eb157124654bd78352f73da9f3f : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c562e259204f493796c480f0c203de30 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aaed279bb1d492da12cb6759df60e2f : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebcef4d821cf4f57badaad7343e48eb7 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d75d454d42dd4534b9b3c2899b3006f4 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/385b2c6785b547fe98e5d14f9a6982f2 : SUCCESS in 15m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/246347f780dd47378b3347f400bda96e : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b431fe52fe414e52bd90afbaf3a962de : SUCCESS in 15m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5b20b37d2e554817b19155612b675884 : SUCCESS in 24m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4740ef10f2214db1a8ab6a36d483d2a0 : SUCCESS in 21m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f675933b495c46fd8857470cd6f6fa87 : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a93aef5b3e642ca8b93355578a67891 : SUCCESS in 24m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e7a62c1d72746f0a8b5157be58e5a7f : SUCCESS in 17m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a990745399d54f1f8035b69e91e8e6c3 : SUCCESS in 1h 15m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2f78a38a6284ea5bb3846f546734d3a : SUCCESS in 42m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/29a53ba12860460d8cf72ed32540de33 : SUCCESS in 44m 02s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb986849997f476781fd8b2fcb59efa3 : SUCCESS in 1h 22m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/12d464efb8004592a56fa7d3edd79d7e : SUCCESS in 47m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/53d4faf0c2654ea88c669b1f2f83e1d6 : SUCCESS in 1h 29m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79177230a67048009f6d66d0ba3a63ba : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/539666315576416aafbfac1c866ffc67 : SUCCESS in 1h 16m 02s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/391cd6d8cbd04f5d8d9ebaa7bc21bdce : FAILURE in 2h 04m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f5eaa97520414ea3d47305f6ce0f32 : SUCCESS in 2h 19m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3313fd0047e3493fab17bc8774cc8886 : SUCCESS in 26m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/45589e8a888046ecbd6253dd0738f132 : SUCCESS in 18m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cf9986f99995495f816eb37574703943 : SUCCESS in 17m 17s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/879e5203b4e949cc9c0813ca883e2041 : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df3b11ec11934b49b9a72e993e304f6f : SUCCESS in 20m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a87a8e152d8b4d008e5a6d5ba41bb604 : SUCCESS in 14m 52s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2a127bf259ad4e57a60411cb1035051d : SUCCESS in 41m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/baf0794fe1c1426fbd0aaf0d742f5d03 : SUCCESS in 15m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/447fee6aadd2415ba04dc11777ac85fb : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7b2421c8c374bcf88603ad03c1e522b : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9a6e92653c0c2fb973b41d031934db7201998337","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-25 20:33:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a8036e830b0ae20a9ab949857c9ed8b2177731a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-26 11:43:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b376c7187282fa9dc7b7a87ffc1cdffd7cf1a99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 14:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/196b443403af4730b66f41eaaab0c31c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40f3e044b2c24a328f6d675fee8473e4 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93e2bb5b211438c8d5204a7791b8592 : SUCCESS in 7m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09b7d482681d4051836682c5d24caef3 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb903a00564a49d68478ec34d5af67c5 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/011cc94e72704d78bbba0ab7587bf282 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f401b7c5642b4d9caee3b324a78f16f8 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0a80394bf9f45d4905e069638d55d2e : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d2d704551164b9b948f540cb27c8efb : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1a12f8dd68cc4114a1d897ed4b9284b2 : SUCCESS in 16m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/24f3e22827114a6d8c9e15b0f30337a7 : SUCCESS in 18m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d188bdc5f0444b6492e6700033954fa5 : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7a6cc09acc134969853d7a56677c6ce3 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/97329fa577d44bd2832cbdd9bec23f77 : SUCCESS in 20m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/944df8c48f494716b8f3fc85061e3d8d : SUCCESS in 16m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/793b475c97114018a267a20e939d6aed : SUCCESS in 58m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e86d1b4d62cf4cbe97a629ac7757ce23 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/896dfae345cf496681da12544701b375 : SUCCESS in 57m 13s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d89a66ef4b64e39bb63f3493e5b749b : SUCCESS in 39m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b4c6379cf6aa485a9ded89baeb2fb963 : SUCCESS in 1h 20m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0f68df5cdfc743f0b37e4c340c591dfe : SUCCESS in 1h 23m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c46657f41f034559b6d915b60b600c4c : SUCCESS in 1h 12m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/beaf144314394c898ad1aa05d9521982 : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/213ecc14e9bf43489322597e1fa8dc06 : SUCCESS in 1h 19m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1ee393cf90d4011895581d2735bf46a : SUCCESS in 2h 11m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b9d13f1ebbf94cbd87350b69a84ec6df : SUCCESS in 19m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea166862ee004f6faba6503f8a9089b8 : SUCCESS in 17m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ac9774c4e53148359c479bf3b129d398 : SUCCESS in 27m 50s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/16cde56617534ea4a384ee09df7de59e : SUCCESS in 22m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/396b45a309d34f0097707b3559af275b : SUCCESS in 21m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8796f882a984528a784c0eee4975d42 : SUCCESS in 20m 07s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e7f8c71186664032805b4414cd09a2e3 : SUCCESS in 41m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ae8835dc91e7401abfbdf8b96b6edf7d : SUCCESS in 21m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ead3c89714cd48708055f9e0c0e82909 : SUCCESS in 21m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff0a7122b6a8484e9367583e862227ed : SUCCESS in 17m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4f57a39a4e8494bf967a8dfbcf77c600659002c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 14:03:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"634a85212fc3a5062c0dbb62e9d96b26f6697caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 16:28:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8bb7f7a861548108f31809e2b9e66a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a83e42cc36ab4e72851d079be6331e99 : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/baadf5c74f1043d19f18a73eea4172ea : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831210711a74487483a541689e2a4005 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25b2662660a34db9bedd4cb2ce3ebae2 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/caa830a099134fbe9bf80586b3e78e69 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b6b901e193149229a4ca300fbc1dd69 : SUCCESS in 6m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9352ea3cbc4452ab68c92605a654023 : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/074e70edb08c42d19aa6eee5099bee33 : SUCCESS in 22m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fddb99ed39264d089afa5527a3f9381d : SUCCESS in 21m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/952cfa5a1c07493fb607b897c837f89d : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/14a46309573e4dd9b4607f2d15113b23 : SUCCESS in 10m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/073f032abf0242f798ccdcd72e97b381 : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed9e41351d63416391690211d2f5bb47 : SUCCESS in 16m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9167a84096944229a0b4268a00f27f6f : SUCCESS in 41m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c74df8bf355a4b0c8031ce81c1c021c7 : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7b48186ba267486ab857c6bfd96b1d5c : SUCCESS in 1h 23m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9bd47edb5b19471fba67a5e4dd1fa933 : SUCCESS in 1h 13m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a9f0fa6997da4b64875bce8a76d994d8 : SUCCESS in 1h 09m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5ebdfa0e30754e4ea9e83616535655ec : SUCCESS in 47m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/431dcc5384a245d2bd6c7d0cb32c118a : SUCCESS in 1h 13m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7fdbfdc690e44077bfad0cf56f792327 : SUCCESS in 1h 11m 15s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/24acbc4513684f4283a88341ae7cb79e : SUCCESS in 2h 20m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5e6bb75fddb4325922ea914a8539c8d : SUCCESS in 2h 07m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7b09a5e82eb3435ab47ea1e8efc70063 : SUCCESS in 15m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54be81770b7f45df9cabe8d096ce7de5 : SUCCESS in 15m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8f09fe2785624608bbbd722d9c3fe8e8 : SUCCESS in 26m 05s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ea1ef413ced343a2a2dea5790715e20a : SUCCESS in 15m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/069f43c7c1394a7381c4d5a975f54b09 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/534e1eb2da8e4161814fa49f9bae5bb2 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/50551c9d8d8c4d85840539cadda25ef4 : SUCCESS in 38m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0f99560bd5d64ef9b73f9964da0ddd26 : SUCCESS in 17m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e22ebaeb7073456f917714f06246eee8 : SUCCESS in 18m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80795eb1f32f4d41bba56bb5908bff1f : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4fa681b5a635005d35d6b2e600bad53a7c1d3994","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 16:28:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af28d2ffc51155550a4f44950fd6a8fcdf06e183","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 16:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9441de0a24e34e92902589074916a56e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd08fe6f6b5407e8fd452e4e8269553 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c403d7bd17aa49e8b5cb6ac9a20dd058 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"143c72c6220d9f4fe2556a5902a4472755a0054e","revisions":{"143c72c6220d9f4fe2556a5902a4472755a0054e":{"kind":"REWORK","_number":1,"created":"2026-09-08 11:40:43.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/86/1004586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/1004586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/1004586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/1004586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/1004586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/1004586/1"}}},"commit":{"parents":[{"commit":"7b31ae9dd595288730d7c008bba0948735ff8b9c","subject":"Add NGS support for Cisco NX-OS switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b31ae9dd595288730d7c008bba0948735ff8b9c"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-09-02 12:02:21.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-09-08 11:40:43.000000000","tz":0},"subject":"Add check and diff mode support to nmstate engine","message":"Add check and diff mode support to nmstate engine\n\nThe nmstate_apply module declared supports_check_mode\u003dFalse, so\n\"kayobe overcloud host configure --check --diff\" skipped the network\nstate application task and reported nothing about the desired network\nchanges.\n\nCloses-Bug: #2164642\nGenerated-By: Qwen3.8-27B-UD-Q4_K_XL(initial review Opus5)\nChange-Id: Id9d4d2643b0faf04d83553e289a40d2de8f00c53\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n(cherry picked from commit 2b7c7ad4ae3a5a6bab71c47f9b0a6a9dfaeff5cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/143c72c6220d9f4fe2556a5902a4472755a0054e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/143c72c6220d9f4fe2556a5902a4472755a0054e"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","ours":"7b31ae9dd595288730d7c008bba0948735ff8b9c","theirs":"2b7c7ad4ae3a5a6bab71c47f9b0a6a9dfaeff5cb","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
