)]}'
{"id":"openstack%2Fkayobe~1004652","triplet_id":"openstack%2Fkayobe~master~Iecd6cc3c6ee521b166e64a8b3dc639ad42946cd3","project":"openstack/kayobe","branch":"master","topic":"net_con_check","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-09 09:29:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2026-09-18 11:19:20.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-18 09:20:55.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Iecd6cc3c6ee521b166e64a8b3dc639ad42946cd3","subject":"Optimize network connectivity check","status":"MERGED","created":"2026-09-08 14:53:48.000000000","updated":"2026-09-18 11:21:15.000000000","submitted":"2026-09-18 11:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004652-net_con_check","meta_rev_id":"de42bb092a8f19f609a525d719b3212232b446d6","_number":1004652,"virtual_id_number":1004652,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 11:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-18 07:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-18 07:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-09-18 09:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-09-18 09:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-18 07:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"date":"2026-09-18 09:23:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 16:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 17:26:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-09-09 09:29:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-10 14:09:10.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-17 11:12:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8c5db0eae3180bd7be0e9882045d24b38d8fddce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-08 14:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab721bd4f19cb10ef697064867556a3cd79683ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:47:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01a2cbdba3ce4ec3a34e6d5f3b0b4cb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11de7adf20fb4ab2900e48859438a7cd : SUCCESS in 10m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f9d65f83bc39444c82d0e93b7d787e1e : SUCCESS in 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a42c942f2e942a587cb09ad429803a4 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30a990e43a44a378fafb03ebe19422c : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79afabed16ee4dc18f91fe0dbfb3231d : SUCCESS in 5m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c49691f71f2f410bb378cfe34f51d8a2 : SUCCESS in 4m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29408d9b75cd41138e3c6154a6b93e5f : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0b496de016a46cc9c2a6bf152c0f58e : SUCCESS in 11m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/227fb3ce72954d329e5b8abacea00122 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b13bad126d12470f8fed7ab549b1af9d : SUCCESS in 13m 03s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/09f748333ab645eea087f38361774536 : SUCCESS in 23m 07s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f57c9c4a2a6946fc90bd22288003ddc6 : SUCCESS in 28m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e0ce1c45049471c82375b538aad989c : SUCCESS in 24m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65eab1d70f9a49238676f1ab8a39ab5d : SUCCESS in 25m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d64adda3541c46f7b9584c5b51fc3287 : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a41544b39024200816872b50f4d6eab : SUCCESS in 31m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c3978360d74047f7b87d103e1065067e : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34ffba58668d4e72b8749eacb3a8be8f : SUCCESS in 29m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4bec734e38504f35b780876b7552272b : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c8f54d3171c94144933ceee4d92e74b2 : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9972e10e0186434086632769cd30071a : SUCCESS in 1h 12m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f82cc80a5884b5cafe48330e1530fde : SUCCESS in 1h 39m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32407fb251bb436fa27adacaf67d4275 : SUCCESS in 1h 35m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9df7ff07f1f94cdcbf8dcb86ae288c08 : SUCCESS in 1h 52m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2a78caf45e424028a464191ced50356f : SUCCESS in 27m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e641b454cc94fd581995ceeeaee4d09 : SUCCESS in 22m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62a1623fb3a94ea1ada60f1b98988c2a : SUCCESS in 21m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/09154e1bb6734a0c9ece7751f7ee5d1c : SUCCESS in 28m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/079bfe62d03a424eac8567a4061375e2 : SUCCESS in 32m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/395f33e1f72740eda51ed5b80fb59546 : SUCCESS in 43m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1cfc77a7e4854357a4ead405014b29a6 : FAILURE in 30m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5ba1b5932714c0782d1be4db1885d61 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"22ba8a284205f884a1e50133fcd63b428388224e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-08 17:01:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c0b8d72d6dd5fa37a9abdbf73e4e127ded68935","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-08 17:26:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"52b2d76e1c0b383035398d767de674ea33881447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:54:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9b6ea9f2fea42d4aec580f0343a02a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1a36af03954b4d9899f29bc4785958 : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7fc783f0460742f58fbafa661be98646 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7782bb574fa4267aea926702757c5d8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dfa6b8b30ea475aa0b42c2b2af924f3 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/588a6870974e4c698e002c5e8a98ca7f : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbf613150a1049fcb6385d88e2d49713 : SUCCESS in 5m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b2f3b67250d44d783d0fb39064162cf : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83abde4bc5dd43dfb379f1c6b045a097 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4dfd9e78e5ff44c4bfb87b39dd98f46e : FAILURE in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47bdf9a379a949328d530f7325c3c2b2 : SUCCESS in 10m 21s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4a1278486e23407699c1779024ff974a : SUCCESS in 13m 55s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/060499d73cac488dbcca7b354700dc48 : SUCCESS in 12m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de5f61948ca4ca9a810e6b52040108f : SUCCESS in 17m 55s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0de4e25c10204d40b0dfc6403eadd5ec : SUCCESS in 21m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca1e8d69b29347b4a1b03bd9419da54e : FAILURE in 22m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd347bc15e574406b419923d4f6af3eb : FAILURE in 19m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cb64b13ef48640e9a1ed8d24268b6ae3 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee3d3e3f42d046498124467618b381b3 : SUCCESS in 23m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/352de73def5c4d8fbb5ead127abd56da : SUCCESS in 1h 00m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d8b3d4852514dc3a29cf0085f9d55ca : SUCCESS in 58m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/043b3e44c592426a8024fa0a3b29d8db : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e6c5cfa3b59442faecba1f13ebbb1c9 : SUCCESS in 1h 32m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a499805ffcfd442eab518925ffa06062 : SUCCESS in 1h 36m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f42c3de71d00416597e83e00617e84be : SUCCESS in 1h 51m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/058d021fc3f7461a92f89e97ea7f93fa : FAILURE in 1h 00m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b70097d63e3b47e8a0d521ab96f18043 : SUCCESS in 17m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ab07ea4a9f843df923ff7e768e3a579 : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3206691e4c9b412b9e525f7422cce67a : SUCCESS in 25m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/89fb7fbf1e3e4ed1aa7a21c3593ab457 : SUCCESS in 24m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d2f7a95c3824c0887094d2b6e8ed12e : SUCCESS in 27m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/80ffab8d96c648e4a982ddd468222e4d : FAILURE in 20m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3bac02ff85740e5968682684cd59708 : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7c96a9f423195de88468551ce40a71558b5a1d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-09 08:45:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"309fd3b7e00d8c7c216d72a6ba451365c2924a80","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-09 08:45:30.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7354504f5a1b68e08bfdbf59540f3c221b9d2dc9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-09 09:29:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fb9422dd678028c77c706b68c22ebc6446e8b538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 10:35:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4129e1dee33143e18e4e0fcf567de927\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e498a4bcb574e02a8873472894e4418 : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6b3190daea345b7aaecc847313721b7 : SUCCESS in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/621e8ea40c784433af9eea1b6e0f9f61 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b04013da0784b1f9ec04c1946469a2f : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bcda98912b6492194607e82ac53815a : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69aa79d06da14d8fae87083459a7b0ca : SUCCESS in 5m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/858b7c4c7b054eb7806c2b321ca3caec : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b838b0ec9804d46aba3f74af585819d : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ffa9db808d44427a877cda6363219ecd : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a01bcc72be7f47e48c3a30114762c718 : SUCCESS in 12m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/64be73453f4d41a3a8c2d13d74c9fdb0 : SUCCESS in 12m 03s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e4864a1a5ad94e68adff6cf450191aaf : SUCCESS in 18m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c4debd11a774b7c8061c0776fdfbc33 : SUCCESS in 18m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01d587d68c114caba003194014389ced : SUCCESS in 10m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f7913321438b4c39a29e089877eb123f : FAILURE in 14m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19904c63f8e945daa38e435bef2966ef : SUCCESS in 22m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f9dae67515a409ba02c1a005d54beca : SUCCESS in 24m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b077864383644dca1ee7611237ba584 : SUCCESS in 19m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/32bf6d91d6f446c9a8bde01dbfc3c2cc : SUCCESS in 59m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1376734a2fb84a65b951ab8fec8f846d : SUCCESS in 1h 01m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/679e0e9f7e034b54a12c992203e06106 : SUCCESS in 1h 09m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdb7f243ca14438da0ee1797c0389b51 : SUCCESS in 1h 29m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/23d76c8fa0394e9391360f72ebbdb072 : SUCCESS in 1h 23m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9eb63cd39531400d86219dff96cb33b2 : SUCCESS in 1h 49m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a1aa323cf8134571b1dcc6a35bb59e21 : SUCCESS in 27m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/114c1a054fa94cc19d06bc9621298a6e : SUCCESS in 16m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5d3e94fee7e4188b59f232ff1e061ec : SUCCESS in 23m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81f844918a84483f9f4829541c298596 : SUCCESS in 22m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3df93e68563f4d4997f3da486d471bf1 : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d4b0d5837104dbfa6bdc2e9e90fc6f4 : SUCCESS in 39m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4299c47d236d4e89825382d3843f002f : FAILURE in 22m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82ba8a1be11c4bb5abd2b3321eea1c63 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a4a7b6ceb5a9fea5b3959b1e28ea5c179a1f41eb","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-09 10:38:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"854089d8bdf70cee39b7206dbd5c53dcdb440631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:35:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7168023f98ce495487d32beb64d27373\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5222f7165a0b4ec7a48560220a6495e9 : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a76729ed35bb41ff84f6a7d8ceb4c44f : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c7bb50afbd540828c7aa91f73944928 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a879bf13763f4be88af7b3fa128d8a05 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/791f6267ed474a0994616af48df14d4b : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b1dd3bb2791445ca08988af85cfa3c9 : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a5dc606ee4146b9bda091211c8af2f5 : SUCCESS in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c4855a7c2284510bf60fea408d6f97f : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/836d04d34be2490ea3f6cfacdb3b853f : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bfae29d2168d4f11867bea73e1df2828 : SUCCESS in 14m 20s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8943e80e201f49749f320902cf23ad52 : SUCCESS in 14m 54s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/08fc6cf65b9c4d4e9d00c85bca7d6c62 : SUCCESS in 23m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa8cc4647ca34a2ab121e9d401eca13b : SUCCESS in 18m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/049d409348f946918546365e08261c91 : SUCCESS in 16m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e92c6f63674d42fa9d5c4dede9d16be8 : FAILURE in 17m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e25c7c91ce924ee6af05c338c62fc8c8 : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0617727ea7944505b01f1a7571964ddc : SUCCESS in 24m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9a445c6b4574a718461cd442f779315 : SUCCESS in 22m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1533973e1ddb4298b886b4ff45991e19 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/02b87bd147ec4c9ca7c3f543d84405d2 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6146a16022fc4ad49983e5e67fd3bc44 : SUCCESS in 1h 10m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbcc6f3b836b4c43839ba6a28c23c8d0 : SUCCESS in 1h 24m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/abd1360e374d499a88c560163e1b1cec : SUCCESS in 1h 40m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8c8ebfd676ee4452b89ef750b3df1523 : SUCCESS in 1h 55m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dc0914dd3ed543298c97fdd777d972b1 : SUCCESS in 23m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/79f48cbd572941eeac4d441a5cf5cf91 : SUCCESS in 22m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b484722b4b6b47c8b25951905ca944ff : SUCCESS in 26m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7043fda07f5425a89df7c52570f4bdd : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c87fc53d0fd54c1583e36601607c7379 : SUCCESS in 41m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b78cb5db842457f8d8e292db5d8f4ec : SUCCESS in 32m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/32185e705172433183c741be39255199 : FAILURE in 17m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1203b023e42d46828bf45215e129dc36 : SUCCESS in 16m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7833b880f2f74f10045fb7e91111cad97ef36195","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-10 08:57:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27e254ad06a2cf379540f1fa2e8465c5e5f154f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 11:11:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/617e2e3c61e846c898db361a24297a43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92324ba6f1614803821de6ac185c934f : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23488209347445bc846dfb75665d0d50 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c23894b16e09442eace57c3edd86f9fa : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b77faa3c054c4bc58255c01add937216 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a937ffe93cc24a2f979617446b876e12 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d456ff51fe2142f193185376acb36b90 : SUCCESS in 4m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5f6f18a2d654a4fb08c19b7f763594f : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ae7eb3fe17547899a244bf5251e0351 : SUCCESS in 10m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4dfe8c91366b4572b50f7775611dbafb : SUCCESS in 16m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/99f47984ccbc49438d0622590e460511 : SUCCESS in 15m 50s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/feac7c5bb58d46289482e08f5fdfa5fb : SUCCESS in 20m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/83348d9f0b674d4ebc8804b535dc817b : SUCCESS in 22m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f666b46be8d42c08fc7cdc407486c45 : SUCCESS in 22m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a283cc99a57d471498eb0b14bb7e9eeb : SUCCESS in 20m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe50a73e28814d7f89465cd74e46296b : SUCCESS in 31m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6b34f754e7f499993a496cb32a73881 : SUCCESS in 31m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e0cff0b4470148dea658c53603f44b86 : SUCCESS in 22m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/900cc7b37f46433a9d0e2456e59ef846 : SUCCESS in 30m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/38d01fbb56bc499688c6c61b79c3d2fd : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1bd215c807eb4a34b31b4c0dcd1bdbda : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0f7216cea95e4fa88c5894d8910dbc51 : SUCCESS in 1h 14m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e50782ac467e45ad880691933a60fc0a : SUCCESS in 1h 51m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e992badb173a4be99219216f6cf7bb2f : SUCCESS in 1h 35m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/64425e11f0f74dcc85c57b5bf0809bd0 : SUCCESS in 2h 12m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d6919c5fe6274b9aa26e9104da1d1b66 : SUCCESS in 24m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c6b8a108a4e742438173a1bae9b1d295 : SUCCESS in 24m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d02852a019104037aa7a687fbde6e36d : SUCCESS in 34m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/004d613bfe4e4b6e9bac4eef38b82f13 : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/97a4eddd0fdd48ad9fe416afe524c743 : SUCCESS in 47m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66de3fd37b764682a75f68895b30056a : SUCCESS in 43m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6db77024c62540c7ac8e6ebd9caf8668 : SUCCESS in 25m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e58c7d548c2467da8fc42e8d753c2d4 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3f89496692bb14caada7393c1bada9988952af2c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-10 14:09:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1bd6bafa96236c1bc1a0d882a5900f97dec7b5b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-10 14:11:37.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"380fa36419b20476ac72d0c287e4dd91ec1e897e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-17 11:12:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b96d14e167eaaebabc17f6b5747c38fcaed623f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 11:12:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45f72adff4b91b5eef6b552aca2ab10fac346af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:00:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f007b27c08364f66bb372773360d6d9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ccf6e690d24a2eb8687b77df19c38a : SUCCESS in 9m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7e5da8ea6e374205b375ca37ac9aedca : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3ea1477bf8456e902b4c285a662835 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3ca49dae600430db5b80456f0850b12 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb9c7a8ddfb647ad98ef6dc9b61a5e9e : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec5b2c75231e49e5915616aeab807187 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83ab8fb5d8cf4347acc9694226494fe9 : SUCCESS in 14m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1369402b40c436cb4c3c074da85b676 : SUCCESS in 12m 19s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/60861203bdea44fa96d5da6022541d38 : FAILURE in 11m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55c7a1fb8a6a40f7939b143df7ddf5eb : SUCCESS in 21m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9e228b9e40541bbb75c15f699ae4035 : SUCCESS in 27m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de3a0a73989d4dbdacd35450dc1070ea : SUCCESS in 22m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ddb04f33d0634420a1f8edf8ba47b436 : SUCCESS in 25m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b95b9052aea4885a82a8dff4a1784a6 : SUCCESS in 23m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2ac4639327ef40e19bedf50fc6728156 : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/eb92eb8a36084eab86a9c386b741e17a : SUCCESS in 50m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d9282af911f0462eb161898771d3f087 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cbf18f3652342b48c8803d0882ddc46 : SUCCESS in 1h 48m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b25f189ba9242bfbc0dc1d59bf1fad0 : SUCCESS in 1h 31m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b6f667f837a4115b9e7570431e51977 : SUCCESS in 1h 53m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e3c09bb34f8403ab860eaff1d7d36eb : TIMED_OUT in 3h 03m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/16a7d2712f0e45529d51cc714b769b6f : SUCCESS in 32m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0be4e69339f54c0384e682ee7f55bbdf : SUCCESS in 24m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3569d1dfd7794d29a0687f15193df1a3 : SUCCESS in 20m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b85f6bf9a0344d190fdefdf3d745380 : SUCCESS in 23m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/06618c4e8ae84febac66eb34d22afc54 : SUCCESS in 32m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d6583f789944b05a5da7fefebae7bf8 : SUCCESS in 40m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82f5311f65474ba6856356987480fc0c : SUCCESS in 24m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78334493948f4237ab034d0eb3960d66 : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f91028cf8cb62d4c562d1b49dcc8d4af5cbb7ddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-18 07:27:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"59d028571e65c155a4ddab254514a72638dd8263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:17:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e01be6e57024a2bbd622d8719d0e587\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927af75268e449698aaf6bb910a0c6e5 : SUCCESS in 5m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/395ff6ae929346258f753e7e78989122 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47db897b9b844cd78e204e591a1ee7ab : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3bb1f1ae3a4bce8060575fb30d6f4b : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/709e5c7a7ea94e63b7dc98b25930176a : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/511155b3157541aca8432e58775dcb41 : SUCCESS in 4m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d29397e5bd4e6fa5147bab4c71f103 : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/280afb31bad44d3eabf3c401d6a2b51b : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f788048ed73549a1a43f62300b7515e8 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c55744da267479291082386db5a8592 : SUCCESS in 7m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/272cd1ade01f408d8d12de7e461358a0 : SUCCESS in 11m 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a52ee381cb5747f5afaa76bb1db4d39f : SUCCESS in 11m 59s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db216c0d704046b5a5923a4073c6da20 : SUCCESS in 10m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d4fe45c0cbbb49b0b345b139cf441a61 : SUCCESS in 17m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6c64aa6aae6d4017b82f253b5439ae99 : SUCCESS in 13m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5a1000510f344d09e01e85e907c6e8c : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a5c7139077ba422c8ee9f4f191f7393b : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69727e57e2134d58bbdb2c2bfae08e5b : SUCCESS in 22m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f6dc6551387d46ec8d7b6a8e0c01578e : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b20c7bd3d09c43ea91e5a889dc4934c1 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/84dd0a161e384113a79f4fdfa3b5cc18 : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/332e4780512546daa714338efbc74861 : SUCCESS in 1h 17m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab979325736c43c1a1b0b87fd30e7d06 : SUCCESS in 1h 44m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/732c396f026f4e978893c766869d31cf : SUCCESS in 1h 48m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8b50b574995244a2b8224510f5f6691c : SUCCESS in 16m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/750d13e9941a49c297b22da5a8d6150e : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4a064ddd57a4b85b2edbb48da5f21c3 : SUCCESS in 23m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3634b808380f44b89c0c42a45137ac52 : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bd5700505fa64312b518e9f6e28617d0 : SUCCESS in 23m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f8a2a050f2b44f3979328077d14265b : SUCCESS in 38m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ea4aff1175e4a29974cef2612cfe81e : SUCCESS in 25m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb31fdc0f06948aca37e078e65e77184 : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":3},{"id":"d5b7978861512372342f71d9a5dff1e4877ff706","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-18 09:20:55.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a18fc1d8213ac48ffd47a0d6e5d872468bbb259d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-18 09:23:24.000000000","message":"Patch Set 3: -Backport-Candidate Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d793abd40dc4bf2b53cbf5310c42211c2269affe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:24:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea35cbee41d8a6d989f341dc8b44751b769c576a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:19:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6472dd250f1e4b3287f7c0021c960ee9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aaa82b5b2504579af9debe4926f60fa : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da30c873366f4d92a95a1966c39977e1 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c93bc4f68e49a2ab018d228a80767b : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd056ec628574a46919e0d0882549ecf : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51cff171631a45029bdb9dc5d5bdea77 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b86841ae01c14a10b9ab10f45cedc7ef : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b1f5a679df449c5922362c87c04ee2a : SUCCESS in 13m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0e4598140cb4bc5a64feb267d1e0396 : SUCCESS in 12m 34s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/31ba388fed924fa4a8b8ebc7569c9995 : SUCCESS in 20m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fb150387c0d416d8fbc2ca192d1a60b : SUCCESS in 18m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d9fac4cd18db40518a88060c4e8737be : SUCCESS in 29m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef885f9c57f642dc9f5aa366a191c889 : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aba1e1bf3cd546ad99ba452f008de42b : SUCCESS in 20m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38468956b75d4e429705a7b5e5958365 : SUCCESS in 21m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/32ef39f39dc94023b9cba76ac734f0ec : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f7699d388a2344b391dd03071c6d3b7c : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/146d7a97660140b99a4484c524438bbd : SUCCESS in 1h 09m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0652ad2419242f28467b2b59d2cf80d : SUCCESS in 1h 23m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3e0a8db715f47159905040252452aff : SUCCESS in 1h 39m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/94dc4e73d91845a3bc9489856453df78 : SUCCESS in 1h 53m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8c2c8704768a454289f7f8e2e5e663df : SUCCESS in 30m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/55a06de707ee42ffa35b4e91f2018eff : SUCCESS in 29m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/563f9e61fb124ceebb8f19042d924599 : SUCCESS in 26m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7f6da399009847a58ade71053f022494 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/19dd68bb40734155b79c0b5bd8eb6996 : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15c631ecf8464d59aada029b64334e67 : SUCCESS in 36m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/99007a0247254485978aca670fc69da3 : SUCCESS in 28m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/050054b33ab843e2bdf745ee0546ab59 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":3},{"id":"09b721ea7a8ba4b73af016b170704f60e854c551","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:19:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"de42bb092a8f19f609a525d719b3212232b446d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:21:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25e5e2c0d38494d96b39e95b72a65f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801f0dc4c24d4cb6bd4360d5ade86d1b : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f7e0b7533374cd899da382f41044a62 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3bfae075b47ab4cdaddace92817a01297a23a916","revisions":{"472fd06cce8b7ab7866a70fb648973849773954f":{"kind":"REWORK","_number":1,"created":"2026-09-08 14:53:48.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/52/1004652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/1004652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/1"}}},"commit":{"parents":[{"commit":"6375e1e967a112babc5f196e47058517372bd1dc","subject":"Merge \"Bump up Ansible supported versions to 13.x/14.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6375e1e967a112babc5f196e47058517372bd1dc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-08 14:49:54.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-08 14:53:47.000000000","tz":120},"subject":"Fix network connectivity check optimization","message":"Fix network connectivity check optimization\n\nAssisted-By: Claude\nCloses-Bug: #2162831\nChange-Id: Iecd6cc3c6ee521b166e64a8b3dc639ad42946cd3\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/472fd06cce8b7ab7866a70fb648973849773954f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/472fd06cce8b7ab7866a70fb648973849773954f"}]},"branch":"refs/heads/master"},"b6150d2b8270fc9593a651a3e9ef4694dabbb900":{"kind":"REWORK","_number":2,"created":"2026-09-09 08:45:09.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/52/1004652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/1004652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/2"}}},"commit":{"parents":[{"commit":"6375e1e967a112babc5f196e47058517372bd1dc","subject":"Merge \"Bump up Ansible supported versions to 13.x/14.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6375e1e967a112babc5f196e47058517372bd1dc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-08 14:49:54.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-09 08:44:52.000000000","tz":120},"subject":"Optimize network connectivity check","message":"Optimize network connectivity check\n\nAssisted-By: Claude\nCloses-Bug: #2162831\nChange-Id: Iecd6cc3c6ee521b166e64a8b3dc639ad42946cd3\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6150d2b8270fc9593a651a3e9ef4694dabbb900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6150d2b8270fc9593a651a3e9ef4694dabbb900"}]},"branch":"refs/heads/master"},"3bfae075b47ab4cdaddace92817a01297a23a916":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-18 07:27:37.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/52/1004652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/1004652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/1004652/3"}}},"commit":{"parents":[{"commit":"c0a6eda87f4ad3ee54358221225896d747f21806","subject":"Merge \"CI: Make Ubuntu overcloud upgrade job periodic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0a6eda87f4ad3ee54358221225896d747f21806"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-08 14:49:54.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-18 07:27:37.000000000","tz":0},"subject":"Optimize network connectivity check","message":"Optimize network connectivity check\n\nAssisted-By: Claude\nCloses-Bug: #2162831\nChange-Id: Iecd6cc3c6ee521b166e64a8b3dc639ad42946cd3\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bfae075b47ab4cdaddace92817a01297a23a916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bfae075b47ab4cdaddace92817a01297a23a916"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6375e1e967a112babc5f196e47058517372bd1dc","ours":"b6150d2b8270fc9593a651a3e9ef4694dabbb900","theirs":"c0a6eda87f4ad3ee54358221225896d747f21806","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
