)]}'
{"id":"openstack%2Fkayobe~1004773","triplet_id":"openstack%2Fkayobe~master~I1f2d440b44aa70025165d361e8a1d70a3be03bc8","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-17 19:24:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f2d440b44aa70025165d361e8a1d70a3be03bc8","subject":"Clarify default values for bootstrap_user","status":"MERGED","created":"2026-09-09 07:32:49.000000000","updated":"2026-09-17 19:26:36.000000000","submitted":"2026-09-17 19:24:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004773","meta_rev_id":"aaa33c92f8b35c5deebbf9253aa86e1155cedd7f","_number":1004773,"virtual_id_number":1004773,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 19:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-17 16:20:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-17 07:39:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-17 16:20:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 07:46:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-09 09:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 16:20:26.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d2c9a227d6956d8e74838bb4f8037b98fd052fec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-09 07:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a381d50e9f77d16cda57279bcaafd36180713bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-09 07:46:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"84e58847e87ebea6b48faa74c2e7576aaf97319b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 09:32:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab5fa2013dc2482a896fdf7a27738005\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc48536eb1d4267910a248f70e61980 : SUCCESS in 5m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/512e132406024945a216be409836fd2e : SUCCESS in 10m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cbd1b88d5b848d7ac39e022a7387682 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbe261ec22947c6bda3a1ad12a670af : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c361c7042e9c4410804fd21cc9e76590 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d935b1c481d424cb6075a4ca0346866 : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f296dae259a24baebc901ed953314c90 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ba3c8bb80a545d19c604ea35e87b27d : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25497323ea604a5c9be017467ad5cbe1 : SUCCESS in 19m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3368f85559754e06bf2ce5a6db312190 : SUCCESS in 12m 09s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/765f00514ea24358ae595e5d49387d06 : SUCCESS in 19m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/722ea67d0d28428685780b48e9ee785f : SUCCESS in 16m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c75c32701d3e47db90efd5ab8e613381 : SUCCESS in 13m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8dcdc075aa964f42a9dbe4551474333f : SUCCESS in 13m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b3674e352249404eab4d49cd7eac6e29 : FAILURE in 27m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21215d4014af41e39d44c768287dd871 : FAILURE in 16m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6480071c8c0b4a5d8c7590b48c083e6c : SUCCESS in 16m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1798cd0bd5cd4a3eb0650ba1ee7bb903 : SUCCESS in 24m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a067c389aa8d4e258a3ece2b2eb5f643 : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d52692f01d1a4ebb8eaf00487d1a1b33 : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/96a14bec7a374a2e8a5ba638d58b8ebf : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfb5c829238449f0ab57c83afcd1ccf5 : SUCCESS in 1h 46m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9018239c8a824e0aba570b3859cad2ab : SUCCESS in 1h 36m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5436dca50ab944559bf5d34dde4f9069 : SUCCESS in 1h 58m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/523de91e825a4afba80ea6250e5f7896 : SUCCESS in 20m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4e7f5cf15cf44ab6b2ec6418e03ffbc2 : SUCCESS in 21m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8baa5a37d9b94f69997012ecfb518106 : SUCCESS in 28m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a826c228090b49f78d781f9cea9ed192 : SUCCESS in 24m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/44324de8de5d43918e59b1c3704a7639 : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/035cdbfc9afa42cd94e2b2f9abe3f907 : SUCCESS in 26m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d414a692d274444fbc719b26103d9bc1 : FAILURE in 23m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb1b49581c8343708d550e094632b7ea : FAILURE in 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"247604edad36ff427c27a2c09072ff39855f19ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-09 18:58:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4e7bf9d516686dc351a3b5ebab5a3f775c5db26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:04:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ff09740fbf401b8fd32c037e8d08f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4fa4b70fc04cd4af407c934512caa9 : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9c06dc5fff204bdea0cf7cedabde6aad : SUCCESS in 10m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5443612ac0a04ac8b31d8e3fcb6238cf : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21144361075a475eab2b8629ee95dd7a : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9019b285b59148a980bb8cf939533d6c : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d4e4a90dbf0491bad73e15796f59908 : SUCCESS in 5m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/786b9aa608bb4a7b8cd576161705bb3d : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77cafe730e9d46549b648b14b3d7ce12 : SUCCESS in 10m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a33d34036c794b2f9a7298d91cc33a85 : SUCCESS in 14m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ead25146e2e246feb785f223ec917b61 : SUCCESS in 12m 30s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/630fde08ed6e46e0863e729965b270dc : SUCCESS in 29m 22s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/199e831a915a4be7a1ed5d82829ff1f9 : FAILURE in 17m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c7efefa4e784114bac3809eaffe476c : SUCCESS in 23m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3fffca240f3c48828fb65b92b20294e9 : SUCCESS in 23m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/69e8a8fabf134f6497fe4c4e53328f9b : SUCCESS in 26m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c73e160b07454b65b5c0ec617ee59d73 : SUCCESS in 29m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/60f706582f9f4b108235003e30ba1234 : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c930b0a79dae4563aeda9188aa609da7 : SUCCESS in 25m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1d2c32f3354e4ed99595484329b542b0 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/46927f8f9b5c4820a78ee0b88ae65830 : SUCCESS in 1h 11m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3d23c794a381427a94fce5970a6ad69e : SUCCESS in 1h 15m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2d51b35d6f5416999006845a2ac9031 : SUCCESS in 1h 39m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6edabec2cf5444a9536de281583ba70 : SUCCESS in 1h 38m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f496b1dd804e4683ad7f29a9ba95ceb4 : SUCCESS in 2h 04m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/768aa0d8e31c4802a0a93d7619cd84b0 : SUCCESS in 25m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a848a7d1831444fe84b0e7f47daacb43 : SUCCESS in 33m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/168b6b3f653e465085c3498e096f67f4 : SUCCESS in 31m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d2fe6ca7c6704396b11b65122f7799d3 : SUCCESS in 28m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/11183bc7233e42e596fc99dc13ba2465 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8157effa6b554ac08bd4f42f6632d265 : SUCCESS in 42m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cc19784967f44a52a2b352dbe161dd5c : SUCCESS in 29m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/852f3fbed88d48a0832240cae803ccea : SUCCESS in 25m 54s","accounts_in_message":[],"_revision_number":2},{"id":"363153600d6f0829f5a47edbd4724c4684877b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-17 07:39:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"696bcb610a6352025a2a7dcecb6799b664991543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:32:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00274246f9b94586baa995a47eeca0d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c0ec74c42b4d018d4e20d422f64760 : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b07973252e23468bb0d15aece87f7b6f : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0a4221394944b7abe9f256070196c66 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191cc7b687054964988d537b32f82a50 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b326e58abbf54bfbb6aabb010a6a2adb : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f27549d514bd471d928a0bf2fddd7814 : SUCCESS in 6m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/601fc7de8f69453894b9a9814b62d1d3 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06251f26f8724907a9c9682d4eedf030 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e7eb946b7fa543948d02247069464c55 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e0500ce2aa849c09a087d8000d800c8 : SUCCESS in 11m 10s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/11877fd9a6624ed1971871716a54f492 : SUCCESS in 22m 22s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3a73baf533824ff88f212b9957f00c73 : SUCCESS in 16m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c7bc9680f534926806f0bfd7dbc25cb : SUCCESS in 14m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c88545c19cba42e9b427006039611bf5 : SUCCESS in 24m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16e0aabcc9794c1caf3d5d4c8450c819 : SUCCESS in 27m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcff14d0e8484bf1804fea73477fb5d3 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c0ea6b2eec0345ecba6823a586c1d1a4 : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a4a223f95d640d2a57e12fce641b419 : SUCCESS in 25m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5ed865f07d714c1397068adc4b765b29 : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/429ef4ab7aa64522bc73f2ff129041f2 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e1b82314338d4af0b7772b558c65c9ed : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/165dbcdd5a6143a08a9272245c1cc56a : SUCCESS in 1h 29m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c2ed81e04594bffb25f1982fd6f30e2 : SUCCESS in 1h 43m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/feb48e131277400fa16ba304a1e522c3 : SUCCESS in 2h 27m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0bc78dfb0be84d10a569cedf955db4a3 : SUCCESS in 23m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/965d4a09c9174f15ace5c845396a4553 : SUCCESS in 22m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/498b1240ab6447d4a5f4f3ce3a706a71 : SUCCESS in 24m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de758cb4a4ba41268fea4d53d8dc1b28 : SUCCESS in 28m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/381434239cf542f1a5b251a9946c38c1 : SUCCESS in 42m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f00be89111f484f9d701b282b445392 : SUCCESS in 35m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/894ddd9c832d4db181e27eae9e56da4c : SUCCESS in 22m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfcac9382bb94626b1e4db41ca964ffd : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4c44d760f7ee99ad9d1fa7f9e7e250716f24919b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-17 16:20:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8bedfc4d914b0871f2e918637f4c3e494d145242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:21:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"848af2af479e4e785b93e76c6760ac4093a02d26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:24:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/046bd9365aa74d6a9ff804e8bd343cfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240d27e283124b718a037f89389f2050 : SUCCESS in 8m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4c4e42d43fc34063b5ce9643b46a93e2 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22cc62f3e34d47968deba7711366d4a0 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/166199dcfcb54d38912c1f4ca70a607d : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af415d4069874c8c865ca9409209f246 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bf557c85073430caff17475813557da : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25a8f79b849d42a088c76a0bd8807897 : SUCCESS in 6m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f137dd484cd648dd94cc455c0ee01f17 : SUCCESS in 14m 15s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9b82c88586ef448cb0b60c47303df397 : SUCCESS in 15m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7c85fd49e7b482eb75d93d50e5967be : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a61e560ffb949199e54d1be9b0b441f : SUCCESS in 14m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/361ee7b3727d44f4a65a70b823212f9e : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5e45f4254ca44e43a86e66a6e837ec25 : SUCCESS in 25m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f9f7b3cc97b402481d6df0081683255 : SUCCESS in 11m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/53633895c7a04454b04ae0492adcd2e0 : SUCCESS in 59m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b081ba5c757b4f018f612c7182e2a105 : SUCCESS in 1h 31m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e732843b121c4b7c981b8a64ca406d79 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0ca7e07c1574edbbd60af1023e3b9c1 : SUCCESS in 1h 25m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a5774be002f414aa376c3d20bc9d9dd : SUCCESS in 1h 34m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c96315f44add41a8b89c4a089a2a2eb7 : SUCCESS in 1h 50m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d26ea4e7cc04afebed5a0541d47fd71 : SUCCESS in 3h 02m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/23b0522b405946d1a86209a473dbf61f : SUCCESS in 28m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2af0e0cf03c415c810240d6ee6f2357 : SUCCESS in 14m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e205696d30048c98612aa7646de5fa3 : SUCCESS in 13m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50341b1d1f9b4a6f98d28e1f0ceca2c0 : SUCCESS in 22m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/496fe3754411438e916980126a00c2fc : SUCCESS in 28m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6eb2d2101a9b498b935bf62c2b55195e : SUCCESS in 35m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc0db80723b44de0b17f9d54aeac6db3 : SUCCESS in 21m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99f9e9cb19f44a33ba4663315be934b5 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":3},{"id":"14e9a3da7143960550950c7055fb98b28b8289ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:24:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"aaa33c92f8b35c5deebbf9253aa86e1155cedd7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 19:26:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6000053a051549738eadfb69ed3a855b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04afe305e8f84edabddb01189b15011f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"07353f93dc2e3549037adb17fd961bb84c6f9153","revisions":{"347ef4951d41e6bf01b974cbf404c51f2e99957a":{"kind":"REWORK","_number":1,"created":"2026-09-09 07:32:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/1004773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/1004773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/1"}}},"commit":{"parents":[{"commit":"e1e52c5b120dc04aef5eac3f2a42034b39e0a73f","subject":"Merge \"Add check and diff mode support to nmstate engine\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1e52c5b120dc04aef5eac3f2a42034b39e0a73f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-09 07:31:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-09 07:31:10.000000000","tz":120},"subject":"Clarify default values for bootstrap_user","message":"Clarify default values for bootstrap_user\n\nChange-Id: I1f2d440b44aa70025165d361e8a1d70a3be03bc8\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/347ef4951d41e6bf01b974cbf404c51f2e99957a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/347ef4951d41e6bf01b974cbf404c51f2e99957a"}]},"branch":"refs/heads/master"},"7651fa5a2629ac9a78c67681cffda88c6b94465c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 18:58:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/1004773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/1004773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/2"}}},"commit":{"parents":[{"commit":"9ff5475ae2c16ab2833830f73002a89026ef23c2","subject":"CI: Download cirros images instead of using cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-09 18:58:38.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-09 18:58:38.000000000","tz":120},"subject":"Clarify default values for bootstrap_user","message":"Clarify default values for bootstrap_user\n\nChange-Id: I1f2d440b44aa70025165d361e8a1d70a3be03bc8\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7651fa5a2629ac9a78c67681cffda88c6b94465c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7651fa5a2629ac9a78c67681cffda88c6b94465c"}]},"branch":"refs/heads/master"},"07353f93dc2e3549037adb17fd961bb84c6f9153":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-17 07:39:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/1004773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/1004773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/1004773/3"}}},"commit":{"parents":[{"commit":"c05292c250703a9583ea46ae8f972b1257247448","subject":"CI: Enable EPEL only when required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c05292c250703a9583ea46ae8f972b1257247448"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-17 07:39:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-17 07:39:17.000000000","tz":120},"subject":"Clarify default values for bootstrap_user","message":"Clarify default values for bootstrap_user\n\nChange-Id: I1f2d440b44aa70025165d361e8a1d70a3be03bc8\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07353f93dc2e3549037adb17fd961bb84c6f9153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07353f93dc2e3549037adb17fd961bb84c6f9153"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
