)]}'
{"id":"openstack%2Fkayobe~1004785","triplet_id":"openstack%2Fkayobe~master~Icabdca7babf71ace60c1db3bbbb3c0016d6df312","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-09 10:57:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Icabdca7babf71ace60c1db3bbbb3c0016d6df312","subject":"CI: Download cirros images instead of using cache","status":"MERGED","created":"2026-09-09 08:46:52.000000000","updated":"2026-09-09 15:14:16.000000000","submitted":"2026-09-09 15:11:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004785","meta_rev_id":"da0f35b0b7b3948b9e2f0d210f63db50d6c8df07","_number":1004785,"virtual_id_number":1004785,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 15:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-09 10:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-09 09:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-09 09:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-09 10:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-09 09:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 09:04:46.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-09-09 09:31:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-09 09:36:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-09 09:36:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-09-09 09:36:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-09 09:36:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-09-09 09:36:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-09 10:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c05a3d7d6cf5437fc5b4a85c7570baac118a5689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-09 08:46:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7be191edd4b33b77bac8198947cebffb34d24b7d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-09 09:04:46.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c64fd8a2deacdd97b1b4e0ff6aa8ff8f0f4ad884","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-09 09:31:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"10bbb67e492bce59ff625bb0b584687a9205f1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 10:41:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aa21b96e9b9471ebe562a8c0646bf16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a45e9697de4fb19ed1effc8a5ed63d : SUCCESS in 10m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/02f709bf7a654ef89b9b4760cc18ed3d : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58708064efc84c6ca9c8b4d30732211d : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2809d60b12c34f00af6ad98e5bcaf106 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a53c0507e6b4d6bbd7b39e7da341ec7 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9999520af01d4fedad0743e9bdf7eaac : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6286ce4d5816472f8d7c6105dc0dd1f3 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6d5e5ca0c974349bb0494767970aaf7 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38555dc7a6c24ac8aaedb4a270770b7e : SUCCESS in 14m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee93253e829b4bdeb655fbb7314e8c9c : SUCCESS in 11m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dc248884083c4a53a56f72e489b87473 : SUCCESS in 17m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/417116a7c5a84077b0d12ff735bfaaef : SUCCESS in 11m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa28792c2ed0455bbf7e73d58cbdae4f : SUCCESS in 21m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bb8ac5a21d9e4b80a309ef691d46e1fb : SUCCESS in 18m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8fb42bdf8e4b479985351ff9f8f6319d : SUCCESS in 12m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b94a5c9ad3bc4044b62b68edae1faf57 : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0d38bd8043d0463b949d61019c5e48dc : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16c4223154314360ade4fbb481725138 : SUCCESS in 20m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/57b761169d2141b9adbe337dd35eecb2 : FAILURE in 1h 26m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a126d28fd03843ac8afa37fecbdf047f : SUCCESS in 59m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/63ee8d9f4beb473a8ad3def458afb27d : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df9da6215a91458f8723765d13f88d7f : SUCCESS in 1h 34m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/933ca03c925146b98e167bc27ef14c91 : SUCCESS in 1h 26m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1f34df122d4543e4afc603e0fe3cd8e7 : SUCCESS in 1h 52m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ce7247fe58f148728dd5b2c21cd72e25 : SUCCESS in 33m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e32be639c5a847c8873f7470fa6269d4 : SUCCESS in 20m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d345aa586d5c4953ace6a2553064ec51 : SUCCESS in 26m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/acc24efcae234a119c4d37bcee01bcd4 : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2b3a2db006824af78ebd18e09aed46da : SUCCESS in 43m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6406d289c3194185a45cfdadc3773349 : SUCCESS in 38m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1196724683c146df908957efa4b4e6f8 : SUCCESS in 19m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6e1e8e3d0e44c6d9bd29e80f8692ea0 : SUCCESS in 22m 03s","accounts_in_message":[],"_revision_number":1},{"id":"66be5e0ea8f2d2563f0dc8ea311c54f69d8436b4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-09 10:56:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8238a3599e16938b9a3cce15ee11f2c574a2a87e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-09 10:57:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f02663f75cc53825de1b2f2f1a39d938e2cd9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c2dda5bd1cd4b7f98e1b63ba2500afb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757e5395510349639e7ed765b25eb6bf : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5bb739935ada4213a4bb1e0a56696e8a : SUCCESS in 9m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c819c3eede9413c8babaf2c08087195 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b522bcf0ff11441fbb0765155e071555 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a0f3bd12f8c4f9590047e748fe10e67 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa0b43b8b95e4311b5914dad68266e5b : SUCCESS in 4m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5936660fb6274919858b18dfe7837ddd : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c771d78215a4d6eae718b423f659a5f : SUCCESS in 9m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69b4443ebba04d859f8209fbc2edc70b : SUCCESS in 13m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd4b139dd7444954a14c58c1e2d54325 : SUCCESS in 13m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df4af7dfd9a34444bf91d7b6a0de7712 : SUCCESS in 30m 22s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e3d785a77e5d42e48a6a6640a757c951 : SUCCESS in 17m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89a0a456a8dd4449885ecf1bf12a295e : SUCCESS in 21m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1b8f5c068b94f1fbbbec90a99d0c272 : SUCCESS in 22m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f02e9b3818324a0db4aa98652e5a43c2 : SUCCESS in 25m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af350b9de61949c29191fc9d858a30db : SUCCESS in 27m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/35d4634f1aa6464b88a5035bc2f50d0b : SUCCESS in 26m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0f9a908025d4a70bffb4df39b12e4e6 : SUCCESS in 20m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ddd45fd3e56b4aeebbb7c953af79e236 : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/989b7a6302c14ca3b591061933a5e961 : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/594fcac28c7943afae10b28e62f679d1 : SUCCESS in 1h 13m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812c6d5e107247c0958e63720e9011a1 : SUCCESS in 1h 37m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe86e56ccc724c468aaedeb3c4f69b83 : SUCCESS in 1h 44m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/77ad7287b23648b5ba68e80a68fa4cca : SUCCESS in 2h 03m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9b8b8afcc9a84b41b69e738ce04f3eff : SUCCESS in 23m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4fbb2b4af42c448ca4b45c47bf5a1686 : SUCCESS in 18m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9dd4eab76a04ce2bf933b4057ea8a4d : SUCCESS in 30m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c30028914b4424ea1c88305654b8bc3 : SUCCESS in 26m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6cd4e75c9a41495f9847dcdda03ac8ae : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4553a8f9f1f467ab57716a70f907eba : SUCCESS in 44m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9a3fd03a290d4f3d93bc81b921e4be0a : SUCCESS in 33m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e52a466c0c34501804a815df22669df : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2f078521534d4c4545720c24329fc63a920bc944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:03:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d669b6a2e4fc5b3f2475f5ed504ad3972df6076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:11:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75685fad44b4fe79f137d8142ea5734\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55eb1bb732b44f76ab74899f4a3b49db : SUCCESS in 10m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de5e0dc97f634ca2ab54e0560103d4c9 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ec9f6c7a9341dd8894a8cb6bb03a30 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cec31f62ec4e463c898c65d70dee84ae : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a4a78de3b91491a9ffba1d7d75a1815 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/08f8a6b0fc144cf88ab79bc64275a418 : SUCCESS in 11m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea9a04aaf18c4a4a8dde69d1bed061a6 : SUCCESS in 16m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ded225fb57c400188ab11e80de980d2 : SUCCESS in 11m 10s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/946dcc013eb44e139da15659347c5b49 : SUCCESS in 25m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ab5464d0f7d47cb9b76d07f919ff421 : SUCCESS in 26m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0ee8e894e1114c069aaec2a8d7fd50ea : SUCCESS in 29m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3dc8ab498164598a295b81f750cf488 : SUCCESS in 30m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/de4e50eab7904d5988aa45e094b8a9a9 : SUCCESS in 28m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/343938b7f1e042239b2312760aa345fa : SUCCESS in 29m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/222b397c0aff4856a0e307dbc2bd29ae : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/24750abf9f354ba887c8a5c724b3a9d8 : SUCCESS in 1h 09m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/57c1044d008e4cfc8c35731563ec6904 : SUCCESS in 1h 20m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5d7b3824ad7463b925d5253b8e370ae : SUCCESS in 1h 52m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/abc4427b49c344eebf93a06927a06851 : SUCCESS in 1h 47m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1806dc055f284f3d8caa08bf47f76ce9 : SUCCESS in 2h 06m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dd61f96acdd04961b158fef1c9ec083e : SUCCESS in 27m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f1359883962b47c689105df56bd32a01 : SUCCESS in 30m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b4509538d1448f883adf83be1b0568a : SUCCESS in 26m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1db4ca49f24a409ca50f33de62cd082f : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5deb381bd54497dab08f9f545f07fc1 : SUCCESS in 49m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d68b819c02294bd3bf602d81481884db : SUCCESS in 33m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/439bac7a4fdd4c33a01cd24d115c7777 : SUCCESS in 32m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f6ca246a6cd4be4934571921adf0a03 : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"47f229326ef8d1d95b357258da5d7271ec03b7f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:11:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da0f35b0b7b3948b9e2f0d210f63db50d6c8df07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c092e5eb944475889dc01b139cdf624\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7fefd29a8d54f69b8262bebd79e6c55 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ff5475ae2c16ab2833830f73002a89026ef23c2","revisions":{"9ff5475ae2c16ab2833830f73002a89026ef23c2":{"kind":"REWORK","_number":1,"created":"2026-09-09 08:46:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/1004785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/1004785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/1004785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/1004785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/1004785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/1004785/1"}}},"commit":{"parents":[{"commit":"e1e52c5b120dc04aef5eac3f2a42034b39e0a73f","subject":"Merge \"Add check and diff mode support to nmstate engine\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1e52c5b120dc04aef5eac3f2a42034b39e0a73f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-09 08:46:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-09 08:46:12.000000000","tz":120},"subject":"CI: Download cirros images instead of using cache","message":"CI: Download cirros images instead of using cache\n\nOld cirros images have been removed from the image cache [1]. Download\nimages while we fix compatibility with new cirros images.\n\n[1] https://review.opendev.org/c/opendev/zuul-providers/+/1000965\n\nChange-Id: Icabdca7babf71ace60c1db3bbbb3c0016d6df312\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
