)]}'
{"id":"openstack%2Fkayobe~1005000","triplet_id":"openstack%2Fkayobe~master~I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a","project":"openstack/kayobe","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-10 12:18:42.000000000","reason":"Change was marked ready for review"},"39327":{"account":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"last_update":"2026-09-11 10:04:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-11 07:52:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a","subject":"Install extra kernel modules on redhat hosts","status":"NEW","created":"2026-09-10 12:10:32.000000000","updated":"2026-09-11 10:04:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"02a57799d0f4e7b5b344d97cd0ebfbcbcb47eee0","_number":1005000,"virtual_id_number":1005000,"owner":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 09:56:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-11 10:04:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 12:11:15.000000000","updated_by":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-10 13:14:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-10 14:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c093ad93c4ad9867ffa1759a28555608ac016c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-10 12:10:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f2cb6337c4988c1a1a49806d8d5c53730800358","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-10 12:10:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"93859d9c9de8fe97d62569a421ff7f3d1ae7d5ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-10 12:14:41.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"898bac51a4912a92d012c691bed6be2f13f19809","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-10 12:18:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"9e9e9eacf1844d9ba8be7882b1f7fbff891b621a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-10 13:14:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7af642a34c6e2741bad25f14716124133436f667","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-10 14:07:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17c5babce4b85876294f391c6c8ff1c9225a9047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:30:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902ad05dbb924001b7bccb23aa55d50a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f87fa90d4a4474a91d62fa5f84506f : SUCCESS in 9m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af8120ae6b6844f4a9b38c8c91b60eaf : SUCCESS in 12m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c48388c4c1e4a00b1a84bdbd1ccd70c : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cddbcbbb4204c36b1ed8609ba9f50bb : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8f8966adf824c0e96c79b028e5a8716 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f5eeff0f01345d586694b86ed4dd716 : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf8adaf94df14af58ed9699db88a11a8 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4b41a51a12a4fbaab76d6693c07fb73 : SUCCESS in 9m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/741d4568624e437d9afb150af6b43d6f : SUCCESS in 18m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cd6236593ed4ce3b391b56369e7b1f5 : SUCCESS in 12m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/485b301c6bb84aaeb647a9c15d0806ab : SUCCESS in 23m 44s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/37316fff1169487293426d75956eaee2 : SUCCESS in 21m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb4c42f518544510ba65af6e38d06bd7 : SUCCESS in 26m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e4432dba3fa4454fb82e4ac2390c6357 : SUCCESS in 24m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/564f3968503546b38707a2a2fe3e759e : SUCCESS in 24m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecc5a462fdcd4b5ba7668d41e86dc68e : SUCCESS in 26m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/44023543987049dd9ac6dd6c2a0ff5d6 : SUCCESS in 35m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3239111330a440b79f2657424f9375be : SUCCESS in 27m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/50871c48fa154b8f9699af70e9a5c74e : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1f8b796bf8314abbb47454bbf11245da : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9989a796f994400bb7a1fb0d2648aa4d : SUCCESS in 1h 17m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6809e146e3d4def9caabc48a706306a : SUCCESS in 1h 34m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e2992ed6fb74b2e8d8e8fd2edd71ee0 : SUCCESS in 1h 39m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a562f7bbdc7f4eadb87d2fa9f4627b72 : SUCCESS in 2h 14m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f9db5c8cc89456f95017dc34388d66a : SUCCESS in 24m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/10983332e1b741fdb41d57a080468358 : SUCCESS in 25m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b5df3c7e25240cdb2aa50824ce64e5e : SUCCESS in 33m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b0dcc0b3dcb411cb843fc3a40e5453a : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/977eb33321a94ccea16b7e2239c2ca2d : SUCCESS in 34m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/388c815bd6dc4853b0affd003b002186 : SUCCESS in 43m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6609e3a2283f49db986f700bc2e0276f : SUCCESS in 32m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e6352387a45416cbb4b7ceecbd3f2e5 : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f6ddfe967d1472344268cb64d51c6a3888b00e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-11 07:51:29.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e0b75181ecd6a7016e5fca6cfbf798e95cd56b9f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 07:52:04.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"2260cf1700c2c6d17df480a92c04fec7cdefabd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 09:56:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad5dd340caf84f588c20f0aadaa1213f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cc1f4b41144ba58105635724f50866 : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/544ed80a9e1146f590c0f71686cf279d : SUCCESS in 14m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbeb56380b884820a13d1f87e135b714 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178bd29b097e48d88144c79eff2f988f : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/991b5b3a32744ed985b64db59b7505bd : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/340a477bf80f4dd69770ad00cc19edfb : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f431faa27c7441182cd83e5199f2b90 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/539d9b7b834c431db542f2ca79cc98e9 : SUCCESS in 14m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/457eae8883d0471d8d9a7e96ddbd9d33 : SUCCESS in 19m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f369810238a4498b8b8cdf3aa07637c6 : SUCCESS in 15m 24s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5210e5899ff1402a863be7c8dcfa4b14 : SUCCESS in 26m 24s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f23703e8730e45ff9109534113fefba9 : SUCCESS in 32m 20s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b41a0850aa6647dabde1984cbde849e9 : SUCCESS in 27m 25s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4fdbcee70f904862bf440ae815d14c88 : SUCCESS in 29m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/63310fc9d1bc42f2ba253b81bfef2617 : SUCCESS in 37m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45f6cb81174440b19d76f61a02d2c4d2 : SUCCESS in 28m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1e2c8e7ddf3d48ca88d8f63458d47409 : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74371215e32a4694952ede0657285267 : SUCCESS in 27m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7410069104ab4b5282dc95fd65fb4afc : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/31d6a06a9a654ff9a91cc9ff540f18e9 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/04b8b371b721404b84540001340782d3 : SUCCESS in 1h 16m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90ea8b7b5d374364b4e34fc7f771345d : SUCCESS in 1h 47m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea29c965157f45f5b7b4266f2827de5b : SUCCESS in 1h 51m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/735907e69b9047ffa19210b6982db8b9 : SUCCESS in 2h 02m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4c992068bca04e92bc1d55ad3703f269 : SUCCESS in 26m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a7c3b44e61b34b1b899a0205dfd85e12 : SUCCESS in 39m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb720360cd6545fcbce5e75ef82863b4 : SUCCESS in 27m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fc6cf32814b48a0800f99106fa2bbfe : SUCCESS in 35m 51s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7f7cac104bbc43c18ba08c67ef1d0112 : SUCCESS in 49m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0be3d77443c04f8dbd14952f14bd02f0 : SUCCESS in 37m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ffd0707169f47fd8c7f49921dacd284 : SUCCESS in 40m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/871f5eb8faae4726aaf11eab483d5e1b : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":3},{"id":"02a57799d0f4e7b5b344d97cd0ebfbcbcb47eee0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 10:04:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1f9834176430b877bfac65cfc0ae3433b3e9905","revisions":{"7ae079f0ee4a08be601b04987dbed4e0411c40ba":{"kind":"REWORK","_number":1,"created":"2026-09-10 12:10:32.000000000","uploader":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"ref":"refs/changes/00/1005000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/1005000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/1"}}},"commit":{"parents":[{"commit":"9ff5475ae2c16ab2833830f73002a89026ef23c2","subject":"CI: Download cirros images instead of using cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}]}],"author":{"name":"RowanJohns","email":"rowanjohns4@gmail.com","date":"2026-09-10 12:01:13.000000000","tz":60},"committer":{"name":"RowanJohns","email":"rowan@stackhpc.com","date":"2026-09-10 12:09:44.000000000","tz":60},"subject":"Install extra kernel modules on redhat hosts","message":"Install extra kernel modules on redhat hosts\n\nChange-Id: I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a\nSigned-off-by: RowanJohns \u003crowan@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ae079f0ee4a08be601b04987dbed4e0411c40ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ae079f0ee4a08be601b04987dbed4e0411c40ba"}]},"branch":"refs/heads/master"},"6aeae22dde8da8878088a840618b42813fc20962":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-10 12:14:41.000000000","uploader":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"ref":"refs/changes/00/1005000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/1005000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/2"}}},"commit":{"parents":[{"commit":"9ff5475ae2c16ab2833830f73002a89026ef23c2","subject":"CI: Download cirros images instead of using cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}]}],"author":{"name":"Author Name","email":"rowan@stackhpc.com","date":"2026-09-10 12:01:13.000000000","tz":60},"committer":{"name":"RowanJohns","email":"rowan@stackhpc.com","date":"2026-09-10 12:14:29.000000000","tz":60},"subject":"Install extra kernel modules on redhat hosts","message":"Install extra kernel modules on redhat hosts\n\nChange-Id: I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a\nSigned-off-by: RowanJohns \u003crowan@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6aeae22dde8da8878088a840618b42813fc20962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6aeae22dde8da8878088a840618b42813fc20962"}]},"branch":"refs/heads/master"},"d1f9834176430b877bfac65cfc0ae3433b3e9905":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-11 07:51:29.000000000","uploader":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"ref":"refs/changes/00/1005000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/1005000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/1005000/3"}}},"commit":{"parents":[{"commit":"9ff5475ae2c16ab2833830f73002a89026ef23c2","subject":"CI: Download cirros images instead of using cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}]}],"author":{"name":"Rowan Johns","email":"rowan@stackhpc.com","date":"2026-09-10 12:01:13.000000000","tz":60},"committer":{"name":"RowanJohns","email":"rowan@stackhpc.com","date":"2026-09-11 07:50:49.000000000","tz":60},"subject":"Install extra kernel modules on redhat hosts","message":"Install extra kernel modules on redhat hosts\n\nCloses-Bug: #2166461\nChange-Id: I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a\nSigned-off-by: RowanJohns \u003crowan@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1f9834176430b877bfac65cfc0ae3433b3e9905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1f9834176430b877bfac65cfc0ae3433b3e9905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
