)]}'
{"id":"openstack%2Fkayobe~1005063","triplet_id":"openstack%2Fkayobe~master~I851e7246030e1995a3cc94340e79c942cc66af2b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-11 07:57:47.000000000","reason":"Change was abandoned"},"39327":{"account":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"last_update":"2026-09-11 07:54:24.000000000","reason":"\u003cGERRIT_ACCOUNT_39327\u003e replied on the change","reason_account":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"}}},"hashtags":[],"change_id":"I851e7246030e1995a3cc94340e79c942cc66af2b","subject":"Install extra kernel modules on redhat hosts Fixes bug: https://bugs.launchpad.net/kayobe/+bug/2166461 Change-Id: I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a Signed-off-by: RowanJohns \u003crowan@stackhpc.com\u003e","status":"ABANDONED","created":"2026-09-10 15:09:08.000000000","updated":"2026-09-11 07:57:47.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ae18283f5e0ed00dc6e1e6dd49f7ef0e3f2fed1","_number":1005063,"virtual_id_number":1005063,"owner":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-10 17:21:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-09-10 15:20:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 15:20:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-10 17:21:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"028805b52c1b751e57e981ffdaa16da3f9938b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-10 15:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"443985aabc50abb45afa2387808bdded24fc51ce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-10 15:20:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c694a8f25be1d3285efd3ccc8e1ec3a073d8ed58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:21:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb81537f118424caf7e1b26c01b0678\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d251098892442ab30c03652850bdbf : SUCCESS in 10m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6491e557777e458a91603908b0bc9e5e : SUCCESS in 12m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d19fdcd0f26484a8fc464a8e60efb32 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e7d49e7ebe4ac9990fc2d18a931bf8 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/241c7589fb6b4695b704a605d2850773 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06d0992cc2bc4c60b52368eccae3a036 : SUCCESS in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a18e5bf3c1094abf980df1d0926436a7 : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9365dcd77f254e59ba2b604d65109f23 : SUCCESS in 12m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/858c53da92fe4aac83e3b4d8ea491adf : SUCCESS in 16m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a83c5353608b4ed6b4abd12a6ba6801e : SUCCESS in 14m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3828eede2f6f4c70a1e78a071e9c70b7 : SUCCESS in 20m 22s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/eed896e1700147d5ac83e66b3f62dfc3 : SUCCESS in 19m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60d8704bd8fe460db283c4a3f2f07e95 : SUCCESS in 29m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de4038b0d1614c8a822f69fd4fc8b04e : SUCCESS in 24m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b65b84356ae4410b8d22a53a661c6ea9 : SUCCESS in 25m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cbccc1408f442d4ad3853982b06cfa7 : SUCCESS in 26m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0fcd0c19f7f44630b1f90bccf547b62a : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/785623dbbb1643088e9efe1ee9f02aac : SUCCESS in 21m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/65ccfa0ca1424e35b6693664d6da562d : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0f66dc1bf4e544eb9faf29258ba6e7cb : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/857a11c574984c91aa0902ea18eedd54 : SUCCESS in 1h 17m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c58f75d460714c7db806b445f23fd5d6 : SUCCESS in 1h 31m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c769ee40b65f482dbfc85e0d62854663 : SUCCESS in 1h 58m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c1442d64750743778255f094d7b5d9cc : SUCCESS in 2h 04m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c8abc01b38624a31800f9d9d65dd2a79 : SUCCESS in 34m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e4715da96d34c9493a3d9a96c5d74c8 : SUCCESS in 38m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c703515698204a069bf16b10870bd442 : SUCCESS in 32m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3d4d908ee8b492eb6e35420d8fd98fc : SUCCESS in 26m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/519056c9ab9540dd8f68c0fc86e663a7 : SUCCESS in 33m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf1a6737bb9346e39fd50c5cee239ea3 : SUCCESS in 53m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cea40f3c784146869064f3de6366bb04 : SUCCESS in 23m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efdfc20ca81046a9a433118c4615e8e1 : SUCCESS in 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7c04bbadd8188c3b595444233269aace672d168a","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-11 07:54:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ae18283f5e0ed00dc6e1e6dd49f7ef0e3f2fed1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"date":"2026-09-11 07:57:47.000000000","message":"Abandoned\n\naccidental duplicate patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"101d56b3748c25d4895e90bf2acba26ab2db4856","revisions":{"101d56b3748c25d4895e90bf2acba26ab2db4856":{"kind":"REWORK","_number":1,"created":"2026-09-10 15:09:08.000000000","uploader":{"_account_id":39327,"name":"Rowan Johns","display_name":"Rowan Johns","email":"rowan@stackhpc.com","username":"rowanjohns"},"ref":"refs/changes/63/1005063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/1005063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/1005063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/1005063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/1005063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/1005063/1"}}},"commit":{"parents":[{"commit":"9ff5475ae2c16ab2833830f73002a89026ef23c2","subject":"CI: Download cirros images instead of using cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff5475ae2c16ab2833830f73002a89026ef23c2"}]}],"author":{"name":"Author Name","email":"rowan@stackhpc.com","date":"2026-09-10 12:01:13.000000000","tz":60},"committer":{"name":"RowanJohns","email":"rowan@stackhpc.com","date":"2026-09-10 15:08:52.000000000","tz":60},"subject":"Install extra kernel modules on redhat hosts Fixes bug: https://bugs.launchpad.net/kayobe/+bug/2166461 Change-Id: I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a Signed-off-by: RowanJohns \u003crowan@stackhpc.com\u003e","message":"Install extra kernel modules on redhat hosts\nFixes bug: https://bugs.launchpad.net/kayobe/+bug/2166461\nChange-Id: I8711e85c3f2439e3d7aa5bcd28530bed8d8dd00a\nSigned-off-by: RowanJohns \u003crowan@stackhpc.com\u003e\n\nChange-Id: I851e7246030e1995a3cc94340e79c942cc66af2b\nSigned-off-by: RowanJohns \u003crowan@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/101d56b3748c25d4895e90bf2acba26ab2db4856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/101d56b3748c25d4895e90bf2acba26ab2db4856"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
