)]}'
{"id":"openstack%2Fkayobe~1005313","triplet_id":"openstack%2Fkayobe~master~I4c8da47c28608f2bb1b768cc7f5f9d7b7c43bbbf","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-11 20:55:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c8da47c28608f2bb1b768cc7f5f9d7b7c43bbbf","subject":"CI: Bump kayobe-overcloud-base timeout to 3 hours","status":"MERGED","created":"2026-09-11 17:10:14.000000000","updated":"2026-09-11 20:56:49.000000000","submitted":"2026-09-11 20:55:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005313","meta_rev_id":"4278ba07af261633e397ceff1b83b15ae81cef4c","_number":1005313,"virtual_id_number":1005313,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 20:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-11 19:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-11 19:10:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 18:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 19:10:36.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"6c3eae0cb2bf1805ef2e02613744a0bd1041a08c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-11 17:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"746efbed4955f61e16be639097e634852f8353d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:51:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdfdeb285314bfb8088f75b8c1c6479\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b470834ed24b38a5143dbc3d2f9b97 : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/269a94cab5e543119b0399ce6e8894d9 : SUCCESS in 10m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a79443d3ba54cddace69ef6630de46d : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f978061421e74308bc0f2a5250056af7 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30fe32bbfe444eee83efb9ed5ea9cd5b : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2112a98a538e4d1da9c081556cd14998 : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d8e5ea49b42424dafe8a7df26ce9b3e : SUCCESS in 3m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3bbbb23944ec47158324c381cae7ec2c : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/44df0cf6d79d4ebe991069f1beaea91c : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/64581a46e5f84c8193135db9b466a7b8 : SUCCESS in 1h 17m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe110918efca4c9a99a63aef5a9a2465 : SUCCESS in 1h 39m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dafafdb405654bffaceda17f058093e1 : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ccf31a5dc5334818ea55cb897051fcf4ad9dcdfe","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-11 19:10:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79913c4a48e0530e70b586d5c67dff97a6155972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:12:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d8aa76c2449d3cf6b38d15fe88b1d1a20e936c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:55:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d782ee7b051b4b80b60ac7c39dc51444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc680f86cb884202bbf7314c65f812eb : SUCCESS in 4m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ded0a60fc86a4ba282d8ae336172b7f3 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d37dadf7564c0bbd4638dc2c2332f4 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2718de623cf0423b926fff8b882d501c : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e96be42210394f7baceae092d448a7d5 : SUCCESS in 4m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e32cf568175a4eb3a9c0012405e18a0f : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/535841e7acec495abb23a4f8219b8081 : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d11c314e0e354967b912dc8594327aa0 : SUCCESS in 1h 17m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7721acc4c524176bce54f7ae99d6ecb : SUCCESS in 1h 42m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ad26ac25541439a8a8981017c5363a0 : SUCCESS in 1h 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"77026899399a2d4a8d2592da7b8143871266d09c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:55:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4278ba07af261633e397ceff1b83b15ae81cef4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 20:56:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbdd03165c2647669785cdafe6574a48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f784b302f7a743c08f788ed2cd94049b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60c53d88b31411566f06b0cfa6c0c779a19afbab","revisions":{"60c53d88b31411566f06b0cfa6c0c779a19afbab":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:10:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/1005313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/1005313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/1005313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/1005313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/1005313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/1005313/1"}}},"commit":{"parents":[{"commit":"3ec0c1258b0be34131dd12aa08a31695b855a815","subject":"Merge \"Never reboot without explicit permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ec0c1258b0be34131dd12aa08a31695b855a815"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-11 17:09:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-11 17:09:15.000000000","tz":120},"subject":"CI: Bump kayobe-overcloud-base timeout to 3 hours","message":"CI: Bump kayobe-overcloud-base timeout to 3 hours\n\nUbuntu jobs have been much slower for the last two weeks. While we\ninvestigate the issue, bump timeout of kayobe-overcloud-base to 3 hours\nto reduce the number of failed builds.\n\nWe are hitting timeouts on the kayobe-overcloud-upgrade-base job too,\nbut we cannot bump beyond 3 hours because this is the maximum in the\nOpenDev Zuul instance.\n\nChange-Id: I4c8da47c28608f2bb1b768cc7f5f9d7b7c43bbbf\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60c53d88b31411566f06b0cfa6c0c779a19afbab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60c53d88b31411566f06b0cfa6c0c779a19afbab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
