)]}'
{"id":"openstack%2Fkayobe~1005337","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~Ia593472e28f5c79f153dd9c772c9a2aa3f1a0438","project":"openstack/kayobe","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-17 03:40:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia593472e28f5c79f153dd9c772c9a2aa3f1a0438","subject":"Fix configuration dump with ansible-core 2.19+","status":"MERGED","created":"2026-09-11 18:58:52.000000000","updated":"2026-09-17 03:41:41.000000000","submitted":"2026-09-17 03:40:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005337","cherry_pick_of_change":1004990,"cherry_pick_of_patch_set":3,"meta_rev_id":"a6734cc3733bb9d82d0ac4e8a30a609bc737dd05","_number":1005337,"virtual_id_number":1005337,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 03:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-11 18:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-11 21:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-11 19:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-11 18:59:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 19:56:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-09-11 21:36:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-11 22:03:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49257c7c32af6563d0e76dbbe85784c691468434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-11 18:58:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"be1c1678ee8522924fffdf7ab40a3d981ab05efd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-11 18:59:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3caf3ab2512ad67022bb3a6103652e677d0d848f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-11 19:56:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dbc9435902649039b3e1e3ef9974611ce52668a5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 21:36:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f21e045c5e04c3cda71afcee27bf642307eb505a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 22:03:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cff1f9bfa5d7423aa7e7bfcb7b881e3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0521a4678534237a151c6c1995f33e4 : SUCCESS in 8m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/800d6d9d83154145ba3bb202e84cc8d4 : SUCCESS in 13m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8415cee796e48afa5969aa4baf9b86b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e79c261015431188cbc3a0a2dc7894 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b92c80ce637147d5bbd47fa64e3ad0c1 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a9ff1f1cc314bde91e482f7ad38c6d5 : SUCCESS in 7m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5df5de8a42b74874bf48c636714cc969 : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/add593806b83451eb5108b45a522041e : SUCCESS in 11m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/932d219b16cd4f06bea08bb91aa49c2e : SUCCESS in 20m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2dc454d7c0944596b18bbc8547f53829 : SUCCESS in 12m 01s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/68f188a5cd6441c3bd7a5ff218b53999 : SUCCESS in 20m 53s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/fa352b3e94fc42a1b06ed159512f0060 : SUCCESS in 36m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a129849f983e4e74953587140d918b8b : SUCCESS in 31m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/894ecfdbacab463da32b1ad7bff3bc16 : SUCCESS in 32m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/902fb8b5095f4af8ae7226d004867f28 : FAILURE in 31m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e7e659ab6a74b78b3427d3524619839 : SUCCESS in 34m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/353f9c5a4723420abe6709f0585a5d71 : SUCCESS in 25m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b90c804f5d4e4ce9acd58bac4cf84a8e : SUCCESS in 24m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5b97366c3a944c40821787f83a8aa7b4 : SUCCESS in 1h 26m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a364e85db10c4a87a221376b9efe07b3 : SUCCESS in 1h 22m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88706e511b814b2f98351306a0f149b9 : SUCCESS in 1h 19m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cdfb286a1e94f8baa36e29fd8d73544 : SUCCESS in 1h 53m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/087a46c06ffc4a939206dd966359adc2 : TIMED_OUT in 2h 03m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/18b940afae294def81a71f90678fe44e : SUCCESS in 2h 05m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/350e97dd24cd4b4fa573076b6872898a : SUCCESS in 2h 00m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/351f1e9a323643c4b49f97c332cbe1e2 : SUCCESS in 2h 50m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4f978b07fab4820b4e2dfe896639408 : TIMED_OUT in 3h 03m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/81b9212d6dfb444fb5984402e2e4268e : SUCCESS in 43m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/40e949b9a155417e9fec33d9af29b91f : SUCCESS in 36m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e879417a8734f2290ca4933a6278e02 : SUCCESS in 33m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ab788be72e64c588665a94108a8eafc : SUCCESS in 30m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc24ba627a244073a6a45f9b835a3f3e : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/63c461cf72b842dc977c9af9886aad05 : SUCCESS in 54m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2dbbeaa81ed40ff9a4c5ecb3739a81e : SUCCESS in 47m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf709923fa6a4c319521e7fc0e42046b : SUCCESS in 34m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c4154c42789d40e69508c10af99c6d94 : FAILURE in 28m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c10484068c24d7fab65de427388be00 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d597462aac7812c36dd8a359d163140b6886fb92","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-12 07:12:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd157bf1a7cc5134231dd87d8a2190ff342f38a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 10:18:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f5721ea6309441782e903c3068b7598\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2365182668f843bfadcea4aea31ad1a5 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9cf8ed7c471e4f458a0a216342bb5e2f : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aaeec054f294aada5898407d4c26b78 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd3634d7533043a283fef5db87874f92 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b89dcf6c417648c690087cf4c76f7e43 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd7600214b9b453597ee2c78669eaee0 : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76df2488b41c4895982b838ff881b3b2 : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4619259d97b941b994e9034ff35fc354 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78a5d738723142f1b006ec7b2ae550f2 : SUCCESS in 12m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd631b6af34749e38a4084499a6b4ba5 : SUCCESS in 12m 28s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2e84b3773b004026927b0a3090b6c64e : SUCCESS in 24m 36s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/cbb4de006cad47de8c2a4edabc80d6f6 : FAILURE in 13m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22a5e712b0c84b4ea18f00153dfd1721 : SUCCESS in 17m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/770a6635393b4c06b4b988a93c07d06c : SUCCESS in 20m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a13cd3dc7a09481f8bbd6deb9fac7d12 : FAILURE in 21m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a83703f022b48e78ae8dbb95b802f83 : SUCCESS in 21m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7a991151940d4443b75451f32e35629c : SUCCESS in 20m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8932b78f21e8455e83bb49910ddb45fb : SUCCESS in 19m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/31223a43232847d7a37aadd83be7f2b6 : SUCCESS in 1h 11m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f27f73934b24b28a6a17f498f1508ad : SUCCESS in 1h 11m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2db179c89c0e4171bef30180ea237e90 : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e5ab716c4904e9b978aef2d82c8d30c : SUCCESS in 1h 29m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/671bee40ff0e44488b1cf9675da4c06c : SUCCESS in 1h 35m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4fd3069d5e8f42ef807125b0f9261a3c : SUCCESS in 1h 47m 27s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c529f94b5798469cbab3a662e5bac36d : SUCCESS in 1h 40m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0a229a9e86c4b5d8ed66e90d8b31f55 : TIMED_OUT in 3h 04m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d85e7f5c53549fdb86ea44685df4750 : SUCCESS in 2h 54m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/54e451cdbc2e4b9a9dbe03ba157414d8 : SUCCESS in 29m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec139ea5d43342aab5ca7eaf06dda1df : SUCCESS in 28m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f21c933a37ec4aadb1ff4c2b732b4294 : SUCCESS in 20m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a63d7760861b4f05b566ddcd0a53da93 : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ac05653219614b4d9a01be99080e417d : SUCCESS in 38m 08s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f4c8892883c74a03b05f1b9b11de8cf3 : SUCCESS in 32m 44s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/316c3fbf2fee4323afbe69a67889be8f : SUCCESS in 31m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/614a951b0a7a459b84414ee854c69933 : SUCCESS in 29m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3d95eba2b27d4809be8bd02afa77cce6 : FAILURE in 26m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b4cdb6c14da4baab4c2a6366ce1d77f : SUCCESS in 23m 21s","accounts_in_message":[],"_revision_number":1},{"id":"08bf99a82f507fd419a5c2e0628180c884212b19","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-15 16:26:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8c0de8e1f3711f85b0ccb2cd4a79c81282d3e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:59:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8e92f94028543c2b145ba169eaf317a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1592f70f660416fa7f7c122db37e1b4 : SUCCESS in 8m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbb83abda11b4ff4ae1185695f2c3b8c : SUCCESS in 17m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcf3a5ce7d5d429897fe4a7739f3dcdc : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336a224959fe45498982a2bfc601ee5f : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a6e8f54d56f446b8f711abf715e074f : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f31351a1e053455797a68403c46737d2 : SUCCESS in 6m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd7cd54530b04e6194f882e2bdc2ae0d : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad7c9396e78f4af9ac8dc5fce75a8dbd : SUCCESS in 16m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/884e92ce1b8646c0b151803391a659e7 : TIMED_OUT in 31m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a933ae6fc1941dcb1ecbdf868072c1e : SUCCESS in 16m 34s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c714f73356ed471d93df975503d12be8 : SUCCESS in 35m 55s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a654022c9310440bb6a9feb4fcdd617f : SUCCESS in 39m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1d5fa2018734c1ebfcd24fe2e32ba5f : SUCCESS in 37m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94531159e2c44a93a4340a8e5600ddd9 : SUCCESS in 42m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/039e4b52e4bd4bdfb12ed1122f5c7b21 : FAILURE in 46m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e247b0d6854d4254b695142301846217 : FAILURE in 41m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74952699c0544130bc6860296a8368c0 : SUCCESS in 35m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c219f96c9cd4267913cf352fc911f93 : SUCCESS in 36m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/98b8301636db410d8c3ce19d2cbc31c3 : SUCCESS in 1h 27m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4ad587557b9c4ec684f3726d9a9bf0c5 : SUCCESS in 1h 33m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1539e2aa5a724eaf8cd37ad8920a0bf0 : SUCCESS in 1h 33m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9ec259223ee4f10ae4f7c60d3b42f96 : SUCCESS in 2h 29m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/364b50356e5b401bba5d39afe1da8500 : SUCCESS in 2h 30m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cfddc43de57e4b798cb2934f7e76a218 : SUCCESS in 2h 23m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b314626a541b459baea71a398dd198fb : SUCCESS in 2h 22m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bec6d9c429814a2ab479144c54bf464b : SUCCESS in 43m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/939ba52acd1240d5831d57d12d86d3dc : SUCCESS in 40m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/713371a32f864cf5a6d119e7fb1334c7 : SUCCESS in 48m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c1e735a304164dd890201a3faddf24fb : SUCCESS in 48m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a708259394904f93bc8db400f8ede39d : SUCCESS in 54m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fe02f96c26954ba8a3ee6e0931976944 : SUCCESS in 56m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ed58b5ad23d407eb06f1b83d8315d19 : SUCCESS in 1h 09m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85637c46be224c8eb1358cabff8c0110 : SUCCESS in 59m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/78a63eb902e943c098e1d39431e77ce4 : FAILURE in 47m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b1dc8a3527f45839cccf3a2c1411f93 : FAILURE in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d9cb0fbf21c6625a7bb0dc4b14fed3ea6243bd92","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 06:03:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98ab99d11fb8050bba72100db0b56df7a8ed796c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 08:42:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34705c3b49e2487dbde67aa426b7219b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e4cd093ea64c8fa2725e4b9ccb162d : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/93edd1095cfe47a394743ca35e48e095 : SUCCESS in 21m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ebcb5f589d64877a319720d58d0854c : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0cfc168aeb4111a87f9ba6dbdc3aad : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8171eb3a8584ec2b0f5dda0186f6e70 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9de37a08c134f0096bfd08699621f85 : SUCCESS in 5m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cf2823a51b6485c8f4338d9c5edb186 : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b6539949053485387da9bfaa1b4ace6 : SUCCESS in 18m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51d89939a02c4ca39486b6e8706974ac : SUCCESS in 33m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12eebb9256084a78bc4bc7a4aa9c3cf0 : SUCCESS in 16m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/861281c440624b6a9b384c56367729b3 : FAILURE in 31m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8ab5e4f53a8741e983d78657c471fce3 : SUCCESS in 38m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de7c03bdadb54a68ab596e876173217e : SUCCESS in 37m 59s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/baf0778f88114474baae7991c796ff9e : SUCCESS in 39m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e79d2763cae749f08b0b943fd2371310 : SUCCESS in 44m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/943f9f40ba5a41a5ae3c1c47b48b05d2 : SUCCESS in 42m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8ae003bf873842d4a9644c73bafbb91e : SUCCESS in 1h 33m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/33e70e1832594a5d8e863cac676702ad : SUCCESS in 1h 41m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d3766ba713e74d11a5cd7e2bc2c83e0e : SUCCESS in 1h 32m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/babb1f45c99f4c49b1517a534c909226 : SUCCESS in 2h 07m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ab89786d67748fa861983cea4c5461d : SUCCESS in 2h 07m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/340d520dc4f84e4bb4e99ac74594d0f5 : SUCCESS in 2h 35m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/49e8a734593f42568eaf638565310f26 : SUCCESS in 2h 37m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/697f1b9d04644f93933321077fa268d1 : SUCCESS in 40m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b478f9a31d44403688cf1e72c4156ea2 : FAILURE in 35m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96c4dd31e8e340eda17f38ef44f7e190 : SUCCESS in 42m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/983ac388d03a417c8819bb3c6a542f8f : SUCCESS in 47m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/63901e3fa40c4d308e22713d8f35461e : SUCCESS in 1h 03m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6811633b138440e2912639a80f9ef05b : SUCCESS in 56m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9fde63ffd9d4e7fb20c517ae0dc9693 : SUCCESS in 56m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f6ee550b1e24e559e6021aeff873855 : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a29798e090ad85de1f7e37c05173d4fbce7330a9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 14:09:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55a50bb74181ea4f4dfb33405474a1eaae4f627c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:15:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6b5393cb1db49b68ffc91855ad15530\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85fb9f852350482898b05bb0dd3079b3 : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d711a13a612646d2862283dd6af6e45a : SUCCESS in 27m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b757d5d1789d449880f49a19d85bf385 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a27adc0dfb4a9e9f90274f8418e613 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6102cfcf77af48b4b5aafe41dc37807d : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26bd50ca03cd495e80127b7ce998debd : SUCCESS in 7m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bbc9ef5dd0840fc83b06bb053699397 : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f4ded7644594c25937c90a1aae244f1 : SUCCESS in 27m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af8b87ab54a34933983c63b054453369 : SUCCESS in 46m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70318f79c2324c7eb1370661eb7e2f1d : SUCCESS in 23m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7d7e3c2ed6344cddb7d2ee96200c38a8 : SUCCESS in 56m 19s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/64ee88c27e8146339abfd5b6336fee42 : SUCCESS in 1h 01m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b798cefc666e4d369db5010d7627ce6f : SUCCESS in 56m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29a46f61b0784067a8945f859e334465 : SUCCESS in 58m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1404ed14130d4a248a85ef2b5118f503 : SUCCESS in 52m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/713628f0ce924bdd8d93c54d2fd233e4 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8d28a8994c7e43b6bc353f004748ff1c : SUCCESS in 1h 56m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1586b3f206ab40b9a35c7169de184312 : SUCCESS in 2h 00m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2ac18f63a0814fa8aed048187c93e968 : SUCCESS in 1h 49m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/214c8069b01c42ea9cabc4f7bdb9da54 : SUCCESS in 2h 36m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/079225c5c55f4c7b95010280d22b8fd5 : SUCCESS in 2h 40m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/25a2d9695ff94a4ea0fb81122a8b16ff : SUCCESS in 3h 02m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/20458e88dc9c45539ed47cb847ae1dfb : SUCCESS in 2h 57m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/61e1b869b114425f8d487ef746122988 : SUCCESS in 56m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b750ce107b874374b1c4f16fe6a6292f : SUCCESS in 54m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc8508b641de4ed085b675886300f8e1 : SUCCESS in 1h 13m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/40bde325ac1549bf9e1360f98d0c2552 : SUCCESS in 1h 01m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ee7c612d969447d3aac46463ec2d271e : SUCCESS in 1h 19m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e5cb1076b2aa407a9a9f77bab293cad6 : SUCCESS in 1h 29m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f3bcbfb8e6644b4bb14a32f79fde2e8 : TIMED_OUT in 1h 32m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2529d473b9649eebbb04f37ea8c632f : TIMED_OUT in 1h 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c1e27c3e50d01069681702bf9bd2e9607da72d1c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 20:46:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58c3e1570187839d840e813eb2ce40641cafc956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 22:37:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22486b6aa7bb49b79709698da822a961\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde2d4967d61467f80caf835ee22d8e0 : SUCCESS in 5m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5067e62191b947d8b7e6cceeb8643883 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b22f9c33d1644d48d5f1c7431a479de : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a549fb2795146d5b00d49b4dec00fc3 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6ec64737a8249afb2dd8924ac981988 : SUCCESS in 5m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae921092770b470c951b2acc6d295784 : SUCCESS in 6m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b322274e0547099379c07fe5454416 : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff1f403a35c14a9fb610d016f8ea890d : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cf8d95f5617475588e26ecd8fdd237f : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/828bec7592fe4e6988ad92691b686808 : SUCCESS in 11m 18s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7e631ad699cc43d9bf11247be8e77c5a : SUCCESS in 17m 30s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/84cec55303af484ebc369cff7700e62b : SUCCESS in 11m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/087642d5e41549cba3370f4550fdb75d : SUCCESS in 10m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ca1f94ff2394e25861d74a47060b6a3 : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/33af4807c0d14996be5d938a659b592b : SUCCESS in 23m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b006d833c8d04c83a4bea7f1dec05de2 : SUCCESS in 10m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c60538b640934973b3081a2f6f4b3617 : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e78836c067204ab68d342e2c6cd81752 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e91e108c04294352a8899d50c3e429a2 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10c4833265b7424c9ef2d0b18023e049 : SUCCESS in 1h 49m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ef801f62385f4126825179357d7b3351 : SUCCESS in 1h 35m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f46a846178b14111b68c64d405560873 : SUCCESS in 1h 41m 49s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/09cda13a1056465b94e12995e0d7c8a6 : SUCCESS in 1h 37m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/caa834d55c1a4f5ea4a42a8ec5ac10af : SUCCESS in 20m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f34735e1088e48d8855e5609242e1c11 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/612b1959a9744c41b3b650a76e00e531 : SUCCESS in 22m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d0abba629414ab09d810648be694570 : SUCCESS in 14m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d684c1a0cd944305a90c8b981e45f38a : SUCCESS in 30m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/102ba16e84804a71b251d4d857af25de : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7124609e25f48bb8b9315750b142b01 : POST_FAILURE in 31m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fe7bcb714214796952dab9fe39075ae : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ac9644cc2efbcde809997f62b4441214e63cded4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-16 23:45:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56b5f0eda96d6939dc9937ff2e4ba0d75a9f6364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 01:35:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31dd85b6e5a74b9ca61df1fd5d8c5832\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc768e619304208a37c397eb0aa81b8 : SUCCESS in 8m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dcbd266a65714d2e85c7246178733e8f : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14bfd74a70434a64981f11c4312099d0 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c5d9f1cb9446bebae04e020dc2fe33 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d962a38cda741fa960542b6582aa96c : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/22c258dd057c4250a06407dd98b8c2fd : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86d1c8dbc5944f69c1c64e7087fe9a7 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7bb5cc22beb24ee5ac48e58e13b9ddad : SUCCESS in 3m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5297dd05ec3a40da934e765823d0cce4 : SUCCESS in 13m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78439a819703407fac5d6422cb62350a : SUCCESS in 7m 02s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/243260feb5c84afe90ebb6244ebfd017 : SUCCESS in 11m 42s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/6787f5e505544e65abc6119a1b30390a : SUCCESS in 11m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/184a4144907741d89d78b5bee39c2a53 : SUCCESS in 14m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0bf54ce727834d1d95346e6304759908 : SUCCESS in 8m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/08e47fde05954a9cb57cfdde29a5408e : SUCCESS in 10m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c923488f435040a29c85a05832b35600 : SUCCESS in 19m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2dd464badc834f189ed82b97960f3049 : SUCCESS in 1h 09m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/99dca3c8549f4b55add2e53f6bbdf399 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f86fb58942b24d2a9534d51385e3080b : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd512e23e1c247279f65b0f0220d86ab : SUCCESS in 1h 38m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/372247b2cf9b4fb7bf7e7892e6873ccb : SUCCESS in 1h 48m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e50ca819faa84f87bc542983c9a33e3a : SUCCESS in 1h 41m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7bc42541f8264652a40f6b07365350e0 : SUCCESS in 1h 36m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a93c0a46a77b49bfbe96638be2ae9097 : SUCCESS in 15m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1931c0562b6543078b2ca7295cf6f275 : SUCCESS in 20m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff4a44d8bfc14bf99c7e2a6809961d42 : SUCCESS in 21m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/311e24e3c6af4a9baf056637a6a404d4 : SUCCESS in 19m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ff5eff67677b4a04a8ccb2910ec9d654 : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9b9d4ba5525240e6b0fd0ff054b9dad4 : SUCCESS in 39m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fa4bf1fc4af4dbe8c0c2f45fdeb6679 : SUCCESS in 32m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/239084da13944a2ea5755c97dda03362 : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c00d5f3b0a70b8f97ea03a04ac6d4f04a1d8335c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 01:36:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24e646cb95db182d576d463b267bf45b110b7995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 03:40:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef160aced9464f63a65b7654a2f1bafb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ef356928a9494b8e3dc8a477ae69ee : SUCCESS in 5m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6cf4686222534b60869e6be8518526cf : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b786447aa314be3800e7635a2861e5f : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f0f0984b5c1471b9e2591f32e5ff3b7 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c7008b72c8465e925bb6c9afa6e533 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2474c309a39e474e8ac0041deded9704 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8474195caba342ce8ee584969aa78173 : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b39f661630047ce9ff6876807db11a9 : SUCCESS in 11m 38s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d6257de14af5411eb50604b0fb004626 : SUCCESS in 11m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7e73ec118e64e89a570f5738248fa51 : SUCCESS in 9m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/436ba8049d5f4704981d20a89f7e4f96 : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2b704027e8e43049e5a026110c71bae : SUCCESS in 10m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8c1e36d2777645af8262966b9543f338 : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a51d45ba39764b11b1c260234dc51fb9 : SUCCESS in 1h 12m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/640d7b61e7a746b2bc26209cc95df4b1 : SUCCESS in 1h 09m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/127c51811aae411c93c389249e4ac13a : SUCCESS in 1h 41m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/19147103347b49f7b0bab7d3be165d8d : SUCCESS in 2h 03m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02410a9c46b74b80bf84ea455f971698 : SUCCESS in 1h 44m 03s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ea2277f7f07e48f4b52f3674da4d6c0d : SUCCESS in 1h 38m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2f3aeba1dcba40be92a00dc34764c055 : SUCCESS in 15m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54ea9a5b22da46eca5c65230cd0eae4d : SUCCESS in 26m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d72c11366af740399b6aaa325ed62731 : SUCCESS in 22m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e1d0e6d6fd34ff49ca90ab15886f5f5 : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/04e49cbb2f7847cf8e87e51e31fc485c : SUCCESS in 17m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1f98af8c72a44f92912efd7cf5c08515 : SUCCESS in 39m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59799bc5214348788ad6daa870b1f007 : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc5b727833a44edbbf7ac7ddbd2161b5 : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a5c51936e63fec19881b3e0dde60fa865d7f547c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 03:40:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6734cc3733bb9d82d0ac4e8a30a609bc737dd05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 03:41:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecc7169155e740a39bd60257059780bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f175ef6f3140f8b476f6da71108cfa : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/066d652fb66b499c991a9a22580de48e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb5873d02f97787cab955bc68fb3c68bb4de7837","revisions":{"fb5873d02f97787cab955bc68fb3c68bb4de7837":{"kind":"REWORK","_number":1,"created":"2026-09-11 18:58:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/37/1005337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/1005337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/1005337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/1005337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/1005337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/1005337/1"}}},"commit":{"parents":[{"commit":"1b09b839dda16d7b414a34437c418945a9bba767","subject":"Merge \"dell-switch: Fix invocation of dellemc.os* modules\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b09b839dda16d7b414a34437c418945a9bba767"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-10 10:28:27.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-11 18:58:52.000000000","tz":0},"subject":"Fix configuration dump with ansible-core 2.19+","message":"Fix configuration dump with ansible-core 2.19+\n\nIn ansible-core 2.19 and later, HostVarsVars templates variables in a\ncontext that does not include the hostvars variable, and treats\nundefined variables as fatal. As a result, kayobe configuration dump\nfailed for host variables that reference hostvars, for example via the\nnet_* filters, or that reference an undefined variable, such as\nansible_facts when facts are not gathered.\n\nAdd a kayobe_config_dump filter that resolves a host\u0027s raw variables in\nthe current templating context using the public Templar API, with\nundefined variables tolerated. Values that cannot be resolved are\npreserved in their raw form, and values that cannot be serialized to\nYAML are converted to strings. Use the filter for both the full dump\nand single variable dumps.\n\nCloses-Bug: #2166964\n\nChange-Id: Ia593472e28f5c79f153dd9c772c9a2aa3f1a0438\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 4fd793c3817d6c9698da23d064ddee9fd8290837)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb5873d02f97787cab955bc68fb3c68bb4de7837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb5873d02f97787cab955bc68fb3c68bb4de7837"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"9ff5475ae2c16ab2833830f73002a89026ef23c2","ours":"1b09b839dda16d7b414a34437c418945a9bba767","theirs":"4fd793c3817d6c9698da23d064ddee9fd8290837","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
