)]}'
{"id":"openstack%2Fkayobe~1005372","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I4c8da47c28608f2bb1b768cc7f5f9d7b7c43bbbf","project":"openstack/kayobe","branch":"stable/2025.2","hashtags":[],"change_id":"I4c8da47c28608f2bb1b768cc7f5f9d7b7c43bbbf","subject":"CI: Bump kayobe-overcloud-base timeout to 3 hours","status":"MERGED","created":"2026-09-12 06:04:14.000000000","updated":"2026-09-12 09:46:42.000000000","submitted":"2026-09-12 09:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005372","cherry_pick_of_change":1005313,"cherry_pick_of_patch_set":1,"meta_rev_id":"90f828ebfd476120661c391d838ae8118e9c8833","_number":1005372,"virtual_id_number":1005372,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-12 09:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-12 06:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-12 06:04:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 07:55:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87d1d0b34b3624033d9de3d14092f6959e66d724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-12 06:04:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cfeece0d9278885d34127c6b5332c3d99543c846","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-12 06:04:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f577246c8c6a460c359a284a9c67a29588f8bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 07:55:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0938d859e334b4099bd44415d2d3e05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a090f003c746969c4bc8a093d16cc7 : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3bcaa5e47e9a42c0b5de0b990b275a28 : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a33fd537ce5c4e0b81272e835c110955 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033e5a5a60a443899dbf51a632ee1a8e : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64d3e87601e54be09957e306f0276418 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b6968dd4b87495eb700a3c98ff7a2d5 : SUCCESS in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3751448fb49f4058822ba628797142e1 : SUCCESS in 6m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/81b8650924554841a9b0016a1128aff6 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b8eafc7ea089488fa97a6ae3fc113833 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/536e6eff9ba54071a0fe8ce539dc8380 : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e91fedd505a648dc9afa818bb2086d12 : SUCCESS in 1h 38m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0338e03213e74f989f1cff7d06297d2d : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d8aa477dde6fb54e37a19f4211eed1f518fd6f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 07:55:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4a3760dd9673a8dffc1e1120f37e7bd0fde7cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:45:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a49e87ecc042e499fe4e9b73ed39cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad52c766a4ac4ba8be1c583b25482727 : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/87278dc4d5df44ddbe7bac3e48a98307 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e7b72686bf45dabfbff7d7e954bcd7 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe2ec2fafdbd41859c8a654f1d36aab4 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8d5bf40d2184967aa7c05b431cd57c1 : SUCCESS in 5m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/38ed5b379e8445e8926209772dc3b6ca : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6444a1a875f64ae2a1a3c99195b2161b : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/64e0b20ab7214a5ba551c07aa8acf555 : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f0f43ff02eb4bc1bcf2553eec29fea6 : SUCCESS in 1h 48m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec0b68ef12244153a9776f478c5de3a9 : SUCCESS in 1h 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7bc33d9d6d944e18ef0cba56a5dd16d356b3dc55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:45:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90f828ebfd476120661c391d838ae8118e9c8833","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 09:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173cf9b502cd4a6197eb5b830309ab03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92bd22b51ed14d12b1024b1c2a67ebb8 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32eef008cfe267fdf4eea762d895881e267f3621","revisions":{"32eef008cfe267fdf4eea762d895881e267f3621":{"kind":"REWORK","_number":1,"created":"2026-09-12 06:04:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/1005372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/1005372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/1005372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/1005372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/1005372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/1005372/1"}}},"commit":{"parents":[{"commit":"0695ce8a7e513d4b655c8a0994db7a03fe284899","subject":"Fix configuration dump with ansible-core 2.19+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0695ce8a7e513d4b655c8a0994db7a03fe284899"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-11 17:09:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-12 06:04:14.000000000","tz":0},"subject":"CI: Bump kayobe-overcloud-base timeout to 3 hours","message":"CI: Bump kayobe-overcloud-base timeout to 3 hours\n\nUbuntu jobs have been much slower for the last two weeks. While we\ninvestigate the issue, bump timeout of kayobe-overcloud-base to 3 hours\nto reduce the number of failed builds.\n\nWe are hitting timeouts on the kayobe-overcloud-upgrade-base job too,\nbut we cannot bump beyond 3 hours because this is the maximum in the\nOpenDev Zuul instance.\n\nChange-Id: I4c8da47c28608f2bb1b768cc7f5f9d7b7c43bbbf\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 60c53d88b31411566f06b0cfa6c0c779a19afbab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32eef008cfe267fdf4eea762d895881e267f3621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32eef008cfe267fdf4eea762d895881e267f3621"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"3ec0c1258b0be34131dd12aa08a31695b855a815","ours":"0695ce8a7e513d4b655c8a0994db7a03fe284899","theirs":"60c53d88b31411566f06b0cfa6c0c779a19afbab","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
