)]}'
{"id":"openstack%2Fkayobe~1005886","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I39f5167379af92bd4747bb1197f9670b010f6fe8","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-17 06:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I39f5167379af92bd4747bb1197f9670b010f6fe8","subject":"CI: Enable EPEL only when required","status":"MERGED","created":"2026-09-16 14:06:29.000000000","updated":"2026-09-17 18:56:08.000000000","submitted":"2026-09-17 18:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005886","cherry_pick_of_change":1005831,"cherry_pick_of_patch_set":2,"meta_rev_id":"fea880c35ba8a820411413fc6e6246e027d464a1","_number":1005886,"virtual_id_number":1005886,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 18:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-09-16 23:50:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-17 02:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-17 02:29:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 17:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 23:50:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"3145b5e747e39a3a427646ac0c7f3fbb33446097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 14:06:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"47d65c8030c10bd31d22d08d01c5f289ab633070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 17:15:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/846a7671685842e6ace72b2c41bd76e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6daa238e498a47269a15444b2508fa15 : SUCCESS in 11m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf5eed68617f44dcb96503399623cb50 : SUCCESS in 26m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6743e05b2aeb439fb8752992d153c511 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f169905dbc43a4b60cf4271e9b1fb6 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa30ddd021a746578cac11e593c48419 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53cd4fc4d2b442429faddbeb63690fdd : SUCCESS in 7m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b33609d0c124432b99ca458923ed0c71 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ea05dac8dd04341a3c2c3ad61a030bb : SUCCESS in 24m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b33089aef5a342fe899aceefe524327b : FAILURE in 53m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed618a8377b44f4d98a9887cc8431a0b : SUCCESS in 22m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/af231d2d42d74ae4a866bd63ce3ed070 : SUCCESS in 59m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97184d2bcafe48dc859424eaccfd09b3 : SUCCESS in 56m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f573146ec90046dd80023ad03cba1966 : SUCCESS in 48m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/413b925fc3cc40988c2dd655375b1c8c : SUCCESS in 55m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/56885b6dbd8242ecb5207ed71c785f82 : SUCCESS in 1h 49m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d884c0457baf4d96af145bffa1c8f34b : SUCCESS in 1h 47m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/12d8bd0893864d339667e9e1df75dede : SUCCESS in 1h 49m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90e96f2f09744960b00578611b4b770c : SUCCESS in 2h 30m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9292f3a608594dfea4e6fe10e8b9f7f5 : SUCCESS in 2h 16m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bc0d34def7184f2dbb3eb09ef5002c3c : SUCCESS in 2h 58m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d624d271a14d4d6e8ad090b6b683dd78 : TIMED_OUT in 3h 03m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fcf8d1be18fc4927acace3c691e0495f : SUCCESS in 47m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac829d7496af442fa5460736eeb20c77 : SUCCESS in 1h 09m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5296c347a332492a8bc07847a1255c93 : SUCCESS in 48m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f5ce3d5960844502a1b1052d82d77400 : SUCCESS in 48m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4d6a01496b4c444ca25657afc18e61e5 : SUCCESS in 1h 20m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ecde4555659479cb443ec78cca0475c : SUCCESS in 1h 32m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3532688c8ccf443fa9a0b5a24f3ab07f : SUCCESS in 56m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/773c7e26c0d2459d8068bf4f73a780c1 : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"691426d0e0c7f8ab2f2be330c325e1d6c573c9e1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-16 20:45:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95dbaaf6e087f2287ad2527234c4585eae215585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 23:22:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5549cca16664feda69ba4a4d6bb75c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf14e3f2f35f461a93a0cbb81f5d8864 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cba90b0dc1714797b6e1ea28aa8c7549 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5a6107cc81d4aa9ad3322134fff98ab : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fa2161aa6514533a615be01f6560c8f : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e866be88efb140fcacc854c7fef7e6b3 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/055416151f3148a08c9daf9c6180da0f : SUCCESS in 5m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02a17470cb1a416bbcc2dec988a4fac2 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0640a08c5f4a43d9b6059df2399d4211 : SUCCESS in 8m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0af15cbcd514445bb93a60c9e2fca55e : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3c6a40c1fc149cb888b62db7dc87c64 : SUCCESS in 12m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f5ddd5de1bde4b8097f1966bfa264ffe : SUCCESS in 10m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df08699e5e0741c1806acc95d674ffa1 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ba217c1e15f0466abb2126cb032bd730 : SUCCESS in 21m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c61d41819184e49b83582f7e39859be : SUCCESS in 19m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1a991ad9db7a46a18a5c157423f56f3e : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/daea156ec5d9491cb8eb3811c0d0f57b : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/786d3aa93c6a420482f3eaf8ea525dfa : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43f01f50f3bc47fda919a651585171bc : FAILURE in 44m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3e67f89cb524c7cb5ec25bf7e1761a3 : SUCCESS in 1h 44m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6df2139e2ef34d52a9a03cd76d830d87 : SUCCESS in 1h 33m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5219bdeeef68437bb36ed09170e18c64 : SUCCESS in 2h 35m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0047247a984f4d3f89ea638a3611324b : SUCCESS in 12m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/591cc6e0a8ad45308b3b3b8379942646 : SUCCESS in 24m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8765aa52609b495091f3d46e84967ae1 : SUCCESS in 24m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0ede7a8734f3439fa27f20d90eb5e859 : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/866415d82191469cbe8dc5a31b07ecf9 : SUCCESS in 19m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00a3a168489d4a099747253ccd98c7be : SUCCESS in 29m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/79f9bb9968874f22987e45420f45f6cf : SUCCESS in 20m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0831c3696cde4f39a77fcad83c869787 : SUCCESS in 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"67402edf39495e65b1efb0f90748c4580f00b721","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-16 23:50:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa751c22dcc144827acf16eb8613f735da8e6d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 02:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de397e772e904f508c21424219a987d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710c7d9cee8e4c428623c80bd3d2ff1d : SUCCESS in 6m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7122746f65cb47bba3fd50937055553e : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/052b12b2569f45498cf607a0b5576152 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783d18b2978c4940885bac27fef0096d : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e3d6471e9764da19163a5dfa59521d1 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a78701815bd24e8485fab9677e18dbc1 : SUCCESS in 8m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12e8eeee5e38407e88b569e4bd944a55 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e72849cd3a84ee492e4a9bfe284fc57 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddd4c9f4c6ad4b17b930043692915479 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0420751d8afb4208b13ec9a7270154d8 : SUCCESS in 6m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5c73034e83b419f938d496f4af4c7f9 : SUCCESS in 18m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d5ef07862ee4b618836fff0a98db72e : SUCCESS in 10m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6d69b22d40e146bb8b94c09ca75b5139 : SUCCESS in 14m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76282709a4094768b9c679efc59f9fd7 : SUCCESS in 20m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f6057ef861d141fb803942d2a21f6048 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/28c12df9ff9f44928e015d2b042bd5ca : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e9d5f4f6c563451a9d2d898abdea36ce : SUCCESS in 1h 01m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1031e12b637f46c29bf7db1ec3926edc : SUCCESS in 1h 51m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2c7faf80230b4af6ae70c2f4cc3e197e : SUCCESS in 1h 28m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2056484516b14a55ba8c42b85736c58f : SUCCESS in 1h 30m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a2d6d6580d94106a8c4bb88318d9211 : SUCCESS in 2h 34m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/72b2c70ba93a4cddab90de203ec6994d : SUCCESS in 25m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ffe4bd5a1ac640529de2b076dad7500d : SUCCESS in 9m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81b4362140e542378c3805e1b2420e1f : SUCCESS in 13m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2dd3ed7bb6834f71b9bea32e1542a69b : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fcb32a7560774624bf6516cfa66f93b5 : SUCCESS in 39m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a93aaf2a9feb49e4a651a6b85c068a6a : SUCCESS in 24m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d9199d22f6aa4e19bcbcf61537e32827 : SUCCESS in 9m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba48c42602e34bc282a4f85c51c68bef : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"daa659e08eae75f1418396e9cabae06eb740f417","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-17 02:29:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b21cefe548f0f4760fdcf72cd41752e2b03b7457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 02:35:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a33ea6e1143d3d49fd4dcf2abb5a4a1875437110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 05:36:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee9288cc6209471db465fc532a8487dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3d0535b26e43199290e820abeb35e6 : SUCCESS in 8m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ef6a01c55cca40a599cb08e0778f81b3 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387bef79f1f8467f805265a16fb90158 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7f13cf125764c6eaf1eb0cfd083a26a : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc928fb0c5404083b7560cf171eb09c0 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6de09c6b12eb474ea776865e940f586f : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddc4180dc7bb47aea575888ba4e92e06 : SUCCESS in 13m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6bc9f2998b1b41b3a782601733eb39e8 : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4d0f04c0152b4c66b2161c935e6c988b : SUCCESS in 15m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32f114fe20004312b3f1cec18d95ee7e : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c8b69f9ead3541d0a76147bcc545c2e7 : SUCCESS in 10m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be868abf18814a569004ac2414c1ad4c : SUCCESS in 17m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bdb967ed66b54196bf68c71a5ca00573 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/068ce5ac31e542529c9fb153c0812bb7 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/84cc168942834ee6b585552a7a10650a : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0f3ccecf5e444b18d31e01af70cc97a : SUCCESS in 1h 41m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ff9ecde766f4703913f3b76040dc42f : SUCCESS in 1h 51m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cbb9151432354358b853f7e6dc45e3af : SUCCESS in 1h 32m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00813ef3d0264c46a0769e6ba0d0351e : FAILURE in 2h 58m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b6019dcbc6ce49d2beb945aae8b88233 : SUCCESS in 11m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ae88170765a4993821250182d0de22d : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dae02d730ee14d4a91c70ee016d23768 : SUCCESS in 20m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47e60f9dec0b45b598351b4058b6d5e1 : SUCCESS in 16m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dd37dd5597e94ab38d8ffba163723d51 : SUCCESS in 30m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6f7c0401b1141f0a8e8496770d4ae81 : SUCCESS in 17m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd46e611ae4f490190960599a98920bf : SUCCESS in 19m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3950a32fe743428ab2a62c4a445b50e3 : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7f2197c6afd175bb4f4eceb76900f084c1f56414","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-17 06:59:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1893df44771b36acf0ab0b7ede2b7d8750555715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:10:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49bf611102454f7ba5885ca521ad07de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84edf2c89ef44566849e2e75176c7e8a : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2eafa48aaf864fc59532496df0048995 : SUCCESS in 8m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/884b3d753b8f4834addb53f0af309f2e : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34584e40265d4f37bf2a47b8c690c87a : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f1b271046d4441b895c3e7ace64fba6 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbbd9585251f4b6fab8f1bb4af147f9b : SUCCESS in 6m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92271463c28b4dcb84187e63ad7af6e7 : SUCCESS in 6m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f37fc0a43664fd3bc5b9823a9a5a657 : SUCCESS in 7m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/279d2ea0c1e24b63a0c2da6ca4167ad4 : SUCCESS in 14m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1317e70e20f9434aabd7d2021985f05a : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/262641d5d8b7489e84bcfa389d0d9f24 : SUCCESS in 24m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b18613fcdbe84c93a845da16c26a6ba4 : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96040d093b01402fb43cc8a2031ae16f : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5768d4a8e4b647269a5981160f300e6e : SUCCESS in 21m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/df111273f3b94103a2a6e3f4a44c4e13 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9ce0af8739064f929e5696c4f07fb26d : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/011bb37a4dfe4cb483643fe539739fc8 : SUCCESS in 51m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10361020189d4b118befac8c694b2eae : SUCCESS in 52m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f72334113fc54e94be1b0a50fedd06fe : SUCCESS in 1h 45m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e7b671b6420947b487df2e84c4871aa2 : SUCCESS in 1h 17m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/134c68ad0a0249028834525f9fde06bf : SUCCESS in 2h 45m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b9a99215c706432396fc883ed47827ff : SUCCESS in 27m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/881d60147a0845d0a198bc0e48a39320 : SUCCESS in 22m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/381bc87940434a1ab1a63e62cf15acdd : SUCCESS in 20m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4994b9c0cb76417284c12f45eb448f55 : SUCCESS in 20m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f4ceddb53b04aa190069cf9f71445a0 : SUCCESS in 33m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e768966c7294d5c8ad34e9c2f216dcd : SUCCESS in 30m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c8c14eef58949f3a29ab51dc9d35e28 : SUCCESS in 17m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c48c6e6981a34bbfb012e1c20bc903dd : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":1},{"id":"76b547ab260a720e9db53769197f7c6c53b224aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:11:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6880590ee9c2fd87c2fee3727bf81803f3e407e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:54:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b8d0350e8b4ee08cecd7123c2e87d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef0ba61b6484a0b9bc06bb95586d480 : SUCCESS in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3cacf65af77a4597af64a011b07dc3a2 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31d0394c4904a539d4a2c418dc9b077 : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07f6c02cac0541809d22f4ccea42c654 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ba6ca65867d4d84b2c35b53de0e8c45 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6bb1f02abcd4b119159e838dec3de8d : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b500e1b86c74093970cf317ecfd4525 : SUCCESS in 14m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5c32893dedb4abaa4e09a7b5ba280a2 : SUCCESS in 14m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f1b325f98023452fa4858a0280c2f000 : SUCCESS in 22m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cb26ef143fe42ed960608e0bed5dff1 : SUCCESS in 15m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6f87ebc103ab4e70b320a7577b0bd5f4 : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22814c18869d4eb29f389fea8e090502 : SUCCESS in 19m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/95c60f610218421d90076c10563ca304 : SUCCESS in 1h 09m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/16d5103108f345e29615e337d9f7ee8b : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e37127399fc54990a82559620d21d434 : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d1ee7014ab54fa49b86b4a20d45fbb5 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa14f34a92ce48dd9649e017dee1164b : SUCCESS in 1h 45m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/db3dac0ab26e4867b89aadaefa87252c : SUCCESS in 1h 33m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a76f399527fa4316b4b1c8423b502566 : SUCCESS in 2h 57m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/852d9bca2f9c4575866566c71c4301b2 : SUCCESS in 20m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/30c89da9732d4e5da993f103e61ca115 : SUCCESS in 21m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67928c198e6c4ec1808baf6b6a5ad25a : SUCCESS in 21m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/824025ca4fd543d08322d00e4566728c : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a6b42f68563b4254983fb4e9c6bee2dd : SUCCESS in 33m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13bc623e0e414b1db7fc1e222d1b46bb : SUCCESS in 32m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc1d68278d974fc5b20c2aeb4e9740cd : SUCCESS in 21m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d806e69f63545fca6269a9d3eb44511 : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"006fa0dc287573d2bac6837fc4d1452cd1904d68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:54:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fea880c35ba8a820411413fc6e6246e027d464a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:56:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ada8a5bb2ee4ee4b804f535c0a5c814\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600fd56fda7040e0a931b24333e314e8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d21d1085c5bf820eb133a77530ce39e170664238","revisions":{"d21d1085c5bf820eb133a77530ce39e170664238":{"kind":"REWORK","_number":1,"created":"2026-09-16 14:06:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/1005886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/1005886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/1005886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/1005886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/1005886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/1005886/1"}}},"commit":{"parents":[{"commit":"406c3ac10c1809001989aa1f7657512a78d3c83c","subject":"CI: Bump kayobe-tox-ansible to one hour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/406c3ac10c1809001989aa1f7657512a78d3c83c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-16 07:22:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-16 14:06:29.000000000","tz":0},"subject":"CI: Enable EPEL only when required","message":"CI: Enable EPEL only when required\n\nWe configure EPEL in the seed jobs only to install gdisk for building\nDIB images. Skip EPEL setup if build_images is false.\n\nChange-Id: I39f5167379af92bd4747bb1197f9670b010f6fe8\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit c05292c250703a9583ea46ae8f972b1257247448)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d21d1085c5bf820eb133a77530ce39e170664238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d21d1085c5bf820eb133a77530ce39e170664238"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"40f873182f46bd46e265a97224348cfdbe49034e","ours":"406c3ac10c1809001989aa1f7657512a78d3c83c","theirs":"c05292c250703a9583ea46ae8f972b1257247448","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
