)]}'
{"id":"openstack%2Fkayobe~1006443","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I1545baeb7e437b4b0928fdb054b4ad34045357dd","project":"openstack/kayobe","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-25 15:02:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1545baeb7e437b4b0928fdb054b4ad34045357dd","subject":"Add support for configuring Dell SONiC switches","status":"MERGED","created":"2026-09-21 06:43:33.000000000","updated":"2026-09-25 15:03:22.000000000","submitted":"2026-09-25 15:02:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006443","cherry_pick_of_change":1004574,"cherry_pick_of_patch_set":6,"meta_rev_id":"4f148fd4937726a92e435ed3ba9a3bf794ffdc32","_number":1006443,"virtual_id_number":1006443,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 15:02:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-25 11:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-21 10:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-25 11:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-25 11:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 06:43:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-09-21 08:40:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 10:52:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-25 11:09:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"6f9769b537fc23e8b2236c016a0c056ea6d161f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-21 06:43:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"054766ca3966cac115e7401f139700853b8f4491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:40:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b6a771c36554362996d3f08aab86c10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ef3176c725411fa5cd8da98fa8aa01 : SUCCESS in 10m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d26fd09db0f2476bb6e6a22a719899a7 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/399ecc3f18394ddf9d4ce03ec671e7da : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e287ba585814fccbe059c3b61f64362 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3bbf917ea134a12b764ebdae0987145 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/68376a54624f4e1686132be351bdd00f : SUCCESS in 8m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1043a72052884f24b873730419a9f7da : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4885e2e86ad494aa430614d6f26c094 : SUCCESS in 3m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7bf0bb55cc14ce1ac073d3e85068e14 : SUCCESS in 6m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a368b2d72014d94a96525d45c2c34a3 : SUCCESS in 6m 42s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a0c2ac7554da41c28686d7783f17e73f : SUCCESS in 16m 14s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/172a6a66bc1b42a99858fb49466ef535 : SUCCESS in 20m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02ad8e0d002440049c62e8fb89ece172 : SUCCESS in 15m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d4c9fb6dce544b7c856ac21d584fc065 : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5b2d941c68374ac18b496727e4fae748 : SUCCESS in 11m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/080f0308e363443ea7b21756cdf5e39e : SUCCESS in 17m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/95febf1975094ac7a25fae12c891c995 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a4a0f20a861e4b808d7cf68b32cc6ec4 : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/03227d7a60d24cc2a1a276260ac20d5e : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb96fb74de34441d99f7e9dd64d8a168 : FAILURE in 40m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/91b1cc9091a64b78b57d8b449a8eeeb4 : FAILURE in 14m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/38ae0a3c3d3340929d2e4168ef2675bf : SUCCESS in 1h 55m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a0d1a75dfc424ba9a36a455086202ccc : SUCCESS in 1h 38m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/945c3c66434243b89157317a0eb250e4 : SUCCESS in 27m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c0f1ca68e3ed4d5d94dab38f98378c98 : SUCCESS in 14m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1042654cb45145e0959e1d0b8b8b7a60 : SUCCESS in 23m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d37745fc74f74188aeecd360c72c3443 : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d3e615e83e9e49bf9653a30d98c54820 : SUCCESS in 37m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f3a176b4ec2c46f3a7bc77df46452863 : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c7e63650dd84d36a47e5bf54060d0bf : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25ff9d6f8d6e44b8afb19b4c7623aac1 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":1},{"id":"502ec290b01f4995e1fe3809fc8a9cd566e774be","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-21 08:49:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"771b01e8258e7a3d8307d5378e65e7fbb1977b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 10:37:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ba3bc0faf524d9eb29e70dd8b24d70b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663fa17602624209b0a15f07c42fd4e5 : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/24629a9a018f42819a6f1b7fc8c7c753 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da978299827d42178362ce2b04fd3a4a : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a304763739e425b91321d1f8c3ad098 : SUCCESS in 6m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bed698287514937b12b42e7f960ea7a : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09a5e082650849d8a69ebec8934ba357 : SUCCESS in 9m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79ee7ec7ba3c4c72a7b40b56ae19ec78 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5637f54511644c64b78059e610b1a855 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/136e476ab1ad49d994f9e7bce401a9a8 : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e39d2b4328541a68a393ca13c5f6b9c : SUCCESS in 11m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/599e39e0e7a2444c88c307673963fc9e : SUCCESS in 17m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b509d361bcdc4a4b99a7569ad10cb6b1 : SUCCESS in 18m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45ef8199ad3440188e8a36ce0db24200 : SUCCESS in 19m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/543b85f49a1d4fe4b2d41a6aad895a5e : SUCCESS in 21m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9ef5e1233e294141bdbf61d421b9299d : SUCCESS in 25m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0852fd44563b4738a0b7f57ac48b77c7 : SUCCESS in 20m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2cfb13af250e4e7f8e76cc2085e55715 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f9d1b202dd34ccea4d04cf419ee3a16 : SUCCESS in 1h 10m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6edddb464a64430f8d09379771e2016a : SUCCESS in 1h 10m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4116997d20542debb2de809c42a7a7b : FAILURE in 16m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b38786e0c9a54632b654f11413585c7f : FAILURE in 11m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b20eb22b78c34f52b5c4378f56435d75 : SUCCESS in 1h 47m 06s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2d105fe2db1841b7a247d20ccd33904d : SUCCESS in 1h 39m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0e56110fe75e4b8991d24423cbe21a08 : FAILURE in 23m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/723da0b9669c4e258a035c4efc5cdd82 : FAILURE in 15m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4d5b8fe717e461d8ad8d0bf454fff99 : SUCCESS in 23m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10ac56e74bc54dc2a9a7a3c28cffd638 : SUCCESS in 22m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e0ba36e384a84c338e8d61c9a4ce1c82 : FAILURE in 21m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0aa03244d33f4fa69c5737fa2083f9e5 : SUCCESS in 34m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7efdf16126543f18d14fef27e38f9be : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf8ea52e84ad4b83943f60dbc27539f6 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9b61c0c1a06b3aaf8335fcfea94c0dd76a825af5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-21 10:52:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"88dfe9296f5ba2874976710499011251be40aba3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-21 10:59:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"502e705c69a2fc2d74dc9ea3718a4e528a71b5ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:57:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1892601d6ad94c38b93dda5ec8b9b75c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623bbc2add184e7685b2fafd1fb0a7f3 : SUCCESS in 10m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3218a8803afd4159b4a608c02a0cec2d : SUCCESS in 8m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3094fcd2fc9547f389831aef3f48fb3a : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2febf62db21846c2bf02f29722318a7a : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57fd3eedc8bf48b7bc4cb8cfd998893c : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2beeffbf5b44deaaf753a778cc0d8a4 : SUCCESS in 4m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a7d392bd2bc455cbcebbbce986d5481 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6cb110669a174db9889b7467d6d9b5ab : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9151274d983c4b9b996e85537685ddd2 : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ed1b4810c42471bab60bc51904e5b3c : SUCCESS in 14m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ee5b0dc640f04b7f8d0ee00563b149da : SUCCESS in 22m 19s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/91b4ceb71ee64034b5eff9a31c16ccec : SUCCESS in 24m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4542137e28f46548af49adeb37c83a2 : SUCCESS in 13m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/12850aca27dc41fc8f981e5387458f20 : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6ec2ea954745478a956d12988805f3a3 : SUCCESS in 13m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/193ad3de0f3649688253f2449dd7ed5b : SUCCESS in 22m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6d6edf4d933547728ee182007ce9b3d4 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2c5dc526e49c4a2aa9dd407dd786470a : SUCCESS in 1h 14m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/be6fac96577548b0948b48ac9a16e0b3 : SUCCESS in 1h 11m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/880017b8edbb416c9165ffe48ad0eec0 : FAILURE in 18m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c9d688777dc4820a4d73da95434e7ce : SUCCESS in 1h 35m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8e0ff94fbf5240499ada6172ee066a07 : SUCCESS in 1h 53m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/75eb398097b640ca804f71663dc5e581 : SUCCESS in 1h 40m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3dc2195ab0b64cc78b4bb94b8d9f593e : SUCCESS in 16m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b481c3e7ff5e45cb97b7bab9345a3142 : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004d5ddc6c4541f1a72b72e17cb2af42 : SUCCESS in 23m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3748d4278664cf98b20f261e2b5392b : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bcea6e6631b446ec8655ff1007b1b39e : SUCCESS in 23m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4d565ca0cf7a49a28df2f00d7d3314b6 : SUCCESS in 22m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28cbf97fa0884edba5f0f46d82cf65f8 : SUCCESS in 33m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7dc3e2d242e41fda29ce194b080c8bd : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a3d3f1182fc2a849cc2004a04f75f9ef3d235bc6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-22 07:56:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92fac0af1ed24c11140da78f08bc87f435c8b20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 10:23:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3e399ce5f164d6892e50385b7e67aeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/067da3814a42463184dcdd8c12846325 : SUCCESS in 10m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1811aa91fbba47e59e04c3669f45671e : SUCCESS in 7m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ee195e43fca42e9a1b905c3eb4be8b8 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae3b564b0f3450697764470198ad7a2 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bf3e988d8814e769d57fd20ac866cbd : SUCCESS in 6m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eafeee0d77154fe09bf44a547fc3424a : SUCCESS in 4m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/945f2918211c425fa8d802738be5be78 : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/463570cdbbd7473e817c0bfcc47723ac : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5c8e4f2273d4224a7734748e8347290 : SUCCESS in 16m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a563b9d23da4500b1f3965f9c94e26d : SUCCESS in 8m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/12b4f6f9d2ee4e5c8e8fd3b2d1f1760f : SUCCESS in 13m 39s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e44d667e72d744ce9b7433a19c155d62 : SUCCESS in 16m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a438a8b00b54fc1805c27acfe119d58 : SUCCESS in 18m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/727a9e04e2fd42c28ceb57938568fd04 : SUCCESS in 18m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/97dc30ca77244b4f8a6d1726a949c59e : FAILURE in 13m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/283ce680098a4207b750952e204c6e8e : SUCCESS in 23m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/40b8f3026a4c405bb20925d67056d65b : SUCCESS in 1h 11m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c324bc0c5d494efe840e3887e864c865 : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2840309d1a634442b39c130afa90b406 : SUCCESS in 1h 10m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4ce8df7afd047d7b140ccde7ad5d1da : SUCCESS in 2h 23m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d4f42a1c520d4d66ad1f03128f9772c8 : FAILURE in 17m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc92034dd7054e15a6f2952f7044be35 : SUCCESS in 1h 50m 21s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cea25cc8fbf844dcb411a984b1e6698e : SUCCESS in 1h 48m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/223ebc7e5ba6495aac406dd5f5831428 : SUCCESS in 29m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ef26cce4b66542e7a48b2367c28f44b5 : SUCCESS in 28m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/307a633b4f5e4976a6352eeffee18c73 : SUCCESS in 24m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/25da6211b6764deb977f0fbd37ef938c : SUCCESS in 22m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fa862830efc94274bde1f82acb61c033 : SUCCESS in 27m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/927353213bad4cd4a596ef4aaf4b448f : SUCCESS in 26m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6de4f6cc36834f01a803ba1ffd1afb4d : SUCCESS in 38m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40f1b6d7e0a9465fb1acf6009b267cba : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fe7cc666816efc81e7e33b74f6efb36cc9f0e1d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-25 11:06:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"460c77f74fc1a3384b5b309594db3c9c5cc341ad","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 11:09:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15c3688bbd07258244c7e2bd6e0ca6d80557864f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 11:09:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a2ed2e86f1cbd282ea636c0dd14f2ffb40114b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 13:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d0f5680056344d58462a94c64873a1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badb2818685c4e8fa886cbfeb719e237 : SUCCESS in 5m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4f88416bc31b45b7bfe177cffe975350 : SUCCESS in 11m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feb4b5b0ac9b48fbac0650dea8d332b9 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab52d37eec247c6b1f1b49b3f58b260 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad40c4ad775348fcb41c2f4f3472a417 : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/273c53582ec44f56ba9f1a3001a2f578 : SUCCESS in 4m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caab725a1ec74b259e28e70b3ed2dc04 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39be7c091802485d8ee420f4d87d0db4 : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/15d412558e644fbe926521fbe9488f03 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aeb6408771f6411a8299c33668b508b3 : SUCCESS in 12m 15s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fb4f6c2f53714550a024aa6f916bf576 : SUCCESS in 13m 51s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/ada2f9be4de441718e30eac9f944298e : SUCCESS in 17m 25s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adf0f30c38694a23973a1a6212c93e84 : SUCCESS in 12m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d7b5b4d1ca2740808f094d0873baba4d : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5501a35ec19b48a38d3d22950151cc7f : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93e15946702140ce9474c7ea5fb2f3cc : SUCCESS in 19m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/81b53fa0f0be4ff0a750e7c33c95034d : SUCCESS in 1h 11m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff71eeecc5084f95ab8cf0003194d252 : SUCCESS in 1h 12m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/126631eec3a843f8bd667750803084d7 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b4841e1ff5d444eb828b8221973c671 : SUCCESS in 1h 20m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e27bc8a594f4ba3927e62999185d9d0 : SUCCESS in 1h 18m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f8073e2491f24dafa2947d07ede40ce6 : SUCCESS in 1h 54m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/436ada230a89403981810228c31c3c1e : SUCCESS in 1h 48m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c46060e528b24999b4142c1361982282 : SUCCESS in 30m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fa83734adf46411e8864e91c0bf094c9 : SUCCESS in 26m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa9c4b1260ce4bbe87330d9deea2bb7c : SUCCESS in 23m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc0d21c696b24c0ca6fdf624a72cf9a4 : SUCCESS in 23m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bc345f956c36433eae282a453f4402bd : SUCCESS in 33m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9918b6e67ea7453cb9f1b07b5e206725 : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7373a71d50a94662a4237a11c1c6fcc1 : SUCCESS in 32m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c617a41828e483dba9a55b67fc471cc : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"88918353d00fb6a31cf49a7b035385bead90376d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 13:02:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"218c5c21e7d85244e687fa8b5ff78cbc3a7fe685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 15:02:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e241001bd686453da8d8cb5359e9d16d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ff9cd8b1534cd494e1775bd7091b31 : SUCCESS in 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2912af573857409ba5ea26da7e07f192 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0e26980fbb487fa9660c6c2b07ad2c : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/878b0589dbd141f993548d225d2c8083 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/988079981aa54949b51c7670f69ebadc : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31900fddbc3d46a19feef247bc2f525b : SUCCESS in 9m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db00627a0b1b466dbc6b797310c6b62b : SUCCESS in 20m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9007dedb61e54585a99f87f73872ebfe : SUCCESS in 13m 00s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/73364657819f42e486dfcc59ab4b697a : SUCCESS in 19m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01897804bc2e445f9f02730c08045fbc : SUCCESS in 25m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f2b20109827b4b9abc835ba927301349 : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89576561b126428e975416439a2ff6c9 : SUCCESS in 20m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/39307a7e009d47e3823f3ecac0e5a2c4 : SUCCESS in 1h 15m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e855b01245c64531a696be252393b5f3 : SUCCESS in 1h 14m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e36f17a64e1b4ea88d647199019191ee : SUCCESS in 1h 14m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff3fe06368c64240b5a7e8ab665e4381 : SUCCESS in 1h 23m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d32a6a7f3f82431e9fc12a1386e13b82 : SUCCESS in 1h 24m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c264fc4f1a964ee9a9032237185ac4a5 : SUCCESS in 1h 58m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f4862a83e8e04f4aa8a5580de6a74774 : SUCCESS in 1h 57m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b1e228a360364906a0b6f47ace6f9eb1 : SUCCESS in 24m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f1f710e494fd430e8f7c9ff8533be31d : SUCCESS in 33m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3664b4d2f9204511bdd3c427ac86aaa8 : SUCCESS in 31m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e88c93279fd4a3c9cb3395268750997 : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/302971334ee24df3a2a7e71536a4410c : SUCCESS in 41m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/412ae29e6a6848a98bbed5f030a95999 : SUCCESS in 50m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecbfbbb5a83a467fa9c1cfd2fc88a220 : SUCCESS in 46m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d251da1c5cb4066845b9695539a044d : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2342e2cd30b23f3af0345bf46967348fb0ce52a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 15:02:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f148fd4937726a92e435ed3ba9a3bf794ffdc32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 15:03:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8fdd3703744dbe949a87858fc5e4a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f926d05e4864b49907de8d8441e3d26 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76d4e861b8f4473e8e1fbdf39b009a2d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b625af7b7c8fdbff279dc1df87df3387a26c87dd","revisions":{"b625af7b7c8fdbff279dc1df87df3387a26c87dd":{"kind":"REWORK","_number":1,"created":"2026-09-21 06:43:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/1006443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/1006443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/1006443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/1006443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/1006443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/1006443/1"}}},"commit":{"parents":[{"commit":"3535065278f9fd1b5b5a1401cd24ea573cb9f1c6","subject":"Merge \"CI: Skip previous release testing for Ubuntu\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3535065278f9fd1b5b5a1401cd24ea573cb9f1c6"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-21 06:43:33.000000000","tz":0},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for configuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n(cherry picked from commit a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b625af7b7c8fdbff279dc1df87df3387a26c87dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b625af7b7c8fdbff279dc1df87df3387a26c87dd"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","ours":"3535065278f9fd1b5b5a1401cd24ea573cb9f1c6","theirs":"a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
