)]}'
{"id":"openstack%2Fkayobe~1006444","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I1545baeb7e437b4b0928fdb054b4ad34045357dd","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-25 14:58:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1545baeb7e437b4b0928fdb054b4ad34045357dd","subject":"Add support for configuring Dell SONiC switches","status":"MERGED","created":"2026-09-21 06:44:29.000000000","updated":"2026-09-25 14:59:30.000000000","submitted":"2026-09-25 14:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006444","cherry_pick_of_change":1004574,"cherry_pick_of_patch_set":6,"meta_rev_id":"89f1c25bbca2382ec0772ad19a74591c3dd0fe9d","_number":1006444,"virtual_id_number":1006444,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 14:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-25 11:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-21 10:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-25 11:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-25 11:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 06:44:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-09-21 08:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 10:52:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-25 11:09:58.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"21f534bc17ce3ce7d9990380e2a0dbf02a4126ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-21 06:44:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f13005d91426c9bc7f3e15ccefebe588cd4b65e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:52:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/467b34fc631f401d8c066178553ee101\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6968d031c47c4f81b0e5a7236fb86839 : SUCCESS in 5m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1120ee039204402fbeab48803fa3e6d1 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20a51ca86c854ff09f2e0d178d0cab12 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c3e153ede04b28948b0cf770f50d1c : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d161ca4088c4608a6be5d83aaad0716 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32d42653f31f4e3494ed9840d4c1143d : SUCCESS in 6m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/224b883f836c4f61be81f905d8d33756 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f446ab130a14ad18eb0bc3b2283dedf : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/185ed34f3687446dbc31c411f4110824 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0203cea500248b7ac914265054b8b2d : SUCCESS in 12m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/98be6fcaf0874df9a4a774d7246b06b2 : SUCCESS in 14m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a82f39f94bd45eabb878a4a41576868 : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8079a5ce053e465491f13f5bbb58de13 : SUCCESS in 11m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb2276f25ee74c56b2573546662f13bd : SUCCESS in 15m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/282796306de74c72978af8cd91c39a5c : SUCCESS in 1h 11m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/edee462fe6554c0784a0af19d311d4cc : SUCCESS in 1h 11m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e5a176013a594096b0fb1f26373aeaef : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39f8e9abb2f548c3b20e64bfae2cc608 : SUCCESS in 1h 46m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f2e68863f3c4a6a99b22b2b1876089a : FAILURE in 12m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/29b4b3156d424df2867187380ad97236 : SUCCESS in 1h 33m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/636a2c112f184ef2bbdd18af6cf8539d : FAILURE in 2h 07m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/319b91a2ed0040f1bef4b957f0d69c1a : SUCCESS in 27m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/850de54c0ea64301a48414f58647b783 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9098242e9ebf4df69850438fea1d7728 : SUCCESS in 23m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/25720783ab6e4c8c8fbeb00f50816670 : SUCCESS in 15m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/13844bfe02ba4aeaa158c4f1be9bf8cb : SUCCESS in 18m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e04a1369288d423d8d82ec81e93b8625 : SUCCESS in 30m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/64a659515f3a4cbfb61419ddbb80dc02 : SUCCESS in 20m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efa178a6f82747758048e8da7af532b3 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9da08a97ea2fe5e8aa23376b07d373d0a6b1bb83","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-21 10:52:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a07090768a26027f1feedc10d4469bc6150a9d0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-22 07:57:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"214617fb5389f07f827b30e50f8afade4d46adb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:36:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72cd726c29ad42ac81e6b628d7afc4e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b765a69893046489d74ca51e7f811eb : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/40654518565c46c3bebce0dbf51cf4f7 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a29d8b70084e47b5a5ccc6164a2414 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec11bc9d52e94e4697bb38752f8a83b1 : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fa6b71d03c846a09d5b8bfa462b96f8 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36de445978b44efe9e5640ebe5135f68 : SUCCESS in 5m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ff4138368864f49afd397236fe1e3ee : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/318054a051a14714b1cc9f2099795f4c : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bc419e062b946cd901fb3bf97816fc1 : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b2eda2c57d3463da390678d51334efe : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4f950695043b4ccabf2176121f5ccc23 : SUCCESS in 18m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c38077c67ce49968f8cdac6d9e377cd : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c97559fb5bd47dba230e8762abece6b : SUCCESS in 24m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57db172c26204b9997dd4d53e0f47818 : SUCCESS in 22m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f8e8b7c0dff54f0a83471d56058f25d6 : SUCCESS in 1h 12m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1408930a537b4d6f91234190fb9b5af1 : SUCCESS in 1h 12m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/03951dd1a8e741e3ac026ecd5813f396 : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bec9d935505413ca64a282dd474dc30 : FAILURE in 15m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fdb5e00f6b1641d190bf309f75110c4a : FAILURE in 24m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc8d4de5fc9747aeb9805c2f7e9e1b56 : SUCCESS in 1h 37m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65f52b89bede47389240d8e7e3a07468 : FAILURE in 13m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c7971de3aad9480185f4da8fca44a81f : SUCCESS in 20m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f32663ea28cc4cbba38f7d2639ba6548 : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00be5e4e012d4161b5dbf0b109edd1b9 : SUCCESS in 25m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87286a96c1044f57879eded2cdf41970 : SUCCESS in 22m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/317b5dd00a6c40158a741076f4041588 : SUCCESS in 35m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e27d96b983574e2396b5e8de8756779c : SUCCESS in 34m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de80ca30da374ba9b2a543520d4383ae : SUCCESS in 22m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8714745fd9e34b3da06c9dc06d6f54f3 : SUCCESS in 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d1ab36f8a48985e85bda952662965b10c8eaa6d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-25 11:06:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c1aac4a326fbb47760bce6d6eca4de4147b0da4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 11:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4b0fd01d875050da7e901ab9dcdbb4700a4ef7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/926755eecb414728b92761fb6bc8960e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb654c40f2c54447af8bfb950ffa5382 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3f974a159244642b0fe40acde1c530d : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba276bcdd7c645bcb29000597f75fea1 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811bd0b8d34e47838c1963e66a9634d1 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b16a9c8be624d60a546c43277858c59 : SUCCESS in 6m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be27d560a5e54599a71f75dc4e735623 : SUCCESS in 5m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98690029cb544b19b01ff7bbf5cfc988 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2ffccc5f88448bd98f7dc3b6d2e3da9 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1c5b978b488477a9e9feba059dde354 : SUCCESS in 13m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ef0dd91309e40e099f9b3b50056722d : SUCCESS in 12m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b99f9b8867854098bdc9a9ce6943cb0b : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5a7aba1df1f453f8a08d6c56bcb421b : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d7511edeb1384a6b9be82e550f07c440 : SUCCESS in 18m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efda585350a74ce483cc0cfca41dbb8f : SUCCESS in 19m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9dbc20bb6630468397afc4dd06f65b64 : SUCCESS in 1h 12m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ccbb4217b644328b01b5fd260a1dd18 : SUCCESS in 1h 11m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/574adf0aae764bbc8930b58f1f748601 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faa8662b60c549c08af37151a35e694c : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df97396f2f214060b190f9864fdc68e7 : SUCCESS in 1h 15m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b25ec1f33784a1fa497e8b7e081b598 : SUCCESS in 1h 35m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f3f8aa98e55440593be8c4661a6c36d : SUCCESS in 1h 50m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fde03f025e684cbaaf80b57140d333f4 : SUCCESS in 16m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eeeed9a857c84990a8d94bd994a74980 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/015ab30b241c421cad74d6110b2d9c48 : SUCCESS in 23m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a2ab8e13ff748609c54e7f7da7c5c7e : SUCCESS in 26m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/364ecc7b964d46e9b42c58a0497b4050 : SUCCESS in 41m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/835b842f32cc4c66a115b3a49a165529 : SUCCESS in 39m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ff3674e4d493412582260a20517903b9 : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/447349471bb3421aa049768322dd463e : SUCCESS in 24m 09s","accounts_in_message":[],"_revision_number":1},{"id":"56a8a8e1a374a947af1263175d811c24bde21e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:58:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a582fe0a1b9e52daff0657ef974167f288b4d6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 14:58:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2f4cde338a420eb0240e0738ad49c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055fab7076114edb851b452abef43a7b : SUCCESS in 5m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5cdc518fab3646969a2557c750e40fa4 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b85ba7f74bc4ed9bedccd47dbf172f1 : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84aa4d9d9e024e6baf2224c4493e70b4 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ce5b08d66a84f2095d3146402efd88c : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5ff90c127be44a6490189599e4cdbc40 : SUCCESS in 13m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a4c126e347349b18c212a6afa26f882 : SUCCESS in 21m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6c1787b85b04f869021ab13f22fd395 : SUCCESS in 12m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b81cc507596040b2b43da54fb1b22309 : SUCCESS in 21m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/618fdaa2e70b4eeea3afe058bc5dadf7 : SUCCESS in 26m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da8658f60a7d44258784e609b957d324 : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95be4a88b8ef4070843d2fbdeb087825 : SUCCESS in 14m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/937992590e8b4c1b916ea82f611cd4cf : SUCCESS in 1h 23m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5c6a47086d6a40c3b050279c64eec0cb : SUCCESS in 1h 13m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a4384f645a994ac8a21cabf5379c94f5 : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3197cecbc8e841db971ee3fb00ffcd49 : SUCCESS in 1h 22m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4972cd732ca744ef9422598513695ab3 : SUCCESS in 1h 18m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bc6b6c752dcf47af9e4eb30806a4f244 : SUCCESS in 1h 48m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6017c92ec5704479b6c2f87550177050 : SUCCESS in 1h 58m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0f32c93d301a49c999138a3d3ddc6350 : SUCCESS in 33m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8b1470b3432245a78b7446d802ac95e0 : SUCCESS in 23m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9881688f79748b0a7ff77f66ab813b6 : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41d687f36d154d1c92a9cc3e7b606ff5 : SUCCESS in 21m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e3c96d4d7b4e4f2db8024d3446e49064 : SUCCESS in 38m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f284f1cb507940ceb46eb2c36a79bc3a : SUCCESS in 34m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3b06323dd8424181befc381e22b2dac7 : SUCCESS in 27m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c130054a7784eeba900c57cbaa98a4e : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7c426b6ef4efa8a4971864718fc799a6a49be256","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 14:58:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"89f1c25bbca2382ec0772ad19a74591c3dd0fe9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 14:59:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b14b0dd5afce43c0b66399ea792dc4cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b782b9a29d43319deb4864db158e10 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2294f1e54c0474891f5ce4321b897d9 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0bc2deffb5fb48c3830720a36bbe3202d6ea738c","revisions":{"0bc2deffb5fb48c3830720a36bbe3202d6ea738c":{"kind":"REWORK","_number":1,"created":"2026-09-21 06:44:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/1006444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/1006444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/1006444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/1006444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/1006444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/1006444/1"}}},"commit":{"parents":[{"commit":"1a0481e80ee71e9c89c5260df7550f73eacafe7f","subject":"Merge \"stable-only: Update chrony role to v0.2.0\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a0481e80ee71e9c89c5260df7550f73eacafe7f"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-08 09:09:49.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-21 06:44:29.000000000","tz":0},"subject":"Add support for configuring Dell SONiC switches","message":"Add support for configuring Dell SONiC switches\n\nAdds support for configuring Dell Enterprise SONiC switches, using\nthe ``dellemc.enterprise_sonic`` Ansible collection. Follows the\nsame conventions as the other Dell Switch roles.\n\nChange-Id: I1545baeb7e437b4b0928fdb054b4ad34045357dd\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n(cherry picked from commit a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bc2deffb5fb48c3830720a36bbe3202d6ea738c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bc2deffb5fb48c3830720a36bbe3202d6ea738c"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"d9cbf9277581b0eb2b9bd241844d75a1301f7a27","ours":"1a0481e80ee71e9c89c5260df7550f73eacafe7f","theirs":"a16e973e64c3e9f214c4cb035d8a2ebb4debc5dc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
