)]}'
{"id":"openstack%2Fkayobe~1006515","triplet_id":"openstack%2Fkayobe~master~Ied168ab752519f464c9863ef6f1cd1486530546e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-29 22:18:02.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-09-29 22:18:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied168ab752519f464c9863ef6f1cd1486530546e","subject":"Remove Swift support","status":"MERGED","created":"2026-09-21 13:00:45.000000000","updated":"2026-09-29 22:18:59.000000000","submitted":"2026-09-29 22:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006515","meta_rev_id":"328a1ce6ba39d4b63e27af8c45841c74a2dd8584","_number":1006515,"virtual_id_number":1006515,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 22:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-09-25 16:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-09-29 07:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-29 07:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-29 07:10:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-28 12:41:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 15:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 16:29:45.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-29 07:10:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-09-29 07:27:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"c9ac76af7a92f75d7ee1a1cfc3cd6ee172c5a7f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-21 13:00:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54ad30c2e61ecef93ede846aecec43a7690e20fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-21 13:01:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d5ec53d638a6e9678531b55f3be94202e1f93af6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:02:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32dfb1a0a62d4a349b459acced859324\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc39fd8253bb4d64b54d5c4cbdb39d88 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a68727a79f48c9b2ad30f9c408fc43 : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d218fe79e7c4466b73f4f8e75ecc231 : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d0960c00cb942f3a15bd799ee78da7b : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69255fe73b584cc0be026cbb9ce535d3 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aee615aa1af641aabcf11b685aac6351 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78c86579f1274022b1ddf4acd4b93633 : SUCCESS in 4m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/348425539e6243278bf98e96c429a717 : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b648098eefa488f8c8f6bd66bcdd51b : SUCCESS in 8m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7bd68bbc8dc476793b325aec73d5e48 : SUCCESS in 14m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fb452f9da6b456895b254055bbba01a : SUCCESS in 12m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3410e48534034fc89d95141a4f9a7970 : SUCCESS in 16m 28s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4f46366c73eb4a8182ab51aa11d7566a : SUCCESS in 18m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb424401de66443e830464ce08528cf3 : SUCCESS in 25m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6bd1f5a2034d469693b28e027d808b90 : SUCCESS in 23m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9e2e8b2cb1724912a7885a95a56e0cb0 : SUCCESS in 21m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6964753a10ed4519b392721e98e19a5b : SUCCESS in 24m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/24dd59630e9c44c8b9f37fc7d62d7a7a : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2178c449fbd04ae39d6bebfbb3b81fb8 : SUCCESS in 23m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f0b6771d82df4c3f8e8912dccddd2a05 : SUCCESS in 59m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7779009e76494b6eba559349b37e03e3 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2919fb7623c94a01be52b7f1e926e7bc : SUCCESS in 1h 11m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ada1808f4354890b0f77bf4aa3996cc : SUCCESS in 1h 52m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a3f31bef24c417aacbd50fda49250de : FAILURE in 43m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3d350a78c99d4d02b2686d9b18c3cae2 : SUCCESS in 1h 59m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/43eaa88b9c4a4dd3882221503c7d50ba : SUCCESS in 27m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/63b2a588867c437f9836de406320e743 : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc7b0cdcbacc474c953cd8dfe28c8d24 : SUCCESS in 28m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ad79e35401e84d3fbc4aa329d970cea1 : SUCCESS in 25m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9a3fd4ebde1b47369994cf400f893182 : SUCCESS in 27m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc1ec38e2b324301b1bd886986b826c3 : SUCCESS in 38m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5e18b0a13294e9ebf82893136c05113 : SUCCESS in 24m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/322f82b00d064703a3deedf006b80dca : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a57b7dec460646f90b976deed372b01ae31854b0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 13:57:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd07475407e5974eb3e5749ee8f6c449fd6fa087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 16:03:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51438ebadb2e450ca3302cc6de691adb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f40e5f696d5448de98365617df102902 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9bd8d44fb9b4e68b64f4d1abff21b9c : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f02b1c3feb3f43b69d138c3676217874 : SUCCESS in 11m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3447b516af0e43eda48bb2ffd4ffef5a : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6db7d850b074486a17c2a2efcf39ffa : SUCCESS in 2m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d92d59f29e41446fa197fed1f45f28e4 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b195cdf43acb49968a2b6a3339977224 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ea4c74aa776417a815ef4732f8682ca : SUCCESS in 9m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e548d1fea85145f583d7299cb9df2854 : SUCCESS in 15m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/634ce682401f43b8b92a99913f63c841 : SUCCESS in 13m 08s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6423eabfa35f47879917765598ea0a53 : SUCCESS in 28m 15s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4dc0062dc28b4be884c52ce62eb0b28f : SUCCESS in 23m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca97b6b11960455a82f92b7e969b3dad : SUCCESS in 23m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5036b9a15284401e982d47d166fda575 : SUCCESS in 23m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dc781470103e4c8d9b81a3860212a26c : SUCCESS in 26m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb65d7114c984dd5aaf820b792f580c8 : SUCCESS in 22m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/de2a51abd64149e090c523df8e0c4fd0 : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab37cb1e71714de39f5ae08af4fc6b7e : SUCCESS in 17m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5e8250fa54734d98971cc9f1acc3d74a : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/636c584c963043fa9913dda10c163be4 : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d79a702e45814e30a130e388e19b66b6 : SUCCESS in 1h 13m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6201b41ce43d4190ba766cb13ac2b2a2 : SUCCESS in 1h 14m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ed9707780954e3c86ce60fdb0b94513 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1b332019f64049abbd20d40f7b48b96a : SUCCESS in 2h 03m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/14887a1007cd4a91828f0637b89c3686 : SUCCESS in 32m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/03c6b6f9acb04a44a9cea611c1981147 : SUCCESS in 20m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46a59964e329409f9e3e39116305394d : SUCCESS in 28m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/268c3084d66942a4845eff978ff0c69e : SUCCESS in 30m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4fe4a6bd86ca46059e86031c9f0d920d : SUCCESS in 29m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b906f04819f7433c8964bad98bfbdedd : SUCCESS in 34m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf2e22aa601f4a8fbca011fb86fbd6df : SUCCESS in 30m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfe4d7b4e2cf4d869c552f3b998507c5 : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":2},{"id":"09a4b7c8d271a80289c84eded8dbb36f83a3bd2a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 16:07:30.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"e0ef48f3b48e208ca6e9200c2b502ff070ad9d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-25 16:28:59.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bdf8febee10bb22e0065020aa9f667e79bb07003","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-25 16:29:45.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba91b60584bb8fbb47171110db2571ee679323f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 18:26:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af0b56be37f24380981559cf16f16cbc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f21eef69b794c009ba71085082a9c84 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30cc2aa2aed471d8c3b191486a91b2f : SUCCESS in 6m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/22ca204980f44b3fbbfa7667ef28ff48 : SUCCESS in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/717f7393683145f08a1118c885471143 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2883f38460f94bc8b7d8b06cc5533ebb : SUCCESS in 5m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa36c395d8244c1ebb687d733f694fad : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efda713b475e4521b53004f0b27ac559 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c27cbf25417b45d19f46b42e0a53a874 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6927f74c900a41e9ae4096dc19a43236 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98e3b0011d8943b8821a7b0e376015d8 : SUCCESS in 12m 47s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3838e00a1fb1455782cb40f443307018 : SUCCESS in 14m 59s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9b21188451194089b57935098ca588eb : SUCCESS in 18m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c5e6f6e136f4c46b1ad07e0c2d39045 : SUCCESS in 20m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4486932a731942d3a4d8afe277ba9ce6 : SUCCESS in 13m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5a2442854cae4657b1579f80086618c3 : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b18c5ec9297441ba234d5a817ed8dc1 : SUCCESS in 15m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/abcfc00b1b564409a655dd50810cd07a : SUCCESS in 14m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6eef3eb2954b4ff09c6aae30d95898da : SUCCESS in 13m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ed7f87d6b58f4b60ab96610dc4f0c93c : FAILURE in 1h 20m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/92d484cc3e8b486fa7c9eb2c6d5678a1 : SUCCESS in 59m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f9f7215b87f24156af322f45022fc303 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2035ed0945aa4189bdc4bce17991fe3e : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b9ebe8ce6c842dc82269b19abb31658 : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d53e5a4c55ac467295c67fc4a391cd4b : SUCCESS in 1h 48m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5c97d537472544c89bc7f6ddffbb7fc7 : SUCCESS in 22m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/11d834e601a043f695df1328e4b5acb9 : SUCCESS in 22m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba2645d423d41b787e842e069b9f120 : SUCCESS in 26m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bdd2d7b8d9e4494ea52b32a281dcdb8a : SUCCESS in 24m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/09bd15493d5a41fca490d6326d4a6c4f : SUCCESS in 24m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa2ab941af7e4ed18b3faa27ca89e31f : SUCCESS in 39m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/415c3c16db5f4364a18a19a0fddf4053 : SUCCESS in 17m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f71ff35569b64f5d9ad91bc0e1a0367c : SUCCESS in 23m 15s","accounts_in_message":[],"_revision_number":3},{"id":"d4201232f284f30ade9bdcfb7b0760e34745dcd3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-28 12:41:31.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bba580dbd5f40264ecfdb53b3ee01a00b39e59d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 14:50:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11544850f025409b830082169e1af740\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/698c3dd6177546f4b4a75bdfd6a6d644 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7ca08d29524c5f91a51efcbb12fafc : SUCCESS in 10m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52b9b9921cf2461f8b5c973205742c99 : SUCCESS in 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe19d15b301b40ef9d2092e73f941f6e : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9efc570fe4994557be0202e548c89fda : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ccb1fa72a8b4093bd315cae4a14ba8c : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a2c6aca33ac4b1a96cbc441b3039fd4 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c55bd32be5c549ac89e8c6039ba5b5b0 : SUCCESS in 11m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab0b5787fbaf4129b14706acac3d2df1 : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0f0b0228fe04aeda22d0737483dd404 : SUCCESS in 15m 24s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d270ebcdadad4db88d59bd3abfe186c1 : SUCCESS in 26m 44s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/ffc4f41d93924cada25954785921b7f1 : SUCCESS in 32m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4de63970e5204a859a3706724e9da02e : SUCCESS in 25m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0aa062c07aff433b8ca33e2f3ff9459e : SUCCESS in 25m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ecc9fe1541254441b6a445db62e621a9 : SUCCESS in 31m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7582bdbe23743609c3eb50058a2f5f3 : SUCCESS in 29m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3a407bb6f61e49499c8aa4c2c5502027 : SUCCESS in 35m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efefb2c8e32f45cb81a6a44d4e5eaddc : SUCCESS in 32m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f05d482eb2fc4c99b1bd36b80c3842de : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/420e7e1db8524953b76c2509301610e3 : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/25a37fa5d765465ea293e5f24abb532b : SUCCESS in 1h 15m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a35f7dac1c4464eb32acd37027b58d7 : SUCCESS in 1h 13m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba3461f6bd6f4bfabba71933255a7aa4 : SUCCESS in 1h 15m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/99733cce206c4902a12a1d2842e5cc59 : SUCCESS in 2h 07m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e4e609456bc54e44bccf1de5fc6322bd : SUCCESS in 34m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/35c48a6e6ad441b9b3dbaf7bc9d6e9b3 : SUCCESS in 25m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/911bf85acba4457cbc4877ef0e6c9f1b : SUCCESS in 35m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4fd44e6c5a254b0a994f29b7e490804d : SUCCESS in 31m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/14b8989c50aa4883b49ee600cc70242a : SUCCESS in 37m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/698ee7f0c88849b28885dbd73dab4a5c : SUCCESS in 51m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a0f6646d179c4edf8611f702f4f6a46d : SUCCESS in 26m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94463b8a138940328c4cf4a75ec87151 : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":3},{"id":"8eda6d73314495d6cfe7c99a7e81d95743aae179","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-29 07:10:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f840f7e1af32c3c73af11505f24fe960703ab204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 07:11:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f83655a365e403714bc4b9332bca81c22a86b801","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-29 07:27:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"73cbc841be52ecaa9dad8ad9cee2ecc4bcaa2bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 09:18:30.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/371042b0eaf04677abd833244a0c73e0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b215e38f45024fad8e7505c7640fd4ff : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e371c41246744fa49c116432fce8fd73 : SUCCESS in 6m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8afd2063693c4b93931322e896e2ffaf : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb57cdd0d2e4907a706f7d3a8b78b48 : SUCCESS in 5m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83085461a80a4c74a585083150518da9 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1532e1dbe98b4df181db658f1cc34b29 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95da00b840794b20bdda4ba5f5949b1c : SUCCESS in 11m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bdac31093ed4661afbcf533a72d0428 : SUCCESS in 17m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16d82a3fcb3445208d55257814b56b3a : SUCCESS in 14m 46s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/feb0b44d734948e483f39c6de6381e00 : SUCCESS in 33m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8865dda2288f4f3e95074317c3003f54 : SUCCESS in 30m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2b565472cb4548938c90509db6babe63 : SUCCESS in 24m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/209503602a6841388e7d4725530d58ff : SUCCESS in 33m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a71e6ac8a62946c09615c49a446eecde : POST_FAILURE in 36m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6a5f53ae9564f0d9b85e723d3faefd6 : SUCCESS in 22m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ef999c6db26e4ffbb6bcdfcb5018d167 : SUCCESS in 1h 15m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/53271a7d3a634333a5e043f73dabc253 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/800cc0080e20457c96a5d4d785fc514f : SUCCESS in 1h 15m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43f4ca66d9724adaa0ea71c674aec521 : SUCCESS in 1h 15m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7aa307c36cc648df8a8f563f724793ea : SUCCESS in 1h 13m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/36c38b9056a34c498f59bdfbac426c64 : SUCCESS in 2h 06m 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c67bd2626aab4c6f97336037cb3a28b3 : SUCCESS in 25m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/064ae3af3ab84acb9c611cc7e228df50 : SUCCESS in 35m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59ef90cffd8742b0845b043f8c3e932b : SUCCESS in 33m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/125afb4a64f94b20a7c2414cb1ef3af4 : SUCCESS in 32m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c85361be9b2145e3b7ad09596b322d54 : SUCCESS in 37m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffc2128176034b99a0e74bf74bece75c : SUCCESS in 35m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c53703be74d04943a74e39837e9da78f : SUCCESS in 31m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f25fb79660784a37a2574c8576bdf2de : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":3},{"id":"98f403a668cbfd46ad78b2d6077af5ec4b2334c4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-29 10:55:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f40dceb6685b6ee67fcb3b7c1f80d6bed9a5dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 20:10:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed94ce968b304a998220d0b96f469f8b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca559285edb44a59afa17c34738b9edb : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833200d39fca49488b0db120f7b9d45d : SUCCESS in 10m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cbe531970bd54434bfc52b18266a0703 : SUCCESS in 14m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53700b353c8b48cf98f1220fdb793b93 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0159fe0acd64f19bb316934130e2938 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/91ea0607801b4f40b75ac24b637771ec : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb906c227664489da7b28146c6aa9ef9 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab38d87c7e2f4158939f22ee9f67c7bd : SUCCESS in 14m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eef551c09c8c41a79e2a6df28dd3d79d : SUCCESS in 26m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f7554130ac841728fb7ab38e43324d4 : SUCCESS in 14m 49s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c5960087f5d646598ebfca33ce911839 : SUCCESS in 32m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/63efa90be1c04239ae928c425e02183e : SUCCESS in 30m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce1ce43c10d94ea6b276c1f2f01b1e16 : SUCCESS in 30m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/71eb3859453844568f99cd86a3937b8d : SUCCESS in 26m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f111ed355104db2acc3608ae96c7477 : SUCCESS in 29m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74261351b2ce45af9de2d9f74d7f9983 : SUCCESS in 33m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9ca90edeedbf4b82bc42167c2b7af73b : SUCCESS in 37m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbbba7ead876423c920db5dd73868a31 : SUCCESS in 31m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6236bea6530647eba0b55c14ef3e6532 : SUCCESS in 1h 10m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a89367a093d342dca9583819139b827f : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1ba00220a5a642d2aa27a78f6175185d : SUCCESS in 1h 14m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e371379ee994c44b9998f54b1bd57f3 : SUCCESS in 1h 25m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ac3e3abb56b4371bce74975b2bd28b8 : SUCCESS in 1h 32m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/16fdfa64a09648d19634e0b344cbbd24 : SUCCESS in 2h 13m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd73feec786443888652b60ad2d0e0ad : SUCCESS in 2h 13m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5988c06b913146b8a1e32494fd47995a : SUCCESS in 31m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54ff993e56bf433383ea461369689f08 : SUCCESS in 28m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40690141648045a98261dc62d01a7b5d : SUCCESS in 38m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b3ce6df5a8f478ca674c016a507b2df : SUCCESS in 32m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/08c0701249ad40419c6a7c0e360a3a1a : SUCCESS in 51m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8c3afb8d5cf4961b638d822d0b7da4e : SUCCESS in 47m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4351484b7f534d8bbf52d3c996cac04a : SUCCESS in 33m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/225dc0b576904752bd28c71a6fc8c532 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"4c410f0b4c5987ee3b6c7b9e056ea9736374e00e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 20:10:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"737091c75ce81787b68e206875b5aedccad1d0f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 22:18:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f4b4eac9504f21955a4735e0bb2d97\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6e32b47ef724b3e8aaf5876b577ca90 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16141c1d61b747ec90552c1db6fa0ae6 : SUCCESS in 10m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/33532d47b1dc4657b85ed7f0ec4ffb72 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6ad8300f964596a5cb738f4edbd91b : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca4b71dbbc6e421ba35e9df3d97bc55c : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00acfae184b742dd86367a29c72eb42e : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/057500970f474d2a997762b843add6aa : SUCCESS in 10m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c9e67b4089641909369d76b50208b9d : SUCCESS in 21m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52ced9cda7d5485692e881f997cef403 : SUCCESS in 10m 05s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c2b24593ff7943f0861ca7201f5542fc : SUCCESS in 23m 02s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85ad0e6893d344678bde9e2c48b75c18 : SUCCESS in 29m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26241dcfc7824dcb90d22f0ce0989696 : SUCCESS in 25m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1be3dcaf72694766ac708cbeb2c80ef1 : SUCCESS in 34m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/04c7cd85e590427e88fee340dd3304f9 : SUCCESS in 30m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d578acf0a2b4cb095978ae69b335af6 : SUCCESS in 36m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/446f63edac3940d6a604cd5b8569a2fc : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/74d842feef50449f80e716732c2f6e59 : SUCCESS in 1h 09m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/67322c23746e4d57be602ab20df1eed5 : SUCCESS in 1h 20m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb6e606360a3485b8caeed1501fb582c : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bc3af545865483ba7ba19120913d239 : SUCCESS in 1h 20m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/82850b8ede9d49b5bd8684a61c59b2aa : SUCCESS in 2h 06m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b72b49ffe054afeba6c0dc68c58e2a2 : SUCCESS in 2h 06m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/82dba1f1055a403f8f8f2c32e4bd78b6 : SUCCESS in 36m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eae1ca27649541a2a498660eb289e9aa : SUCCESS in 32m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/564385c7d10c49e3bac362b2ea9f5094 : SUCCESS in 33m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8505f26b883c42469178cf5a75df9ed0 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/781543923f6e4848bb4561c68bf8ad58 : SUCCESS in 47m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5b71a4643934dbdbf8aa31084c86ece : SUCCESS in 49m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/37b71679b42f411f8a634e6669cec2b0 : SUCCESS in 26m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/750097d3d7564568ad6179dc5138b7f3 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":3},{"id":"61d30230c249fc0dd10ef2e839c9eb83da0f86f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 22:18:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"328a1ce6ba39d4b63e27af8c45841c74a2dd8584","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 22:18:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5362bc1ec7ff4a31aca1e07196c2b6df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da836689308440088ed84e5b618ac9fa : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb9f7955f6b64e4a9251414e3b2d37fc : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e4ab4eba193ae5a00cafd353371da08295084802","revisions":{"a5b0774c75e0666cadc3a0512bf0a070e8ecfb18":{"kind":"REWORK","_number":1,"created":"2026-09-21 13:00:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/1006515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/1006515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/1"}}},"commit":{"parents":[{"commit":"93b2819d8fb67b56da2e5c080807864cb5055d38","subject":"Merge \"Add support for configuring Dell SONiC switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93b2819d8fb67b56da2e5c080807864cb5055d38"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 12:51:12.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 13:00:35.000000000","tz":0},"subject":"Remove Swift role","message":"Remove Swift role\n\nKolla Ansible has dropped support for Swift [1], so remove the\ncorresponding support from Kayobe, including the swift-rings and\nswift-block-devices roles and associated integrations and CI scripts.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/945113\n\nChange-Id: Ied168ab752519f464c9863ef6f1cd1486530546e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5b0774c75e0666cadc3a0512bf0a070e8ecfb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5b0774c75e0666cadc3a0512bf0a070e8ecfb18"}]},"branch":"refs/heads/master"},"52d6a3698fccfb9f935f583a08a52526689fdf1e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-21 13:01:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/1006515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/1006515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/2"}}},"commit":{"parents":[{"commit":"93b2819d8fb67b56da2e5c080807864cb5055d38","subject":"Merge \"Add support for configuring Dell SONiC switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93b2819d8fb67b56da2e5c080807864cb5055d38"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 12:51:12.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 13:01:25.000000000","tz":0},"subject":"Remove Swift support","message":"Remove Swift support\n\nKolla Ansible has dropped support for Swift [1], so remove the\ncorresponding support from Kayobe, including the swift-rings and\nswift-block-devices roles and associated integrations and CI scripts.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/945113\n\nChange-Id: Ied168ab752519f464c9863ef6f1cd1486530546e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d6a3698fccfb9f935f583a08a52526689fdf1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d6a3698fccfb9f935f583a08a52526689fdf1e"}]},"branch":"refs/heads/master"},"e4ab4eba193ae5a00cafd353371da08295084802":{"kind":"REWORK","_number":3,"created":"2026-09-25 16:28:59.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/15/1006515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/1006515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/1006515/3"}}},"commit":{"parents":[{"commit":"93b2819d8fb67b56da2e5c080807864cb5055d38","subject":"Merge \"Add support for configuring Dell SONiC switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93b2819d8fb67b56da2e5c080807864cb5055d38"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 12:51:12.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-09-25 16:28:50.000000000","tz":60},"subject":"Remove Swift support","message":"Remove Swift support\n\nKolla Ansible has dropped support for Swift [1], so remove the\ncorresponding support from Kayobe, including the swift-rings and\nswift-block-devices roles and associated integrations and CI scripts.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/945113\n\nChange-Id: Ied168ab752519f464c9863ef6f1cd1486530546e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4ab4eba193ae5a00cafd353371da08295084802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4ab4eba193ae5a00cafd353371da08295084802"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
