)]}'
{"id":"openstack%2Fkayobe~1006569","triplet_id":"openstack%2Fkayobe~master~Ic343f22b1f3ac8af9a245664a985849b6d38cfdc","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-23 13:13:22.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-23 13:13:22.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-23 10:00:24.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ic343f22b1f3ac8af9a245664a985849b6d38cfdc","subject":"Encourage agents to identify themselves","status":"MERGED","created":"2026-09-21 15:15:50.000000000","updated":"2026-09-23 13:17:29.000000000","submitted":"2026-09-23 13:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006569","meta_rev_id":"d80d5bfc7c23852e330e6b11a73fc99b4f826ad9","_number":1006569,"virtual_id_number":1006569,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 13:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-09-23 09:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-23 09:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-23 10:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 16:55:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-21 17:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 09:42:11.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-23 09:45:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c11081c335c94259ac487a1442887e94836e3aee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-21 15:15:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6351bc56697a6ab4d9594e0090cb8b098a8022b4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-21 16:55:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df90d16d9a8c692dc10197bf7f2c5c790ba291a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:16:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36b3cfd7b95940d9bc331b7a6be9cd75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc078d33cae4452db7e33fbb50a90119 : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38aefbb94c0440d4ada8dc69d0cfc216 : FAILURE in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/860b9750f6484cdbb06bec8f3d5beec6 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85727a880914c71be0ed8c5fbf9790a : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5e235153ca4456183b14c560fea4034 : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2059c33a5804aad9ee01cc8f40e649e : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f4c1ee06a2439d8ce237f5370bd03c : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9ceddf95e08466da9b93f00cc0680ac : FAILURE in 7m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5e71d0326cc431393e4cb5ac01d7254 : SUCCESS in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5011895aacc74f239a1f620d1186360b : SUCCESS in 12m 40s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5528f4ef1ac94ea9bff5b72d7973ece5 : SUCCESS in 24m 46s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/df80ae6faeb94b67bb41e89158ce542d : SUCCESS in 24m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e037e7909e0f4ceab5cc1829e28667fb : SUCCESS in 16m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ab58a752aa24643ad227cf6722f4fd8 : FAILURE in 8m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f0c8332752e24feeb2eaa6e9d1723aa7 : FAILURE in 6m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4eb68a99f484ed2864172a60c0ab52a : FAILURE in 6m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3711bd359a664330a11877f44b59ebae : SUCCESS in 15m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b65ec6e27ae148b1837202f039a8c1a3 : SUCCESS in 26m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5f3aa6fb1041417f80a02ef97f8f3e3b : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c4f71b7d8a974668b05bdb2914088d6c : FAILURE in 8m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6813c47af7214d8b9bef37798f4f637e : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cc479f7eb5446fbb8118faf298b0a5e : FAILURE in 40m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/751534cb3e6c434cb297f61e48375e01 : FAILURE in 25m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bf1c5fc6b8514901b02dde6959786260 : SUCCESS in 1h 56m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1f6e0fac735c49048ab2de616fecec77 : SUCCESS in 18m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/53e220b8eaa845f0b647ef89d0f00d98 : SUCCESS in 25m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d196ceb73ea04261ad609c7c7518787f : SUCCESS in 24m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06cfba7bd612488697e88ae9bed5ab0e : FAILURE in 5m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/deb063b334bd4eaeb545a83ead17d8f8 : SUCCESS in 26m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63c408d6c5464ce9bc465d8de6886a62 : SUCCESS in 24m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3dc5c086f0a74beb9e9a963ce2c635cf : SUCCESS in 25m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6809ce1cca994233bda72d65802e36ba : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"dca078565c3dbf485d6b6e0fa151c3abef3a0118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-23 09:37:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"65346e09f398d0ee99aef3e851dde178aefa80e3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-23 09:38:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c7c30f66857d1fd1cf8819f74b7d19a2e1e8fd3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-09-23 09:42:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61a4b2c56bd22231542d92e1323d88d609e29785","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-23 09:45:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"19c28cc955faeb60e95aa31204e682b69f88888a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-23 09:50:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"821b88fdf4290027973d0f553cc4a9be4540ffda","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-23 09:52:05.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81231c4cdcc0da2c1bee9ade5197884e4ed69860","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-23 09:55:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8cc5be6f5b37bd2f8a96f63c70a0946d95511c63","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-23 10:00:24.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bbcdc233b093657282314b8f721bb2af196038a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 11:08:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100fb9ffc0c9420a9a7430d6e66707e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10173fd760f8487fa7d099683d53ddab : SUCCESS in 10m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d378faaeb6284a4d9cdc8e0752641617 : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f92a4593f7ee4971ad181706c96f0089 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fa9f256d2394ba7bfbbfcac636e306a : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0f8b41f54934bc39f918648491fb755 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f28508149114ea3a72ab8cc63e2c14e : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9b727e5ae04ed94d8191ac2db7a887c26eff398d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 12:55:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"84704dd248c5b17c99c6401f6250dd8dc12ff53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:13:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b586b1e95c67400ba410e2c568d1fedc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becbcc7f77c84c4fad9176d8ec5670e3 : SUCCESS in 5m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fcb302fec8604be298dc5f627991a110 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75315887725244afac3bec86e454c96d : SUCCESS in 7m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4297efbd6b84c3e919a9597d6321171 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23ff93907fdf4a6aa329731c2802bd7f : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"15bc8bff202cdcadb6a8a3fb58f08d8e1311b16b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:13:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d80d5bfc7c23852e330e6b11a73fc99b4f826ad9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:17:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a02a2e8f05b4201b52b969d087b810f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4acc6461bc264b7c82d31bb9c08d6f34 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a19c474a569fbf0988553029adf49c03deb3286","revisions":{"2e1d0380b1aa106973d30037407adb7db45958cc":{"kind":"REWORK","_number":1,"created":"2026-09-21 15:15:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/69/1006569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/1006569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/1"}}},"commit":{"parents":[{"commit":"93b2819d8fb67b56da2e5c080807864cb5055d38","subject":"Merge \"Add support for configuring Dell SONiC switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93b2819d8fb67b56da2e5c080807864cb5055d38"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 15:12:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 15:15:41.000000000","tz":0},"subject":"Encourage agents to identify themselves","message":"Encourage agents to identify themselves\n\nCommits created in conjunction with generative AI should be marked as\nsuch to keep us compliant with the openinfra AI policy[1].\n\n[1] https://openinfra.org/legal/ai-policy\n\nChange-Id: Ic343f22b1f3ac8af9a245664a985849b6d38cfdc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e1d0380b1aa106973d30037407adb7db45958cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e1d0380b1aa106973d30037407adb7db45958cc"}]},"branch":"refs/heads/master"},"9623c58043457bb057ab9438883b79c79c459135":{"kind":"REWORK","_number":2,"created":"2026-09-23 09:37:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/69/1006569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/1006569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/2"}}},"commit":{"parents":[{"commit":"93b2819d8fb67b56da2e5c080807864cb5055d38","subject":"Merge \"Add support for configuring Dell SONiC switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93b2819d8fb67b56da2e5c080807864cb5055d38"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 15:12:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-23 09:37:41.000000000","tz":0},"subject":"Encourage agents to identify themselves","message":"Encourage agents to identify themselves\n\nCommits created in conjunction with generative AI should be marked as\nsuch to keep us compliant with the openinfra AI policy[1].\n\n[1] https://openinfra.org/legal/ai-policy\n\nChange-Id: Ic343f22b1f3ac8af9a245664a985849b6d38cfdc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9623c58043457bb057ab9438883b79c79c459135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9623c58043457bb057ab9438883b79c79c459135"}]},"branch":"refs/heads/master"},"6a19c474a569fbf0988553029adf49c03deb3286":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-23 09:50:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/1006569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/1006569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/1006569/3"}}},"commit":{"parents":[{"commit":"8c4d908b251fb86c270886f1e4d1813875b52d48","subject":"CI: Ignore .md files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c4d908b251fb86c270886f1e4d1813875b52d48"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-09-21 15:12:20.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-09-23 09:49:57.000000000","tz":120},"subject":"Encourage agents to identify themselves","message":"Encourage agents to identify themselves\n\nCommits created in conjunction with generative AI should be marked as\nsuch to keep us compliant with the openinfra AI policy[1].\n\n[1] https://openinfra.org/legal/ai-policy\n\nChange-Id: Ic343f22b1f3ac8af9a245664a985849b6d38cfdc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a19c474a569fbf0988553029adf49c03deb3286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a19c474a569fbf0988553029adf49c03deb3286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
