)]}'
{"id":"openstack%2Fkayobe~559706","triplet_id":"openstack%2Fkayobe~master~I2268a1ddf9a2870c713f32a40689e1686365aabd","project":"openstack/kayobe","branch":"master","topic":"story/2004759","hashtags":[],"change_id":"I2268a1ddf9a2870c713f32a40689e1686365aabd","subject":"Add seed VM provisioning CI job","status":"MERGED","created":"2018-04-09 11:11:51.000000000","updated":"2020-06-17 14:37:02.000000000","submitted":"2020-06-17 11:17:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"559706-1592392644203-25594a5d","meta_rev_id":"6f4b5a1dcac81c7548c36b9b1ce521dbbb92de8d","_number":559706,"virtual_id_number":559706,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 11:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-16 16:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-17 09:25:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-17 09:25:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 15:36:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-17 09:25:43.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-17 11:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80c63f4ba69a356e0172ac9aa7259ca30eb0ab76","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-09 11:11:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdac84c0a3ef2dd66e54bad51728f6f59b10df16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-09 11:25:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/1/check/kayobe-seed-centos/3f42fc3/ : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0c9196753fdc056671f932325c577f0ce6532b33","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-09 11:28:16.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ec28f237fbc64d6e9e3c279ea6fd9021550356d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 08:36:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d431d5cc776196814d8bd96fa7401fa5d75c5d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 08:47:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/da6a956/ : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9bb74cc3b580fa25779113da1f6af45465e862fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 09:05:40.000000000","message":"Patch Set 2:\n\nrecheck\n\nstackhpc.libvirt-vm role issue in KVM mode.","accounts_in_message":[],"_revision_number":2},{"id":"1b2d075beff6740197c45b0f86f66fc0467a0afe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 09:15:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/93a40d4/ : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":2},{"id":"76370abd50d0466da6e0fd34a85ee22e3a52e8d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 09:32:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5f8723a879ef09f5a4709e8013a9676a4224da50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 09:42:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/6feff2e/ : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c3758412c3924980affb9a183ed08ef4762be729","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 10:08:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e311f108477b772a80c4eaf3f83d060981be69ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 10:18:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/0168fe0/ : FAILURE in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"701c37ca32284d92d8ad148c68dea11851c4abe1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 10:41:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"58e6178c632253d2286df72ed7570e54bfb51e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 10:54:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/f9b68ac/ : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4e1b6dbd3010842773a5628aefc1a1a1e2e7e309","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 12:30:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7329e858c91236fe740acd75438ad84626a397d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 12:42:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/7e6514c/ : FAILURE in 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"16831a11f1ab8955f07398c7c8be7ed7bf5e89ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-27 12:57:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4ffd2134642b6a4084e018fad0ec1618ee5aa275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 13:13:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/ccf970f/ : FAILURE in 8m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7242a2a65f9571dc9dca43fac7de0ca4cefa5098","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-04-28 16:57:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64305a2b5294586a68e9e98fc222d6898b609508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-28 17:16:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/42b75c4/ : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":2},{"id":"63402f7aaee4352d7af5cd96775aea4992f204b8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 12:59:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"86d0bbfe58dc53ff7be50306cf9dc5f8cca3aa48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 13:14:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/0c3ef73/ : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"cf3c07cbca5dba3507578360c922efa667bd7493","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 13:16:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8c2abf8c7b8d283975333cbda2d9eef505b63cf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 13:23:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/06/559706/2/check/kayobe-seed-centos/dd08aea/ : FAILURE in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"879e782bcce4c0d78abd899f07bf527c80a58b2c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 17:50:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fbbff999ce3fd32e7fb5368d88971a3080f70c86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 18:03:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/3/check/kayobe-seed-vm-centos/6c38459/ : FAILURE in 11m 45s","accounts_in_message":[],"_revision_number":3},{"id":"1c4ee557c38e7943fad0b2fa8706e435c6ba88d6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 18:09:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"49b14b9ba1c3a8d15aa9509efb913ef8c7e2341e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 18:19:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/3/check/kayobe-seed-vm-centos/f4fa823/ : FAILURE in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b1ad99692ba78da7dff59230441b174c27f9144e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 18:30:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e741dbcf2cbee18779631c599cc06dba4773133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 18:36:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/4/check/kayobe-seed-vm-centos/e4d0b60/ : FAILURE in 4m 37s","accounts_in_message":[],"_revision_number":4},{"id":"e121f9b715e2f36873b54eaf427ab94d48746dfe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 18:42:12.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d5662ff751206e748a46cca089083506b21ebd3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 18:50:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/4/check/kayobe-seed-vm-centos/1bd9f24/ : FAILURE in 6m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f5914b73fa167a813360e5d82a60faeadbfad62c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 19:18:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7edcffa5f2653d712b0e38f707e26ab0f51b3899","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 19:26:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/5/check/kayobe-seed-vm-centos/cdf4d14/ : FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c4889d8f96d98d6279f34833c6f649dc06c7d8b0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 19:40:01.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4a8f2935b391329b36db0bac4f2547b6a429207b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 19:56:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/5/check/kayobe-seed-vm-centos/cc613f5/ : FAILURE in 15m 29s","accounts_in_message":[],"_revision_number":5},{"id":"232f2c7415e5d3c6e5f271ccce0635f09778ca8a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 20:56:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"20688deef14a97b3c9a34def257be50bb0c75e67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 21:15:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/5/check/kayobe-seed-vm-centos/303ba66/ : FAILURE in 14m 52s","accounts_in_message":[],"_revision_number":5},{"id":"7249d599e02db9ad75a9852a6e2ffdd56af7e3ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-02 10:11:24.000000000","message":"Abandoned\n\nNot currently working on this, can be reinstated if necessary.","accounts_in_message":[],"_revision_number":5},{"id":"db19b3cc934e55e7564f57cec1c0d8e478bb90ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-03 14:14:34.000000000","message":"Restored\n\nUsing to test a change to the libvirt-vm role.","accounts_in_message":[],"_revision_number":5},{"id":"e438a907d315c0e4f6252af1b54888c5d351d373","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 14:14:47.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"9e2edd507a987cebf978db5045000109a656fd09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-03 14:16:19.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a024d61fb50e9743d64c059b2dd9169d25e13415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 14:16:31.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"5537e4eb607e0423d4809fe9c6ba03693faea72d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-03 14:17:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4584779836d2084f2d23993ca5f71bba0d0950d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 15:56:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/7/check/kayobe-seed-vm-centos/41c7818/ : FAILURE in 22m 55s","accounts_in_message":[],"_revision_number":7},{"id":"b13dfd01b99f49d31ca5cdaea187978611d59007","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-03 15:59:21.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b5b7d42943edd784c58793deaf5eda6123579dc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 16:47:34.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/7/check/kayobe-seed-vm-centos/e0f3239/ : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":7},{"id":"01a32a15ffbdc7f7516b3cc2a1b1d360783d3fbb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-03 16:49:22.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"004a3f60bb6c6ecad4ebb66598f8beed5caaf660","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 17:22:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/7/check/kayobe-seed-vm-centos/cbe9e3f/ : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":7},{"id":"8d086e1ad3a4205c4de4538a4f381918c6f3bf52","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-03 17:27:33.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"06da0123a76522e3a53166b79e68ca0dd6dd4eab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 17:48:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/8/check/kayobe-seed-vm-centos/32709a1/ : FAILURE in 15m 34s","accounts_in_message":[],"_revision_number":8},{"id":"cb53620508cacdc634f29c479e1d995f38664883","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-08-31 16:29:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"149b27b719907e7937a537bf541b9e596fa17d77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 19:01:07.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/9/check/kayobe-seed-vm-centos/b0a31f1/ : RETRY_LIMIT in 2m 14s","accounts_in_message":[],"_revision_number":9},{"id":"f6932182fb9c9ee031161ab58ab16a097a50a44c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-09-02 08:08:16.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"cf57c49f282bf5b5e0fe216f328e10869e8ca8ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-02 08:27:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/10/check/kayobe-seed-vm-centos/4968550/ : FAILURE in 15m 08s","accounts_in_message":[],"_revision_number":10},{"id":"1981dc555d343bc599639add9476f913f17fb24e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-09-03 08:13:01.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"d235a9a7e4b29d0a9a30d674476755f7749e19d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 08:30:52.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/10/check/kayobe-seed-vm-centos/0ed064e/ : FAILURE in 16m 42s","accounts_in_message":[],"_revision_number":10},{"id":"e5b5586ac17fd6f767fde388d248604ee2afea9d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-09 16:13:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1e47257d306d56c660f1b5a8ba5fef6474af4605","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 16:32:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/11/check/build-openstack-sphinx-docs/b0a7fde/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/06/559706/11/check/openstack-tox-cover/ee34013/cover/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/11/check/openstack-tox-py35/0ce4637/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/11/check/openstack-tox-pep8/e0bf322/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/11/check/openstack-tox-py27/815b956/ : SUCCESS in 4m 46s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/11/check/kayobe-seed-vm-centos/1232e47/ : FAILURE in 14m 24s","accounts_in_message":[],"_revision_number":11},{"id":"56495731072aecb357ce501629636fad83c8c28b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-09 16:42:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"aadaee1948d972bc7de93e5ffce8471d4ae9ea31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 17:16:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/12/check/build-openstack-sphinx-docs/54bae27/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/06/559706/12/check/openstack-tox-cover/3caeaf2/cover/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/12/check/openstack-tox-py35/581c397/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/12/check/openstack-tox-pep8/77c0577/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/12/check/openstack-tox-py27/ce4032d/ : SUCCESS in 4m 22s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/12/check/kayobe-seed-vm-centos/7e342d7/ : FAILURE in 8m 47s","accounts_in_message":[],"_revision_number":12},{"id":"8769ac4763f26a95083b9ce7c9afdd4bc874901b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-09 17:34:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3f17650db89ff19f4eef6a9308b1aa5ce9fd2c13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 18:04:24.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/13/check/build-openstack-sphinx-docs/a8b81f5/html/ : SUCCESS in 6m 57s\n- openstack-tox-cover http://logs.openstack.org/06/559706/13/check/openstack-tox-cover/b6f8c2b/cover/ : SUCCESS in 8m 45s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/13/check/openstack-tox-py35/d35f569/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/13/check/openstack-tox-pep8/c7c56b5/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/13/check/openstack-tox-py27/156f11a/ : SUCCESS in 4m 53s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/13/check/kayobe-seed-vm-centos/21065fb/ : FAILURE in 19m 47s","accounts_in_message":[],"_revision_number":13},{"id":"2cc8029e1dcd251583c87e74d151559ca9a54b51","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-11 20:15:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a58b747b9c746d804643c3051db5e882c2f822cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-11 20:23:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/14/check/build-openstack-sphinx-docs/a563e55/html/ : SUCCESS in 4m 00s\n- openstack-tox-cover http://logs.openstack.org/06/559706/14/check/openstack-tox-cover/5f069c1/cover/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/14/check/openstack-tox-py35/c155696/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/14/check/openstack-tox-pep8/42f8547/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/14/check/openstack-tox-py27/a9dfb23/ : SUCCESS in 4m 42s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/14/check/kayobe-seed-vm-centos/3dfd247/ : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":14},{"id":"5c7016cad1a34f2aa7219ebd85a4687a57e2bc71","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-11 20:36:00.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"3b1838b096f7521bd2aa9bcb4ee0f31c8e42a5c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-11 20:43:28.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/14/check/build-openstack-sphinx-docs/c9f8266/html/ : SUCCESS in 5m 18s\n- openstack-tox-cover http://logs.openstack.org/06/559706/14/check/openstack-tox-cover/79d9338/cover/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/14/check/openstack-tox-py35/ce926ef/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/14/check/openstack-tox-pep8/521ade9/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/14/check/openstack-tox-py27/6c31126/ : SUCCESS in 5m 12s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/14/check/kayobe-seed-vm-centos/46c942b/ : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":14},{"id":"60da761d29efbeec1d4d6fefa445c29befc45294","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 09:11:50.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"f1f43c548199dcd02fb89ebeb91a393dd1ecc848","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 09:22:14.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/14/check/build-openstack-sphinx-docs/41341c3/html/ : SUCCESS in 4m 08s\n- openstack-tox-cover http://logs.openstack.org/06/559706/14/check/openstack-tox-cover/709dfc8/cover/ : SUCCESS in 7m 23s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/14/check/openstack-tox-py35/bb7ee7b/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/14/check/openstack-tox-pep8/eef2d54/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/14/check/openstack-tox-py27/d49f597/ : SUCCESS in 7m 32s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/14/check/kayobe-seed-vm-centos/4b3aeab/ : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":14},{"id":"e4236e00a2c178cbcbd4c559b075b678906eea6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 09:36:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"acddc68becc6de37ecacc2fec3eeb57289f8f015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 09:51:05.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/15/check/build-openstack-sphinx-docs/781279a/html/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/06/559706/15/check/openstack-tox-cover/18c5e84/cover/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/15/check/openstack-tox-py35/fe02901/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/15/check/openstack-tox-pep8/4042072/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/15/check/openstack-tox-py27/a894004/ : SUCCESS in 7m 01s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/15/check/kayobe-seed-vm-centos/7bd2b1f/ : FAILURE in 9m 20s","accounts_in_message":[],"_revision_number":15},{"id":"072c4d7f3231e45bd0948ad290cc384cd378d9ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 10:06:28.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"34c4ef06e8fb802568a34e100819fccb6fd94cc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 10:25:59.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/16/check/build-openstack-sphinx-docs/fcbc654/html/ : SUCCESS in 5m 01s\n- openstack-tox-cover http://logs.openstack.org/06/559706/16/check/openstack-tox-cover/e8d2163/cover/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/16/check/openstack-tox-py35/182f9c8/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/16/check/openstack-tox-pep8/5587991/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/16/check/openstack-tox-py27/737a9f7/ : SUCCESS in 4m 26s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/16/check/kayobe-seed-vm-centos/8a04ba4/ : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":16},{"id":"6a3032b9e81848e098d034d6ba2a0a54b88cf4b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 10:40:00.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"1229c4be0884ca050a66875f8a6f67ecef246c6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 10:57:07.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/16/check/build-openstack-sphinx-docs/e1d6c5f/html/ : SUCCESS in 6m 47s\n- openstack-tox-cover http://logs.openstack.org/06/559706/16/check/openstack-tox-cover/0081d65/cover/ : SUCCESS in 6m 08s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/16/check/openstack-tox-py35/c229b30/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/16/check/openstack-tox-pep8/1f390e7/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/16/check/openstack-tox-py27/9d8d8f3/ : SUCCESS in 7m 01s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/16/check/kayobe-seed-vm-centos/f51d861/ : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":16},{"id":"5a499352328f2c2d2dc2b8a568f0900955abf3d2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 11:09:23.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"24c03cd487e62ea7343622ba60703539a7487a19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 11:32:53.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/17/check/build-openstack-sphinx-docs/f88896c/html/ : SUCCESS in 4m 07s\n- openstack-tox-cover http://logs.openstack.org/06/559706/17/check/openstack-tox-cover/cee2fb8/cover/ : SUCCESS in 7m 04s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/17/check/openstack-tox-py35/10d2d28/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/17/check/openstack-tox-pep8/8da244e/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/17/check/openstack-tox-py27/e8814a9/ : SUCCESS in 8m 22s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/17/check/kayobe-seed-vm-centos/b7ea728/ : FAILURE in 18m 42s","accounts_in_message":[],"_revision_number":17},{"id":"ac0bc55da3a8388ef5c09b1ea4648f58bbfd5e61","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 11:36:35.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"529263aa4902b01682c6dfc5ef50362b6c28b5f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 11:55:35.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/18/check/build-openstack-sphinx-docs/a1df253/html/ : SUCCESS in 4m 41s\n- openstack-tox-cover http://logs.openstack.org/06/559706/18/check/openstack-tox-cover/3ae87fc/cover/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/18/check/openstack-tox-py35/41f5512/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/18/check/openstack-tox-pep8/42688c2/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/18/check/openstack-tox-py27/a929cf7/ : SUCCESS in 4m 57s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/18/check/kayobe-seed-vm-centos/6596b48/ : FAILURE in 17m 41s","accounts_in_message":[],"_revision_number":18},{"id":"d61a1f2c622729afa638ce83b1b140dc174cfc8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 12:02:37.000000000","message":"Patch Set 18:\n\nrecheck\n\nVM kernel trace (QEMU)\n\nhttp://logs.openstack.org/06/559706/18/check/kayobe-seed-vm-centos/6596b48/primary/system_logs/libvirt-consoles/seed-console.txt.gz","accounts_in_message":[],"_revision_number":18},{"id":"d514afdf78a592263cfddfa88112362c9f78aa1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 12:22:50.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/18/check/build-openstack-sphinx-docs/41c2cae/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/06/559706/18/check/openstack-tox-cover/1f32820/cover/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/18/check/openstack-tox-py35/10ae5cc/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/18/check/openstack-tox-pep8/7227bac/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/18/check/openstack-tox-py27/98b93be/ : SUCCESS in 4m 35s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/18/check/kayobe-seed-vm-centos/473a326/ : FAILURE in 18m 58s","accounts_in_message":[],"_revision_number":18},{"id":"d44e1f163e3e8cf234ea485ebd5baaa5f1ed97c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 12:27:23.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"0fb14e029dfd2b19b22c08caacb4b9aa818f9e8a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 12:28:19.000000000","message":"Patch Set 19:\n\nrecheck\n\npip config missing trusted-hosts\n\nThe repository located at mirror.regionone.limestone.openstack.org is not a trusted or secure host and is being ignored. If this repository is available via HTTPS it is recommended to use HTTPS instead, otherwise you may silence this warning and allow it anyways with \u0027--trusted-host mirror.regionone.limestone.openstack.org\u0027.\\n  Could not find a version that satisfies the requirement docker (from versions: )\\nNo matching distribution found for docker\\nYou are using pip version 9.0.1, however version 18.1 is available.\\nYou should consider upgrading via the \u0027pip install --upgrade pip\u0027 command","accounts_in_message":[],"_revision_number":19},{"id":"ce506dcfc6cef494b8d2da463989c27b6b5fd6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 12:54:41.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/19/check/build-openstack-sphinx-docs/a849007/html/ : SUCCESS in 4m 23s\n- openstack-tox-cover http://logs.openstack.org/06/559706/19/check/openstack-tox-cover/75a3178/cover/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/19/check/openstack-tox-py35/0f8bfbe/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/19/check/openstack-tox-pep8/5392fdf/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/19/check/openstack-tox-py27/f5c25a4/ : SUCCESS in 7m 21s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/19/check/kayobe-seed-vm-centos/5cfc5ec/ : FAILURE in 22m 42s","accounts_in_message":[],"_revision_number":19},{"id":"6257fae03afc6c4d388c5601484ff64d55e48f02","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 13:01:57.000000000","message":"Patch Set 19:\n\nGetting \u0027sudo: a password is required\u0027 when accessing seed. QEMU.\n\nhttp://logs.openstack.org/06/559706/19/check/kayobe-seed-vm-centos/5cfc5ec/primary/ansible/seed-deploy","accounts_in_message":[],"_revision_number":19},{"id":"25aaa6bd55d36fd69e5bf4036141cf2703651168","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 14:12:15.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"b1a8a155f535c115263e8a4b847fa34a17bf667f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 14:41:26.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/19/check/build-openstack-sphinx-docs/5d60b7b/html/ : SUCCESS in 4m 46s\n- openstack-tox-cover http://logs.openstack.org/06/559706/19/check/openstack-tox-cover/f4c817c/cover/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/19/check/openstack-tox-py35/2b344ca/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/19/check/openstack-tox-pep8/30da88d/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/19/check/openstack-tox-py27/470a716/ : SUCCESS in 6m 39s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/19/check/kayobe-seed-vm-centos/259eec1/ : FAILURE in 23m 33s","accounts_in_message":[],"_revision_number":19},{"id":"863fb9dfcb7f99c726024ab616f9026b8d674e3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 14:51:33.000000000","message":"Patch Set 19:\n\nfailed: sudo: a password is required QEMU","accounts_in_message":[],"_revision_number":19},{"id":"3378fb6da338f491f10f14642d66107e2c16eee7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-12 14:51:57.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"16335e1d93f2eecd54842db57abc109bf15560a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 15:23:06.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/19/check/build-openstack-sphinx-docs/52904ed/html/ : SUCCESS in 4m 10s\n- openstack-tox-cover http://logs.openstack.org/06/559706/19/check/openstack-tox-cover/1d1a886/cover/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/19/check/openstack-tox-py35/76b246d/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/19/check/openstack-tox-pep8/ebc0c5a/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/19/check/openstack-tox-py27/d82e84d/ : SUCCESS in 4m 47s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/19/check/kayobe-seed-vm-centos/e42dc20/ : FAILURE in 25m 34s","accounts_in_message":[],"_revision_number":19},{"id":"95234eb625f6b53206bb8531c0aa309d27532454","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-13 12:43:24.000000000","message":"Patch Set 19:\n\nFailed: Timeout (12s) waiting for privilege escalation prompt:","accounts_in_message":[],"_revision_number":19},{"id":"745bbe45b625ad251b34658c280ff181fea0a957","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-13 12:43:29.000000000","message":"Patch Set 19: Workflow-1","accounts_in_message":[],"_revision_number":19},{"id":"9f4a8341d850ff1e15666a954f75eceaab366ec9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 10:14:16.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"52adba6437fc1533416f8422d20f33334d77c14a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 10:14:27.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"db44e4c9b476b74f5feab95c9f112ca914ca7d5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 10:17:59.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"261cd9a2de353d6ffe63ae0c24d264069b5e710c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 10:30:57.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/20/check/build-openstack-sphinx-docs/9b9c5d0/html/ : SUCCESS in 4m 16s\n- openstack-tox-cover http://logs.openstack.org/06/559706/20/check/openstack-tox-cover/9d0f25d/cover/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/20/check/openstack-tox-py35/b6a4131/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/20/check/openstack-tox-pep8/9c8c4f1/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/20/check/openstack-tox-py27/616d85f/ : SUCCESS in 4m 35s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/20/check/kayobe-seed-vm-centos/8ef0e5e/ : FAILURE in 10m 00s","accounts_in_message":[],"_revision_number":20},{"id":"91a79ecee3cea9699d58e052c731cca305a20f7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 12:25:00.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"33d8b9fcb76cc74ac9fe1eb0463abd4e188914c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 12:53:54.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/21/check/build-openstack-sphinx-docs/c1e8504/html/ : SUCCESS in 4m 20s\n- openstack-tox-cover http://logs.openstack.org/06/559706/21/check/openstack-tox-cover/8ddd9ff/cover/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/21/check/openstack-tox-py35/00c1da5/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/21/check/openstack-tox-pep8/3ee6ef2/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/21/check/openstack-tox-py27/c805a59/ : SUCCESS in 4m 08s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/21/check/kayobe-seed-vm-centos/02b474e/ : FAILURE in 23m 56s","accounts_in_message":[],"_revision_number":21},{"id":"b6f9a4243712e6e321b1585e8b119e7f4291deba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 19:39:08.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"4f9b83826e30611a3bc1eecd0e0e34e4526d1af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 20:03:10.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/06/559706/22/check/build-openstack-sphinx-docs/9e445b4/html/ : SUCCESS in 4m 48s\n- openstack-tox-cover http://logs.openstack.org/06/559706/22/check/openstack-tox-cover/00e0205/cover/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/06/559706/22/check/openstack-tox-py35/9557962/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/06/559706/22/check/openstack-tox-pep8/0707f17/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/06/559706/22/check/openstack-tox-py27/f9e82eb/ : SUCCESS in 8m 01s\n- kayobe-seed-vm-centos http://logs.openstack.org/06/559706/22/check/kayobe-seed-vm-centos/f3dbbb7/ : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":22},{"id":"9d0ceb320f16623c3a7e85829c277d9ae53957bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-12 18:00:13.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"9ae2b3afa407947b3461270a40af2db534c4c63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:13:06.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05dc1de4a04b4ebc8ae014af3b8d631e : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/65830d2fd4bd4ce89670249667c3e32f : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17bae29a2170461d99ee4ae88521fd39 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517d262385a749d8bda19ba88863b070 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c9d2b1fc5244d7aa547d6e285e31f7c : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3498ecbda377408eb26bf9b8b14a8657 : SUCCESS in 6m 52s\n- kayobe-seed-vm-centos https://zuul.opendev.org/t/openstack/build/49596bd0bc884f1b90b837bff56e2c40 : RETRY_LIMIT in 5m 10s","accounts_in_message":[],"_revision_number":23},{"id":"88cd59698904d7818cfc6e7a32599691908cb4e7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 10:40:24.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"92c344aa1f6cea1cd9f3909bf6848706580be689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 10:48:08.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos https://zuul.opendev.org/t/openstack/build/3f36477f48fa4240ae7df80a68049eec : FAILURE in 6m 42s","accounts_in_message":[],"_revision_number":24},{"id":"9697f322779ebfb2c3ba8c31a42eb29d5ad4ff4f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 11:26:41.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"66bf32bd53ced8befb833421d9175d689541e712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 11:44:28.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos https://zuul.opendev.org/t/openstack/build/26423a0fbe3241578849fb5c3c14d97e : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":25},{"id":"216a0ceae567fbc7c52777718ca418521f301fbf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 12:27:18.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"6f4e766b25d20ecdb4e7ddd14a33dda7c27d64df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 12:34:42.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos https://zuul.opendev.org/t/openstack/build/610aea523bcd46428e51ae3f33cb3492 : FAILURE in 6m 54s","accounts_in_message":[],"_revision_number":25},{"id":"22284ec4a7efc4d21c175a0ff42d831cb6d91ef7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 12:37:43.000000000","message":"Patch Set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"5ceaf4f4fca59c3b6ef0faaa089a195d41f34f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:09:08.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos https://zuul.opendev.org/t/openstack/build/68ea53cbe2ee4e67af0562ebb0d80f86 : FAILURE in 30m 04s","accounts_in_message":[],"_revision_number":26},{"id":"e485543c23c6ccf5faeaf54db3ab5b7092ad6dfe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 14:03:55.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"46b1949df917baed2575f0a49425fec7cae6b69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 14:36:22.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos https://zuul.opendev.org/t/openstack/build/8813e984a088451b8ea31f7332e5cbb0 : FAILURE in 24m 17s","accounts_in_message":[],"_revision_number":27},{"id":"95838fb9ee1e57bab6f993af9b4b2c73ca5cd8b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:15:33.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"9d18904b2a5ffca2b9596d98dc3cf431d5c47cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 09:26:39.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4ac8fbce06624e348e80cf3646e9ff1c : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":28},{"id":"f4a62f08bccb2099354724440eae085bd45ee46e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:49:47.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"1b84cbc34df58b2833f485b4adf3ab61775e124b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 10:00:55.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/585a38c01a1a40e7963c1cf20a79e705 : FAILURE in 10m 23s","accounts_in_message":[],"_revision_number":28},{"id":"e8589d526da4d06c40492cc8a2f0d82ef3df36a8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 10:08:05.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"54ee1f214a9f9e7b925ab2255d5a2fe085142f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 10:31:01.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c49578fc4a5e46f7a0bffc5989dc3bfd : FAILURE in 18m 10s","accounts_in_message":[],"_revision_number":29},{"id":"2bee8a2563dc0957479408b51b51f3c00f0c0206","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 10:37:07.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"46af7a8dfd2e88a360d7ce71c3456ee200d0047d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:13:34.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/83f8bef7f77444f796569eb0b1be4b98 : TIMED_OUT in 1h 31m 45s","accounts_in_message":[],"_revision_number":30},{"id":"f14edd840a3447a8d398ff2573cef087f3054433","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 13:36:50.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"4320d3743c69de8a56553152762ddb001269976e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:15:01.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9132213b2e78462aab2711e92477e389 : RETRY_LIMIT in 4m 47s","accounts_in_message":[],"_revision_number":31},{"id":"3ea5160b82fc38697f5841a1e6b20ffede435b07","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 14:29:47.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"8a3d6c03ffe301bbfc8e629add98018db5a8c488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:51:05.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1f44a6b014a546ecb9457ce4e144fe7c : FAILURE in 20m 42s","accounts_in_message":[],"_revision_number":32},{"id":"37f289d033f012c29c06e239820025f045902341","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 14:55:26.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"ea69981041ddc576b1c4d68937fcb717aac447eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:29:00.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/135dd40725904cb6a2d20a7a4e95489d : TIMED_OUT in 1h 32m 04s","accounts_in_message":[],"_revision_number":33},{"id":"a314f458625896dfc996ea53628f26815495a1be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 17:08:28.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"d069f3e558e8154eef24714dbf370fcbd227122b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:32:13.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/453092c6540e48e6b1a790a4d4e3fbb6 : FAILURE in 1h 17m 43s","accounts_in_message":[],"_revision_number":34},{"id":"5cd6eda579355e63c279e42624abcbf4832f013f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 08:26:28.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"5ebde7573549294588e75cfef6ad7c25b99c568e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 08:30:51.000000000","message":"Patch Set 35: Workflow-1\n\nWaiting on https://github.com/jriguera/ansible-role-configdrive/pull/8.","accounts_in_message":[],"_revision_number":35},{"id":"2fd4e298a7f27aaa7cd0e014a6e0f0496c4f4180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:18:05.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eee69db1666b46c8b3044eb21daa90e6 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c389f1df17e4a26885d390e0b5c79a4 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/875b82d9883f432f9af6c36d12cf7f6a : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19a6ede9c9d940b89b9bb1714cad128e : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be8243e6e07b43c09e85e7eaea3109e9 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d31e56db4e4b7ca8aa326fb388a738 : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d6590218524444ea55cc94eea619384 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7887fb5fe0746aaba4ddc9e009da02e : SUCCESS in 7m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5dbf2272a27045f3989076c6d9112953 : SUCCESS in 37m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d46ed4743665415dac9791d1f81f4b04 : RETRY_LIMIT in 4m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/565d1babf61841f3a5fc974fe5756421 : RETRY_LIMIT in 5m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/800c1ed1748443ada13332c9523f06d6 : FAILURE in 1h 47m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f4ae8191f1674e8787d7ff184d003f9a : FAILURE in 35m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0f73b4ca83f54616ad26cefa14e8765e : SUCCESS in 28m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f2e68eeab67d40209e89c84056f5cfb5 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":35},{"id":"9b6d4a59ed653ef8f03e55afc14c2c3ba255c858","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 10:35:09.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"7a78992c059bc09ad2f071c194d9f12a7ae498af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 12:57:37.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60a04b18507344f2852b9168ed7ed508 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f3a298b3ba4e4c85a4ee16d747a5a2 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5124b595182b402c8967781269a61c82 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d8bec69c7154ae3bd11e6d4d6a2dd09 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1640be4426eb4e9c899a6478aceb67fa : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11893a9eaa6e45df9c57bcdcf41c7731 : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9650dd7536ce456d9d0775b989426a41 : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0e2f079c2994c8d9824725f3536f0cc : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/004066db62234580abcfecdc2fa380ed : SUCCESS in 37m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9610f552e84740e486c5ab4b2d3df802 : SUCCESS in 1h 15m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8756aa2681e74a84bd7b2e77c9071b9b : SUCCESS in 18m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f6a18ee0282c46f1802b84f9865f14cb : SUCCESS in 2h 12m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/123a434a8c02462c83c32b22c9bbbbbb : SUCCESS in 19m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/83dc604adf5d448ea48467158205d2b7 : SUCCESS in 27m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/02ba3eea50f24977a72cebae703f80dd : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":36},{"id":"3c45f57e148b58e59b9f93c9e37be225192070bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 12:59:40.000000000","message":"Patch Set 36: Workflow-1","accounts_in_message":[],"_revision_number":36},{"id":"c19c4f48c9686c882bcfe64104254f1eff721d83","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-10 08:01:15.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"7a8ec7241b845ee451c37506933085765448a079","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-10 08:01:46.000000000","message":"Patch Set 37:\n\n2 years, 2 months, 1 day later, it\u0027s ready to go.","accounts_in_message":[],"_revision_number":37},{"id":"c23469f873e91dc7800e1f4f0147b1d56a0f3587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:05:29.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0050545e174c4ef0ad287559b9f54ddb : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001b8d375db141a08eaa958d700272a1 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0af78c2e87a84e40842189ca4b0bc642 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e57e860df5db42aabb27729bb44de571 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d6bdd69a38f4532b12df367230cc96e : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1856007716ef4329916384f3181d059d : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb96a7d3daee474a9f365c9654dd16d4 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/492a49c088174c07bf927b005f474632 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4adf808a24d64644b62675422d80e936 : SUCCESS in 34m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2c79d13f0c6a46fc806c7306937c2fb1 : SUCCESS in 1h 26m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a0f8ae19c316483bac6e55abf032a60a : SUCCESS in 16m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d379b76b9a634abd94e522fc338f64cc : SUCCESS in 1h 57m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d952f6ae9ca249adbe0e6672a4bc10f1 : SUCCESS in 18m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b4198f4281134108b8f3671685be1f48 : SUCCESS in 28m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c3e3924d032941208d4f5f58bb3c4fd0 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":37},{"id":"5df627936321c58e5f4f038672b2d996e658814d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-10 15:36:51.000000000","message":"Patch Set 37: Code-Review+2","accounts_in_message":[],"_revision_number":37},{"id":"e7e22ad062c6a9c8b99aaa476de1c40e5b98ea87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 16:19:56.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"f97caa55edfe046a8cfc8f35866e724bfbfaf51b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:51:43.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/970425ea1e6348a584458bb2195ed39b : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dcac21024f8416bb13dbf9e633e01fe : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5475fc49eb3843369d94de8a79c94f02 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f077cebc602b4a9ea82db48e2b322bdb : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d9fbcc978e45b8ba0448aeda3414ad : SUCCESS in 9m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18f872f8d44042f6906e3916c7801c99 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e794f2318cf54dda8c451524c8e6bfe7 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e4240dc94f046a28899f867d23ee657 : SUCCESS in 36m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/084bc37fe3864d769f44c7a0a86cb84d : SUCCESS in 1h 10m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b0918686d579440a9bd3f84b2558b1f3 : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/715883ff05f543b793aecc55a3485a66 : SUCCESS in 1h 51m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9bf6b9508dc54038ac70f8976d00babe : SUCCESS in 19m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/448f3006979047d980e6a23956529e72 : SUCCESS in 26m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/91c9ab5fb59243038c603db12f84e6b9 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":38},{"id":"e682cdf4a967bddb807f91c30923e276fa9c004e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 09:25:43.000000000","message":"Patch Set 38: Code-Review+2 Workflow+1\n\nLooks good, I see this pass:\n\nTASK [Wait for SSH access to the seed VM] **************************************","accounts_in_message":[],"_revision_number":38},{"id":"a71ff048cee260df511e9de4139a7c628c9af440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:26:04.000000000","message":"Patch Set 38: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":38},{"id":"3aa53262bc467356a469e5f1f2e2d2bcdd023088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:17:23.000000000","message":"Patch Set 38: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68ec131d505466f9a2d20402c40787b : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd7ba8b6aa4c4c1e83c8019887043232 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbd6fbc73bce43f1b5a1ed703eaadda8 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188033ecee28443fbf98d31b027f5d4d : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e979fc1a141946d1a7c800f402a0d6ec : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0d98c51af1e427b942b65e70ac77b28 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e162623f8eee4e07b43c84e83b500e1e : SUCCESS in 36m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1f116592c7844576b5b9d985f7dc747d : SUCCESS in 1h 36m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f58e9fd88e024fcd941fe6fd7b5cbb97 : SUCCESS in 19m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9f9b4aa579ef4b16b564110c9073908c : SUCCESS in 1h 49m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/eefae291cb5d4849ab27f11ac24023a2 : SUCCESS in 17m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/599865b680c342a1868a12ba26b33ae3 : SUCCESS in 23m 13s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9c0e3bfd85444ff3bf44a3beafb00eb5 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":38},{"id":"12e5e1927e59ddacff29323c19bcd30fe496734c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:17:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":38},{"id":"8228e19e8aebd787aafd9ce0bdda70b63d258435","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:20:44.000000000","message":"Patch Set 38:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d81865f77b748edaf768832be45ca0d : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":38},{"id":"dea03da70d976fe7e8ea3bae841f7526f1a4e64e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 14:37:02.000000000","message":"Patch Set 38: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0d064bb77af828d4a5eeb1052b093bc9f9b9f2a5","accounts_in_message":[],"_revision_number":38}],"current_revision_number":38,"current_revision":"c16597aa2daf76d7dd31991c5db938eee291a098","revisions":{"6b724ec2769ad7fcb0250b354791d07ba27861c9":{"kind":"REWORK","_number":1,"created":"2018-04-09 11:11:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/1"}}},"commit":{"parents":[{"commit":"cd6c366cc63cf6b34b9a9427bd339e6b949938ef","subject":"Add a job for seed deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd6c366cc63cf6b34b9a9427bd339e6b949938ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-09 11:11:09.000000000","tz":60},"subject":"WIP: Use the Zuul VM as a seed hypervisor","message":"WIP: Use the Zuul VM as a seed hypervisor\n\nModifies the kayobe-seed-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b724ec2769ad7fcb0250b354791d07ba27861c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b724ec2769ad7fcb0250b354791d07ba27861c9"}]},"branch":"refs/heads/master"},"f09b5d18737238209eebb0e9d2d5aaebc7e6a3d3":{"kind":"REWORK","_number":2,"created":"2018-04-27 08:36:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/2"}}},"commit":{"parents":[{"commit":"1c976bc125784fa41503ebf2cd7aea5fe4a17170","subject":"Update release notes to mention TLS certicate bundle handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c976bc125784fa41503ebf2cd7aea5fe4a17170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-27 08:36:43.000000000","tz":60},"subject":"WIP: Use the Zuul VM as a seed hypervisor","message":"WIP: Use the Zuul VM as a seed hypervisor\n\nModifies the kayobe-seed-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09b5d18737238209eebb0e9d2d5aaebc7e6a3d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09b5d18737238209eebb0e9d2d5aaebc7e6a3d3"}]},"branch":"refs/heads/master"},"c926249ce3ba4a5c0675997cbbdf2f330e4e966b":{"kind":"REWORK","_number":3,"created":"2018-05-03 17:50:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/3"}}},"commit":{"parents":[{"commit":"1c976bc125784fa41503ebf2cd7aea5fe4a17170","subject":"Update release notes to mention TLS certicate bundle handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c976bc125784fa41503ebf2cd7aea5fe4a17170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-05-03 17:50:16.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c926249ce3ba4a5c0675997cbbdf2f330e4e966b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c926249ce3ba4a5c0675997cbbdf2f330e4e966b"}]},"branch":"refs/heads/master"},"d996030c5baa6a2783509989f7961ad68ff87cc9":{"kind":"REWORK","_number":4,"created":"2018-05-03 18:30:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/4"}}},"commit":{"parents":[{"commit":"1c976bc125784fa41503ebf2cd7aea5fe4a17170","subject":"Update release notes to mention TLS certicate bundle handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c976bc125784fa41503ebf2cd7aea5fe4a17170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-05-03 18:30:10.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d996030c5baa6a2783509989f7961ad68ff87cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d996030c5baa6a2783509989f7961ad68ff87cc9"}]},"branch":"refs/heads/master"},"5efac2c7c5241ca1cfdde270e39b4dac3445cc00":{"kind":"REWORK","_number":5,"created":"2018-05-03 19:18:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/5"}}},"commit":{"parents":[{"commit":"1c976bc125784fa41503ebf2cd7aea5fe4a17170","subject":"Update release notes to mention TLS certicate bundle handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c976bc125784fa41503ebf2cd7aea5fe4a17170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-05-03 19:18:38.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5efac2c7c5241ca1cfdde270e39b4dac3445cc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5efac2c7c5241ca1cfdde270e39b4dac3445cc00"}]},"branch":"refs/heads/master"},"9dfed93245dffb0248b82238246e6984b7f17224":{"kind":"REWORK","_number":6,"created":"2018-08-03 14:16:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/6"}}},"commit":{"parents":[{"commit":"1c976bc125784fa41503ebf2cd7aea5fe4a17170","subject":"Update release notes to mention TLS certicate bundle handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c976bc125784fa41503ebf2cd7aea5fe4a17170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-08-03 14:16:16.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9dfed93245dffb0248b82238246e6984b7f17224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9dfed93245dffb0248b82238246e6984b7f17224"}]},"branch":"refs/heads/master"},"28850bbb4d59a98e7522c1fca5b9881a19c82fa4":{"kind":"REWORK","_number":7,"created":"2018-08-03 14:17:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/7"}}},"commit":{"parents":[{"commit":"34e410dd465cc4201f8b98ab3bf376b0cc7182a8","subject":"Merge \"Fix path to virtualenv in usage guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34e410dd465cc4201f8b98ab3bf376b0cc7182a8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-08-03 14:17:27.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28850bbb4d59a98e7522c1fca5b9881a19c82fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28850bbb4d59a98e7522c1fca5b9881a19c82fa4"}]},"branch":"refs/heads/master"},"002051aa2305bf116ed3a25d4c7718d9eca67a64":{"kind":"REWORK","_number":8,"created":"2018-08-03 17:27:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/8"}}},"commit":{"parents":[{"commit":"34e410dd465cc4201f8b98ab3bf376b0cc7182a8","subject":"Merge \"Fix path to virtualenv in usage guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34e410dd465cc4201f8b98ab3bf376b0cc7182a8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-08-03 17:27:22.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/002051aa2305bf116ed3a25d4c7718d9eca67a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/002051aa2305bf116ed3a25d4c7718d9eca67a64"}]},"branch":"refs/heads/master"},"12a4cbad0cb9db28616ec3a06286b49dab4a908d":{"kind":"REWORK","_number":9,"created":"2018-08-31 16:29:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/9"}}},"commit":{"parents":[{"commit":"679de79cfc9374cf28661e019b2f2e6d94cb1714","subject":"Switch inspection store to use stable docker image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/679de79cfc9374cf28661e019b2f2e6d94cb1714"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-08-31 16:29:29.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12a4cbad0cb9db28616ec3a06286b49dab4a908d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12a4cbad0cb9db28616ec3a06286b49dab4a908d"}]},"branch":"refs/heads/master"},"afa336057a7a6674401175164afb24ad35ebcab3":{"kind":"REWORK","_number":10,"created":"2018-09-02 08:08:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/10"}}},"commit":{"parents":[{"commit":"679de79cfc9374cf28661e019b2f2e6d94cb1714","subject":"Switch inspection store to use stable docker image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/679de79cfc9374cf28661e019b2f2e6d94cb1714"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-09-02 08:03:35.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afa336057a7a6674401175164afb24ad35ebcab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afa336057a7a6674401175164afb24ad35ebcab3"}]},"branch":"refs/heads/master"},"78a99be3215fd145d1475209b6d71a61b2c988ee":{"kind":"REWORK","_number":11,"created":"2018-11-09 16:13:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/11"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-09 16:13:03.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78a99be3215fd145d1475209b6d71a61b2c988ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78a99be3215fd145d1475209b6d71a61b2c988ee"}]},"branch":"refs/heads/master"},"a0f286257600268a7fd5728cd8bfefc046c2c031":{"kind":"REWORK","_number":12,"created":"2018-11-09 16:42:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/12"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-09 16:42:42.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0f286257600268a7fd5728cd8bfefc046c2c031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0f286257600268a7fd5728cd8bfefc046c2c031"}]},"branch":"refs/heads/master"},"9f3b004b7e95ac34f87e6cd99ff9adfebc4c0e03":{"kind":"REWORK","_number":13,"created":"2018-11-09 17:34:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/13"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-09 17:34:28.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f3b004b7e95ac34f87e6cd99ff9adfebc4c0e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f3b004b7e95ac34f87e6cd99ff9adfebc4c0e03"}]},"branch":"refs/heads/master"},"1b453f9ab2e5ca399c185c5f7afc1bbe5211bdac":{"kind":"REWORK","_number":14,"created":"2018-11-11 20:15:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/14"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-11 20:14:42.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b453f9ab2e5ca399c185c5f7afc1bbe5211bdac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b453f9ab2e5ca399c185c5f7afc1bbe5211bdac"}]},"branch":"refs/heads/master"},"096d6bd74978317fdafd263ec3825bb061789e15":{"kind":"REWORK","_number":15,"created":"2018-11-12 09:36:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/15"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-12 09:36:26.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/096d6bd74978317fdafd263ec3825bb061789e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/096d6bd74978317fdafd263ec3825bb061789e15"}]},"branch":"refs/heads/master"},"168b896d8e128647acbd0d87749214a7bfb25d54":{"kind":"REWORK","_number":16,"created":"2018-11-12 10:06:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/16"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-12 10:06:17.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/168b896d8e128647acbd0d87749214a7bfb25d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/168b896d8e128647acbd0d87749214a7bfb25d54"}]},"branch":"refs/heads/master"},"70c5759a1c8a1762e22d1ea5f96dd95a30479d47":{"kind":"REWORK","_number":17,"created":"2018-11-12 11:09:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/17"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-12 11:09:15.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70c5759a1c8a1762e22d1ea5f96dd95a30479d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70c5759a1c8a1762e22d1ea5f96dd95a30479d47"}]},"branch":"refs/heads/master"},"4a2b8418c6952da621fdbef2d5aa141d06e73bdb":{"kind":"REWORK","_number":18,"created":"2018-11-12 11:36:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/18"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-12 11:36:26.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a2b8418c6952da621fdbef2d5aa141d06e73bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a2b8418c6952da621fdbef2d5aa141d06e73bdb"}]},"branch":"refs/heads/master"},"bcddeeacae95a1a3789c5bca79e4ba71865cc95b":{"kind":"REWORK","_number":19,"created":"2018-11-12 12:27:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/19"}}},"commit":{"parents":[{"commit":"ac79d0d09a58251c4f235907ca5d744c7c1a8a4d","subject":"Add note about LLDP workaround for switches on Dell OS 9.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac79d0d09a58251c4f235907ca5d744c7c1a8a4d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-12 12:27:13.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bcddeeacae95a1a3789c5bca79e4ba71865cc95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bcddeeacae95a1a3789c5bca79e4ba71865cc95b"}]},"branch":"refs/heads/master"},"a6971af0cf3536a482fa4efaa937f5696bac29bb":{"kind":"REWORK","_number":20,"created":"2019-01-11 10:14:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/20"}}},"commit":{"parents":[{"commit":"bf3d31aaa273af38693ea3aba96f6e56117b1c5e","subject":"Merge \"Perform ironic online data migrations prior to seed upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf3d31aaa273af38693ea3aba96f6e56117b1c5e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-11 10:14:00.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6971af0cf3536a482fa4efaa937f5696bac29bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6971af0cf3536a482fa4efaa937f5696bac29bb"}]},"branch":"refs/heads/master"},"600726081a4bfdd5cfa6aba51d2c08cd7d6b25ee":{"kind":"REWORK","_number":21,"created":"2019-01-11 12:25:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/21"}}},"commit":{"parents":[{"commit":"bf3d31aaa273af38693ea3aba96f6e56117b1c5e","subject":"Merge \"Perform ironic online data migrations prior to seed upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf3d31aaa273af38693ea3aba96f6e56117b1c5e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-11 12:24:44.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/600726081a4bfdd5cfa6aba51d2c08cd7d6b25ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/600726081a4bfdd5cfa6aba51d2c08cd7d6b25ee"}]},"branch":"refs/heads/master"},"370c25b92179d1c15023e589f2c023aa54dabaa5":{"kind":"REWORK","_number":22,"created":"2019-01-11 19:39:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/22"}}},"commit":{"parents":[{"commit":"ba1f85c22fa839df546633edd67b83c61627abf8","subject":"Update stackhpc.libvirt-vm and stackhpc.libvirt-host roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba1f85c22fa839df546633edd67b83c61627abf8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-11 19:38:57.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/370c25b92179d1c15023e589f2c023aa54dabaa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/370c25b92179d1c15023e589f2c023aa54dabaa5"}]},"branch":"refs/heads/master"},"09191ecee0f94da8089b4720a357eb2b20087db4":{"kind":"REWORK","_number":23,"created":"2019-12-12 18:00:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/23"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-12 18:00:04.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09191ecee0f94da8089b4720a357eb2b20087db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09191ecee0f94da8089b4720a357eb2b20087db4"}]},"branch":"refs/heads/master"},"e1bd4771057ea47a0d20ae8448c5638aa81257c2":{"kind":"REWORK","_number":24,"created":"2019-12-13 10:40:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/24"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 10:39:54.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1bd4771057ea47a0d20ae8448c5638aa81257c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1bd4771057ea47a0d20ae8448c5638aa81257c2"}]},"branch":"refs/heads/master"},"2ab08a5c14f60d602dd04579ba2b2178a17506b0":{"kind":"REWORK","_number":25,"created":"2019-12-13 11:26:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/25"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 11:26:33.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.openstack.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ab08a5c14f60d602dd04579ba2b2178a17506b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ab08a5c14f60d602dd04579ba2b2178a17506b0"}]},"branch":"refs/heads/master"},"c21868e988fa0e4175673fab8c52f1ba71a5443a":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2019-12-13 12:37:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/26"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 12:37:39.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.opendev.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c21868e988fa0e4175673fab8c52f1ba71a5443a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c21868e988fa0e4175673fab8c52f1ba71a5443a"}]},"branch":"refs/heads/master"},"28efe94bab11e00ac8df19820146d2db588ef669":{"kind":"REWORK","_number":27,"created":"2019-12-13 14:03:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/27"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 14:03:51.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nDepends-On: https://review.opendev.org/617161\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28efe94bab11e00ac8df19820146d2db588ef669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28efe94bab11e00ac8df19820146d2db588ef669"}]},"branch":"refs/heads/master"},"7f0ed5ca78b614cf64bb8751402e69c4667a4af6":{"kind":"REWORK","_number":28,"created":"2020-06-08 09:15:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/28"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 09:14:53.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f0ed5ca78b614cf64bb8751402e69c4667a4af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f0ed5ca78b614cf64bb8751402e69c4667a4af6"}]},"branch":"refs/heads/master"},"62a4aa445e382d05ae767c50c29848426d4fb050":{"kind":"REWORK","_number":29,"created":"2020-06-08 10:08:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/29"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 10:08:00.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62a4aa445e382d05ae767c50c29848426d4fb050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62a4aa445e382d05ae767c50c29848426d4fb050"}]},"branch":"refs/heads/master"},"8676138cf525c92f41a5b396db2c4f50e5ca9c99":{"kind":"REWORK","_number":30,"created":"2020-06-08 10:37:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/30"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 10:37:03.000000000","tz":0},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8676138cf525c92f41a5b396db2c4f50e5ca9c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8676138cf525c92f41a5b396db2c4f50e5ca9c99"}]},"branch":"refs/heads/master"},"4aa372e2de01cc0b12d6aeec33aa43a8e90eb354":{"kind":"REWORK","_number":31,"created":"2020-06-08 13:36:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/31"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 13:32:38.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4aa372e2de01cc0b12d6aeec33aa43a8e90eb354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4aa372e2de01cc0b12d6aeec33aa43a8e90eb354"}]},"branch":"refs/heads/master"},"3337fd30b64034f99d799858b802085dc4b2399d":{"kind":"REWORK","_number":32,"created":"2020-06-08 14:29:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/32"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 14:29:27.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3337fd30b64034f99d799858b802085dc4b2399d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3337fd30b64034f99d799858b802085dc4b2399d"}]},"branch":"refs/heads/master"},"80a76513e89feb470cfd4aa94171e50ca8631740":{"kind":"REWORK","_number":33,"created":"2020-06-08 14:55:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/33"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 14:54:56.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80a76513e89feb470cfd4aa94171e50ca8631740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80a76513e89feb470cfd4aa94171e50ca8631740"}]},"branch":"refs/heads/master"},"ef527be439be0484baf656fe710b6dd5d7f9c86d":{"kind":"REWORK","_number":34,"created":"2020-06-08 17:08:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/34"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 17:07:41.000000000","tz":60},"subject":"WIP: Add seed VM provisioning CI job","message":"WIP: Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nDepends-On: https://review.opendev.org/617161\nDepends-On: https://review.opendev.org/734040\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef527be439be0484baf656fe710b6dd5d7f9c86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef527be439be0484baf656fe710b6dd5d7f9c86d"}]},"branch":"refs/heads/master"},"d08295c5e886bf16f31420ee4feb4185264d5ee6":{"kind":"REWORK","_number":35,"created":"2020-06-09 08:26:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/35"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-09 08:22:50.000000000","tz":60},"subject":"Add seed VM provisioning CI job","message":"Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nThe job uses a Cirros image for the seed VM rather than the usual CentOS\ncloud image. This is to reduce bandwidth required to download the image.\nIt does mean that the resulting seed VM cannot be used as a seed, but\nnested virt would make this slow and unreliable anyway. Cirros does not\nload cdrom drivers by default, so we add the configdrive as a disk\nrather than a cdrom device.\n\nDepends-On: https://review.opendev.org/617161\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d08295c5e886bf16f31420ee4feb4185264d5ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d08295c5e886bf16f31420ee4feb4185264d5ee6"}]},"branch":"refs/heads/master"},"12e8e6c3bb5cf2bb58ee1e9db784c735c56f7506":{"kind":"REWORK","_number":36,"created":"2020-06-09 10:35:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/36"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-09 10:34:58.000000000","tz":60},"subject":"Add seed VM provisioning CI job","message":"Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nThe job uses a Cirros image for the seed VM rather than the usual CentOS\ncloud image. This is to reduce bandwidth required to download the image.\nIt does mean that the resulting seed VM cannot be used as a seed, but\nnested virt would make this slow and unreliable anyway. Cirros does not\nload cdrom drivers by default, so we add the configdrive as a disk\nrather than a cdrom device.\n\nDepends-On: https://review.opendev.org/617161\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12e8e6c3bb5cf2bb58ee1e9db784c735c56f7506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12e8e6c3bb5cf2bb58ee1e9db784c735c56f7506"}]},"branch":"refs/heads/master"},"36d9485c6488226038adaca9aaf27e099372320a":{"kind":"REWORK","_number":37,"created":"2020-06-10 08:01:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/37"}}},"commit":{"parents":[{"commit":"b876d2e67e859d208234f132626e491303f5179b","subject":"Fix seed VM interface ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-10 08:01:11.000000000","tz":0},"subject":"Add seed VM provisioning CI job","message":"Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nThe job uses a Cirros image for the seed VM rather than the usual CentOS\ncloud image. This is to reduce bandwidth required to download the image.\nIt does mean that the resulting seed VM cannot be used as a seed, but\nnested virt would make this slow and unreliable anyway. Cirros does not\nload cdrom drivers by default, so we add the configdrive as a disk\nrather than a cdrom device.\n\nDepends-On: https://review.opendev.org/617161\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36d9485c6488226038adaca9aaf27e099372320a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36d9485c6488226038adaca9aaf27e099372320a"}]},"branch":"refs/heads/master"},"c16597aa2daf76d7dd31991c5db938eee291a098":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2020-06-16 16:19:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/559706/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/559706/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/559706/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/559706/38"}}},"commit":{"parents":[{"commit":"f7d85a5f2f8ab89426faf4a9f23704bcfd5bd54a","subject":"Merge \"Remove python 2 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7d85a5f2f8ab89426faf4a9f23704bcfd5bd54a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 16:19:47.000000000","tz":60},"subject":"Add seed VM provisioning CI job","message":"Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nThe job uses a Cirros image for the seed VM rather than the usual CentOS\ncloud image. This is to reduce bandwidth required to download the image.\nIt does mean that the resulting seed VM cannot be used as a seed, but\nnested virt would make this slow and unreliable anyway. Cirros does not\nload cdrom drivers by default, so we add the configdrive as a disk\nrather than a cdrom device.\n\nDepends-On: https://review.opendev.org/617161\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c16597aa2daf76d7dd31991c5db938eee291a098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c16597aa2daf76d7dd31991c5db938eee291a098"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
