)]}'
{"id":"openstack%2Fkayobe~566101","triplet_id":"openstack%2Fkayobe~master~Ie4814983613a40fb6ff14aa63ddfd59f4ed62daf","project":"openstack/kayobe","branch":"master","topic":"disable_live_restore","hashtags":[],"change_id":"Ie4814983613a40fb6ff14aa63ddfd59f4ed62daf","subject":"Disable live-restore by default","status":"MERGED","created":"2018-05-03 15:50:39.000000000","updated":"2018-05-04 14:07:18.000000000","submitted":"2018-05-04 14:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"566101-1525442838178-4c9a8829","meta_rev_id":"f7cd0d7a7ab6b4065e5d5afb348904199867fa05","_number":566101,"virtual_id_number":566101,"owner":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2018-05-04 14:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2018-05-04 13:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2018-05-04 12:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2018-05-04 13:27:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-04 12:35:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2018-05-04 13:27:07.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2018-05-04 14:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f62fab753e20c395c2595b85fd52948f2d4e198c","author":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"date":"2018-05-03 15:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed6b64efb832be828bea67f34adf0b06aebe8507","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-03 15:53:29.000000000","message":"Patch Set 1: Code-Review-1\n\nThe default value in the release needs updating.","accounts_in_message":[],"_revision_number":1},{"id":"2bb8ca43e0d02cc0d3ef5d7bf4615205b05513b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 16:30:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/566101/1/check/openstack-tox-pep8/8ed5aa7/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/01/566101/1/check/openstack-tox-py27/aefc338/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/01/566101/1/check/openstack-tox-py35/52d7e9c/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/01/566101/1/check/openstack-tox-cover/9cae7ac/cover/ : SUCCESS in 3m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/01/566101/1/check/build-openstack-sphinx-docs/75e74c6/html/ : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/01/566101/1/check/kayobe-tox-ansible-syntax/51a4bc2/ : SUCCESS in 6m 10s\n- kayobe-tox-ansible http://logs.openstack.org/01/566101/1/check/kayobe-tox-ansible/a380da9/ : SUCCESS in 5m 45s\n- kayobe-tox-molecule http://logs.openstack.org/01/566101/1/check/kayobe-tox-molecule/0bf4f18/ : SUCCESS in 19m 53s\n- kayobe-overcloud-centos http://logs.openstack.org/01/566101/1/check/kayobe-overcloud-centos/2383476/ : SUCCESS in 36m 40s\n- kayobe-seed-centos http://logs.openstack.org/01/566101/1/check/kayobe-seed-centos/8a2d770/ : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"398a83978e37246df0995109ab3c2e18311141e1","author":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"date":"2018-05-04 12:27:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94ccba8f09dcf4665f2848113714fdf1e74d31a1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-05-04 12:35:39.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Kevin","accounts_in_message":[],"_revision_number":2},{"id":"dd7638e2d3b1ddf1100063ce6097ae86f2ef27b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-04 13:10:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/566101/2/check/openstack-tox-pep8/640ca18/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/01/566101/2/check/openstack-tox-py27/a382ea3/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/01/566101/2/check/openstack-tox-py35/0b45c71/ : SUCCESS in 4m 19s\n- openstack-tox-cover http://logs.openstack.org/01/566101/2/check/openstack-tox-cover/e971952/cover/ : SUCCESS in 4m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/01/566101/2/check/build-openstack-sphinx-docs/ade9f6b/html/ : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/01/566101/2/check/kayobe-tox-ansible-syntax/a113e7e/ : SUCCESS in 6m 03s\n- kayobe-tox-ansible http://logs.openstack.org/01/566101/2/check/kayobe-tox-ansible/789c41b/ : SUCCESS in 7m 51s\n- kayobe-tox-molecule http://logs.openstack.org/01/566101/2/check/kayobe-tox-molecule/f6890b7/ : SUCCESS in 22m 12s\n- kayobe-overcloud-centos http://logs.openstack.org/01/566101/2/check/kayobe-overcloud-centos/0e09c36/ : SUCCESS in 42m 09s\n- kayobe-seed-centos http://logs.openstack.org/01/566101/2/check/kayobe-seed-centos/ac7b4dc/ : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":2},{"id":"53ef7dc9eb6592256409134507082a078c9b81c0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-05-04 13:27:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI can\u0027t read the RedHat solution (it\u0027s pay walled for me) but if it\u0027s not working it seems sensible to disable by default.","accounts_in_message":[],"_revision_number":2},{"id":"d0f1d2d997492e5e4da19a31876977df0a141539","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-04 13:27:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5da1fb6a48bce08c4e0ccec325920cbfb3db4cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-04 14:07:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/566101/2/gate/openstack-tox-pep8/b33fd0b/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/01/566101/2/gate/openstack-tox-py27/5c344f4/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/01/566101/2/gate/openstack-tox-py35/bae0f0d/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/01/566101/2/gate/openstack-tox-cover/d7cbf04/cover/ : SUCCESS in 4m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/01/566101/2/gate/build-openstack-sphinx-docs/68aff79/html/ : SUCCESS in 5m 20s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/01/566101/2/gate/kayobe-tox-ansible-syntax/e87809b/ : SUCCESS in 4m 13s\n- kayobe-tox-ansible http://logs.openstack.org/01/566101/2/gate/kayobe-tox-ansible/f5573aa/ : SUCCESS in 6m 18s\n- kayobe-tox-molecule http://logs.openstack.org/01/566101/2/gate/kayobe-tox-molecule/6992ce5/ : SUCCESS in 26m 26s\n- kayobe-overcloud-centos http://logs.openstack.org/01/566101/2/gate/kayobe-overcloud-centos/4755439/ : SUCCESS in 38m 58s\n- kayobe-seed-centos http://logs.openstack.org/01/566101/2/gate/kayobe-seed-centos/84dc5c5/ : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c4f749e5e758db35181c1c268bfb180b2e0485f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-04 14:07:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b850bb90f719948b69822cb61ddf1e07a845ce11","revisions":{"02a4a8525778ef84913a8aa851034e012be0a1f2":{"kind":"REWORK","_number":1,"created":"2018-05-03 15:50:39.000000000","uploader":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"ref":"refs/changes/01/566101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/566101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/566101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/566101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/566101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/566101/1"}}},"commit":{"parents":[{"commit":"f86c0ff48fcfce9304d3dae519be27671d7f134f","subject":"Use queens version of dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f86c0ff48fcfce9304d3dae519be27671d7f134f"}]}],"author":{"name":"Kevin Tibi","email":"kevintibi@hotmail.com","date":"2018-05-03 15:49:22.000000000","tz":120},"committer":{"name":"Kevin Tibi","email":"kevintibi@hotmail.com","date":"2018-05-03 15:49:22.000000000","tz":120},"subject":"Disable live-restore by default","message":"Disable live-restore by default\n\nLive-restore is not support by redhat[1]. We need\nto disable option by default.\n\n[1] https://access.redhat.com/solutions/2991041\n\nChange-Id: Ie4814983613a40fb6ff14aa63ddfd59f4ed62daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02a4a8525778ef84913a8aa851034e012be0a1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02a4a8525778ef84913a8aa851034e012be0a1f2"}]},"branch":"refs/heads/master"},"b850bb90f719948b69822cb61ddf1e07a845ce11":{"kind":"REWORK","_number":2,"created":"2018-05-04 12:27:10.000000000","uploader":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"ref":"refs/changes/01/566101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/566101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/566101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/566101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/566101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/566101/2"}}},"commit":{"parents":[{"commit":"f86c0ff48fcfce9304d3dae519be27671d7f134f","subject":"Use queens version of dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f86c0ff48fcfce9304d3dae519be27671d7f134f"}]}],"author":{"name":"Kevin Tibi","email":"kevintibi@hotmail.com","date":"2018-05-03 15:49:22.000000000","tz":120},"committer":{"name":"Kevin Tibi","email":"kevintibi@hotmail.com","date":"2018-05-04 12:26:50.000000000","tz":120},"subject":"Disable live-restore by default","message":"Disable live-restore by default\n\nLive-restore is not support by redhat[1]. We need\nto disable option by default.\n\n[1] https://access.redhat.com/solutions/2991041\n\nChange-Id: Ie4814983613a40fb6ff14aa63ddfd59f4ed62daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b850bb90f719948b69822cb61ddf1e07a845ce11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b850bb90f719948b69822cb61ddf1e07a845ce11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
