)]}'
{"id":"openstack%2Fkayobe~620074","triplet_id":"openstack%2Fkayobe~master~I0665806708d507d9c76f7b5075ceed5837af1597","project":"openstack/kayobe","branch":"master","topic":"kolla-host-vars","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-13 12:42:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-02 15:11:57.000000000","reason":"A robot voted negatively on a label"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-05-16 11:49:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0665806708d507d9c76f7b5075ceed5837af1597","subject":"Test control plane deployment using tenks","status":"NEW","created":"2018-11-26 14:26:47.000000000","updated":"2024-05-16 12:07:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"2b5c24debdd58acdd888418248713a5348f4e72e","_number":620074,"virtual_id_number":620074,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-16 12:07:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"reviewer_updates":[{"updated":"2019-01-11 10:35:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-02-01 17:57:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-01-03 18:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-13 12:42:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-05-16 11:49:53.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"94263b664f236d5768f698c4038abd56600fdddd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-26 14:26:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8b72d44739b3f8407b5279ce8c3de8b959b5dc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 15:39:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/620074/1/check/build-openstack-sphinx-docs/ec9fbf8/html/ : SUCCESS in 4m 41s\n- openstack-tox-cover http://logs.openstack.org/74/620074/1/check/openstack-tox-cover/4a9960f/cover/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/74/620074/1/check/openstack-tox-py35/299f8f6/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/74/620074/1/check/openstack-tox-pep8/921a0f2/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/74/620074/1/check/openstack-tox-py27/09a02d1/ : SUCCESS in 4m 56s\n- kayobe-seed-centos http://logs.openstack.org/74/620074/1/check/kayobe-seed-centos/fb1b9af/ : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"050bdcc3a59adeb57e1c7486ae8e470d428d6879","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-26 16:00:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"897f629bdff279f8ea7d9f720935c6b8e072c70a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 18:42:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/620074/2/check/build-openstack-sphinx-docs/318b0a3/html/ : SUCCESS in 11m 50s\n- openstack-tox-cover http://logs.openstack.org/74/620074/2/check/openstack-tox-cover/5703232/cover/ : SUCCESS in 5m 38s\n- openstack-tox-py35 http://logs.openstack.org/74/620074/2/check/openstack-tox-py35/b23294b/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/74/620074/2/check/openstack-tox-pep8/d9bf999/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/74/620074/2/check/openstack-tox-py27/6d5a7de/ : SUCCESS in 34m 07s\n- kayobe-seed-centos http://logs.openstack.org/74/620074/2/check/kayobe-seed-centos/8589887/ : FAILURE in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b15ed3a6ecc5d2fa14d43ea4231bd1d4861580c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-26 19:16:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee90425de3c782009a4277dac4e31161953882a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 23:33:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/620074/3/check/build-openstack-sphinx-docs/09ccfb4/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/74/620074/3/check/openstack-tox-cover/fefd6fe/cover/ : SUCCESS in 6m 11s\n- openstack-tox-py35 http://logs.openstack.org/74/620074/3/check/openstack-tox-py35/4e43d80/ : SUCCESS in 6m 10s\n- openstack-tox-pep8 http://logs.openstack.org/74/620074/3/check/openstack-tox-pep8/6f401c2/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/74/620074/3/check/openstack-tox-py27/87c5301/ : SUCCESS in 4m 45s\n- kayobe-seed-centos http://logs.openstack.org/74/620074/3/check/kayobe-seed-centos/e37ad97/ : FAILURE in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e367ff74b0626373f51da355897db46bdbb1f99e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-27 10:10:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"40871959c656f0b0d6771e04ebd1ebf7bfb2ebea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 11:47:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/620074/3/check/build-openstack-sphinx-docs/ce300da/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/74/620074/3/check/openstack-tox-cover/f47356c/cover/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/74/620074/3/check/openstack-tox-py35/6ac7c5b/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/74/620074/3/check/openstack-tox-pep8/88ee7e1/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/74/620074/3/check/openstack-tox-py27/829cb04/ : SUCCESS in 17m 47s\n- kayobe-seed-centos http://logs.openstack.org/74/620074/3/check/kayobe-seed-centos/66c1bb2/ : FAILURE in 36m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b14e48fb0605fbf837d8494493976b11653964a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-27 14:08:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e2f548166fd8088adb151893f97ae845a62f18e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 15:14:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/620074/4/check/build-openstack-sphinx-docs/7785568/html/ : SUCCESS in 5m 02s\n- openstack-tox-cover http://logs.openstack.org/74/620074/4/check/openstack-tox-cover/df7ca9b/cover/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/74/620074/4/check/openstack-tox-py35/8e9abbd/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/74/620074/4/check/openstack-tox-pep8/2d259e1/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/74/620074/4/check/openstack-tox-py27/c32f62d/ : SUCCESS in 4m 39s\n- kayobe-seed-centos http://logs.openstack.org/74/620074/4/check/kayobe-seed-centos/4371289/ : RETRY_LIMIT in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"088f2fe8b20ec9051d256b53a26721fa18101af3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-27 15:20:32.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"01d608f71951ea2c94be94deea994a805a13f00b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 17:38:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/5/check/kayobe-seed-centos/c6a3dc5/ : FAILURE in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1d721c4b228d0b35b37b75180d19d51f8a645c1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-27 19:10:30.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"43ad02a45c0d959397d319f96a2c69b3437f2a94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 21:31:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/6/check/kayobe-seed-centos/911922c/ : RETRY_LIMIT in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0f6e0368a218fd0387d194c9a4ec2fddee083641","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-28 13:52:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"285febdcafc70f14252300f62c9133ec20747e05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 15:43:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/7/check/kayobe-seed-centos/5f040aa/ : FAILURE in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"01ad261a26147df3282e242ef9e722ac802847f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-28 16:06:18.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0b6454fcdba8e1814c7f6cf90f7904536967419a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 19:33:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/7/check/kayobe-seed-centos/efaca64/ : FAILURE in 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1cbede9d0574b6dba2abbced9c17a0d35b8e9fd9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-29 09:41:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9f7028e9cc1b0b5a5534066bcfd4b0792745e678","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-29 12:33:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/8/check/kayobe-seed-centos/63ac2aa/ : FAILURE in 55m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"740e4f47470f33e55c8917254f9ec0a506699040","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-29 13:17:36.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7b4362943eb72f32bb117738b98983552d8f0450","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-29 17:32:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/8/check/kayobe-seed-centos/ecce2cd/ : FAILURE in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aea9dfd9f60982116e3db694162bbdeb92d4fcbc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-29 17:41:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4197e5966b3f1a98ce43c3df663cd21aa8220494","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-11-29 18:03:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c89120532cb37b97d9a34088a05e7b0082665b1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-29 22:08:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/10/check/kayobe-seed-centos/3b5d94a/ : FAILURE in 41m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c9cebbdbea4f91c91b2fcbba638c4a42789478d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-02 21:38:39.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"853b1c81f5cca73d6a1d20c1c5b899c305cfdac3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-02 21:41:38.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"93024893edf561615e8f586c328cbd3625821523","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-02 22:22:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/12/check/kayobe-seed-centos/99667db/ : FAILURE in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a6a706cecea9c7ba170bf7080dad491d13fd9543","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-07 14:45:48.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"79afbdd46dee911dded392ba06db575dc3ac703b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-07 14:47:25.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"39423670334b278b0e16ccd71d60aa296efa7ddd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 14:59:14.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/14/check/kayobe-seed-centos/b608f84/ : FAILURE in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6a58743950aaf36b452c97e85a22c7cce85e9075","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-08 13:58:54.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"4c29743f74305533575ca11dbc1c694f315a7982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-08 14:50:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/14/check/kayobe-seed-centos/2c37110/ : FAILURE in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8c5b8d921b4ad5d2d7fcded8782efaf755a50ce3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-08 15:17:03.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"5d5a5132c2b86691d0750192f22c925c39b8e50f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-08 16:02:55.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/15/check/kayobe-seed-centos/cdc2022/ : FAILURE in 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3b509f14a9ce9b64ef9fd02a6f1d4f3292adddf1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-08 16:33:14.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"8045e6cad11ea39d95d7b9c7ff5c6d7f06de8703","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-08 17:16:00.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/16/check/kayobe-seed-centos/a82915a/ : FAILURE in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"36dfa5d095161d024ba4f981242c4add4841f69e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-08 17:24:28.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"0a49bad5776114881ba62404132446969ae372bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-08 18:50:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/16/check/kayobe-seed-centos/fe51322/ : FAILURE in 1h 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"166cf2aee407a356e1bc0903c97da6827cecbef1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-09 10:28:16.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"17858188e12d23f372b9aa9a6b522e88186cc2d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-09 11:53:54.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/17/check/kayobe-seed-centos/eb39527/ : FAILURE in 1h 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"fec22de1bca949f39959f5add070e979ab45b15a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-13 10:17:00.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"a2fc6d7209b1403a638e2ee2794eff516b2a09e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 10:32:59.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/18/check/kayobe-seed-centos/feaa877/ : FAILURE in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c2e0aa4748191e70875b23a2d9d6b7e1dc43447c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 17:23:44.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e89e97553c810d4dc44cdc3a610502cf0fb66966","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 17:48:02.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/19/check/kayobe-seed-centos/f517bad/ : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":19},{"id":"e38d20c489cced808c356baae89a03544928304a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 17:56:12.000000000","message":"Patch Set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"13b365a9645d2d564922414f8ea90f1b07e0ca79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 18:34:06.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/20/check/kayobe-seed-centos/a147f61/ : FAILURE in 35m 03s","accounts_in_message":[],"_revision_number":20},{"id":"f0267a96dc1f1ee6a718c511d94958dcec07fa2f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 18:58:00.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"52be50490c6e2403103c8aa391c599ddce06d811","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 18:59:11.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"049eaee47281a34a40ebdb88fe6e48e28dc85fcf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 19:01:25.000000000","message":"Patch Set 22:\n\nLast failure: no route to host during inspection.","accounts_in_message":[],"_revision_number":22},{"id":"7b6bc22d358a1d46bb6230937a9d2d54fc9dc07e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 19:05:29.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"0ab8c11992fbd055c82356e09d27fe339edca841","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 19:46:28.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/23/check/kayobe-seed-centos/14e68f4/ : FAILURE in 37m 52s","accounts_in_message":[],"_revision_number":23},{"id":"327b7e9861959c1f86677bb64f3ac793f858ee73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-20 19:50:33.000000000","message":"Patch Set 23: Workflow-1\n\nrecheck\n\nPXE failed:\n\npxelinux.cfg/default... ok\n\nConfiguring (net0 52:54:00:e3:15:41)............... ok\n\nhttp://192.168.33.5:8080/ipa.vmlinuz............... Connection timed out (http://ipxe.org/4c0a6035)\n\nCould not boot: Connection timed out (http://ipxe.org/4c0a6035)\n\nPXE boot failed. No configuration found for MAC 52:54:00:e3:15:41","accounts_in_message":[],"_revision_number":23},{"id":"076545981bf4de1ce64b6a60acf7315134a22b36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 20:46:41.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/23/check/kayobe-seed-centos/92ebae8/ : FAILURE in 47m 50s","accounts_in_message":[],"_revision_number":23},{"id":"a6de8fd71c7ce841c7bb11bb97bc9bd1500de4ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-21 09:14:13.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"e8f49f5245b3c9c39b991d46ad743c7c3282768e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-12-21 09:14:19.000000000","message":"Patch Set 24: Workflow-1","accounts_in_message":[],"_revision_number":24},{"id":"448dbb31db213de84830092be7298a21912faf9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 09:54:09.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/24/check/kayobe-seed-centos/a0200bc/ : FAILURE in 36m 36s","accounts_in_message":[],"_revision_number":24},{"id":"ee1a5fff7fd17da0d14ee9e0274ff477be3af156","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-01-11 10:35:43.000000000","message":"Patch Set 24: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"7cc4681712313db120fa62294297f3ea7b37819d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 10:42:40.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"3d412a4ee9c37be485f19dec7692a024e73def53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 11:21:28.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/25/check/kayobe-seed-centos/2fd1e82/ : FAILURE in 37m 12s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/25/check/kayobe-seed-upgrade-centos/7c3f6b8/ : FAILURE in 22m 28s","accounts_in_message":[],"_revision_number":25},{"id":"cdccf3baf3ca313f5c092c903793a6cc48ca184e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-01-11 14:15:19.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"b923006adbdd6382f1e472c48ae08f474fda603e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-01-11 14:18:48.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"be1e4f0aa45c4503a3fca3ec1f378b39d24f0a5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 16:33:06.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"bc65ffd668b1ba6487b0e9991653f9c2054cfd73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-11 16:45:51.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a67da7c7b21e37ed779f7ee660d277faa7a7bdf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 17:37:09.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/26/check/kayobe-seed-centos/49dc87d/ : FAILURE in 45m 13s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/26/check/kayobe-seed-upgrade-centos/55bc69e/ : FAILURE in 24m 03s","accounts_in_message":[],"_revision_number":26},{"id":"41bb15c88313ff5049dbbe9ece737b02d867e560","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-29 10:54:13.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"a107a1a6828c4f0a17d5d4fc3923a20ebce99dda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 11:39:21.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/27/check/kayobe-seed-centos/f0e64ab/ : FAILURE in 35m 04s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/27/check/kayobe-seed-upgrade-centos/dda7926/ : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":27},{"id":"f45bd4ab515d7a9076643e23e0c992c5c277bf87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-29 12:16:34.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"bc172c84e68f957ae3b771b8d5024dd85c48810f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-29 12:18:24.000000000","message":"Patch Set 28:\n\nFailed during inspection, ramdisk could not reach inspector API:\n\nCRITICAL ironic-python-agent [-] Unhandled error: ConnectionError: HTTPConnectionPool(host\u003d\u0027192.168.33.5\u0027, port\u003d5050): Max retries exceeded with url: /v1/continue (Caused by NewConnectionError(\u0027\u003curllib3.connection.HTTPConnection object at 0x7fe6491a0050\u003e: Failed to establish a new connection: [Errno 113] No route to host\u0027,))\n\nCould try skipping inspection.","accounts_in_message":[],"_revision_number":28},{"id":"700b8c532c7ea8bac02d946d834ffc035115c72f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 13:06:00.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/28/check/kayobe-seed-centos/a192cdf/ : FAILURE in 37m 22s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/28/check/kayobe-seed-upgrade-centos/61e4973/ : SUCCESS in 25m 37s","accounts_in_message":[],"_revision_number":28},{"id":"560c88c7162f9cffa3b082f341a7459f7ad45a90","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-29 13:34:48.000000000","message":"Patch Set 28:\n\nrecheck\n\nIPA can\u0027t be downloaded during PXE boot.","accounts_in_message":[],"_revision_number":28},{"id":"844e17ed0b61f5c5034f766c69329e37b7fa74ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 14:28:10.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/28/check/kayobe-seed-centos/4a85c3d/ : FAILURE in 40m 13s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/28/check/kayobe-seed-upgrade-centos/8512481/ : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":28},{"id":"2ff29938d591f08178b4ab96eadf583b8227b3bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-01-31 16:58:48.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"80e1c4f92334fca67eea5cde125b9742fe0667c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 17:55:09.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/29/check/kayobe-seed-centos/f887e6d/ : FAILURE in 48m 14s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/29/check/kayobe-seed-upgrade-centos/41a32f8/ : SUCCESS in 26m 56s","accounts_in_message":[],"_revision_number":29},{"id":"4054a01e8d78831fe2e69dad0e545354301a43af","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-02-01 17:57:22.000000000","message":"Patch Set 29:\n\nLooks like another deploy failure:\n\n```\nTASK [Wait for the ironic node to become active] *******************************\nFAILED - RETRYING: Wait for the ironic node to become active (60 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (59 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (58 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (57 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (56 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (55 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (54 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (53 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (52 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (51 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (50 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (49 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (48 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (47 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (46 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (45 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (44 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (43 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (42 retries left).\nFAILED - RETRYING: Wait for the ironic node to become active (41 retries left).\nok: [controller0 -\u003e {{ hostvars[seed_host].ansible_host | default(seed_host) }}]\n\nTASK [Set a fact containing the final provision state] *************************\nok: [controller0]\n\nTASK [Fail if any of the nodes are not available] ******************************\nfatal: [controller0]: FAILED! \u003d\u003e {\"changed\": false, \"msg\": \"Ironic node for controller0 is in an unexpected provision state after provisioning.  Ironic provision state: deploy failed. Expected: active.\\n\"}```","accounts_in_message":[],"_revision_number":29},{"id":"cff5f210c02681efaafb6458eaf401f5c61d0132","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-29 17:28:59.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"1d0f9206da968022deabf81502d5bd3def120152","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 18:07:55.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/30/check/kayobe-seed-centos/df8b240/ : FAILURE in 24m 38s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/74/620074/30/check/kayobe-seed-upgrade-centos/5432f3d/ : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":30},{"id":"f2a5c5d87b762baaa69b0fa8a2aa7fe715970b1a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-29 18:31:42.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"7744d2127e3e027abb604a973a513056ba5807d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 19:47:47.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/31/check/kayobe-seed-centos/f1804e3/ : FAILURE in 1h 04m 03s","accounts_in_message":[],"_revision_number":31},{"id":"8b9dd4204d575eb79fc289c7a74fc132a749641f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-30 08:51:58.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"0fb3f16d0f32f3c5aeff3761dcecd320741b47ad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-30 08:52:19.000000000","message":"Patch Set 31:\n\nDevice too small: http://logs.openstack.org/74/620074/31/check/kayobe-seed-centos/f1804e3/primary/kolla/ironic/ironic-conductor.txt.gz#_2019-03-29_19_35_57_474","accounts_in_message":[],"_revision_number":31},{"id":"f009711778dbbef09308865274078c6fbc1a72ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 09:37:21.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos http://logs.openstack.org/74/620074/32/check/kayobe-seed-centos/ab1933b/ : FAILURE in 37m 47s","accounts_in_message":[],"_revision_number":32},{"id":"b94729127ab8fb1c35fe626535316ae6aff10407","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-12 17:57:05.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"8262919525aca56de620d348a1c0f6d7b230fb7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:18:14.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a14aa37aa7be47678bd021755c0645f0 : FAILURE in 19m 28s","accounts_in_message":[],"_revision_number":33},{"id":"6a117ae5a3e0cb3efef97b5f149fff4f0447f061","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 10:48:57.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"44bda7978e14c40c64ea31cca27f61d725f6cb33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 11:25:25.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4d8f3991d4a44210a0acbcc3e59acc42 : FAILURE in 24m 44s","accounts_in_message":[],"_revision_number":34},{"id":"ae208d698d12ee9a4867a5811dcf97c02156ea5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 11:29:43.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"236105f83926ff01869cd3025a87a1464e510829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 12:21:44.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/00a01981fbcc482894646c140e778b71 : FAILURE in 45m 55s","accounts_in_message":[],"_revision_number":35},{"id":"d1ae624d1ed2c6bb950bdc280d7d9228a61de999","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 12:33:30.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"e806f3fc9fb449d7935babcbe4864736139522e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 14:01:17.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f67fb2538bad464ca58411b941a7f241 : FAILURE in 1h 25m 02s","accounts_in_message":[],"_revision_number":36},{"id":"33f129ce6735cd6bd82e5cb38d4a2448d436b2ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 14:34:05.000000000","message":"Patch Set 36:\n\nNow fails during provisioning of the controller:\n https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_f67/620074/36/check/kayobe-seed-centos/f67fb25/primary/ansible/seed-test\n\nnode e3a0459e-d4a1-4132-9b2c-75399690b0e3 command status errored: {u\u0027message\u0027: u\u0027Command execution failed: Failed to create config drive on disk /dev/vda for node e3a0459e-d4a1-4132-9b2c-75399690b0e3. Error: Unexpected error while running command.\\nCommand: parted -a optimal -s -- /dev/vda mkpart primary fat32 -64MiB -0\\nExit code: 1\\nStdout: \\\u0027\\\u0027\\nStderr: \"Error: Can\\\u0027t have overlapping partitions.\\\\n\"\u0027, u\u0027code\u0027: 500, u\u0027type\u0027: u\u0027CommandExecutionError\u0027, u\u0027details\u0027: u\u0027Failed to create config drive on disk /dev/vda for node e3a0459e-d4a1-4132-9b2c-75399690b0e3. Error: Unexpected error while running command.\\nCommand: parted -a optimal -s -- /dev/vda mkpart primary fat32 -64MiB -0\\nExit code: 1\\nStdout: \\\u0027\\\u0027\\nStderr: \"Error: Can\\\u0027t have overlapping partitions.\\\\n\"\u0027}\n\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_f67/620074/36/check/kayobe-seed-centos/f67fb25/primary/kolla/ironic/ironic-conductor.txt.gz","accounts_in_message":[],"_revision_number":36},{"id":"447709caf22782f6ea9a11f1733ee476bfd957a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 12:16:02.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"e11248366e8e21c89cd517904df64a58c0f1a156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 12:56:45.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ea929e9da0654efdb093bcfd223d191e : FAILURE in 39m 46s","accounts_in_message":[],"_revision_number":37},{"id":"14ad4c1eb44ace8959aa27ab247c0dad5a5c059f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 13:49:21.000000000","message":"Patch Set 37:\n\nFailed due to https://storyboard.openstack.org/#!/story/2007068.","accounts_in_message":[],"_revision_number":37},{"id":"99f8b4b4c048d84b2d842bcf54a4417d3a406015","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 15:18:25.000000000","message":"Patch Set 38: Commit message was updated.","accounts_in_message":[],"_revision_number":38},{"id":"d6e18644f579dd927fbc6486c090fab271256ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 16:44:09.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9a0810f3707348dbb726f5571e342202 : FAILURE in 1h 24m 41s","accounts_in_message":[],"_revision_number":38},{"id":"c36fb4f9b28c4d7f6980093fb535dd51d341e58a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 16:57:12.000000000","message":"Patch Set 38:\n\nDeploy successful but cirros gets an IP from DHCP rather than the config drive.","accounts_in_message":[],"_revision_number":38},{"id":"5b95a811d9bf1cd4c133f7f35ecb5ec48f07547f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 16:58:01.000000000","message":"Patch Set 39: Published edit on patch set 38.","accounts_in_message":[],"_revision_number":39},{"id":"189141c2ab8297bbb0faf250e011b3334f2a0e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 18:20:08.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f53cb2e469fb47598a27f377e5f5ee71 : FAILURE in 1h 21m 04s","accounts_in_message":[],"_revision_number":39},{"id":"852cd3a01238a3e6ec3d923e40f0a2e0dc0df073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 13:56:09.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"79318586d4f968b79d778f7e04ce1608ea8a8120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:23:37.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/babf71b0704f45048fffe309c77d22cb : FAILURE in 18m 38s","accounts_in_message":[],"_revision_number":40},{"id":"6b4a80062903f6b4d3a0d9013b42dacc78ec94ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 14:46:16.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"41f14ce64ccd9f77ce9fdd945a131000b321d995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:14:13.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ec272da94f0040528a9faa487fba2d45 : FAILURE in 30m 17s","accounts_in_message":[],"_revision_number":41},{"id":"a702aa8e6de6ffc3f8490e2b2310748c464fd22a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 08:15:10.000000000","message":"Patch Set 42: Published edit on patch set 41.","accounts_in_message":[],"_revision_number":42},{"id":"f791d085bff46dc9eb3e7cfcf311434820a06b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:27:16.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9d11e50d50e349ad8af941d4592f74a6 : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":42},{"id":"d8354b289c250211b3fd447ddc5fad9cc8ca0034","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 14:30:53.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"3f7b7a834be08a468e6549136964ad5b4d475b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 15:11:57.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a5a2848bb06c42589c6f808f21cf2a9a : FAILURE in 14m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c58862ae423c407d88c4a068f4128a06 : FAILURE in 32m 27s","accounts_in_message":[],"_revision_number":43},{"id":"badce1e246c7ba2eed16a6b8df701a419ef0badb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-12 16:56:29.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"a36e4f22f12fda6617d0aac70cba2709ea7b926e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 17:56:22.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e638ab332e674ef0bead7e8fec9ff47c\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e79441b6b05044f982766a8d6c0f1b56 : RETRY_LIMIT in 7m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c22ea709d55342fd973652d1341faaa5 : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":44},{"id":"7423948e9677e0bbc8ec7ad15f84e907a15d7030","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-09 15:02:46.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"69001e92b6c6c92478284bd83490f0e30e5b4e27","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 11:37:01.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"0581111763d917ac6273e68ce3ef2741989afa3e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 11:43:09.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"2f49af9a2a07054383bce766471d182bd866fc4a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 11:49:53.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"2b5c24debdd58acdd888418248713a5348f4e72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 12:07:20.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acad889cf5924e7ea9f756c8dfd933d2\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/af69ca5702bd46da852e816e77d3193b : FAILURE in 14m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d82e4f5833a044718b64952d2abb7e85 : FAILURE in 10m 51s","accounts_in_message":[],"_revision_number":46}],"current_revision_number":46,"current_revision":"8bbd465ca7baa6ede9fde467243453673597f9a9","revisions":{"ca467a8cd59867c4eae00e2fc0eb6ca363cb8d1f":{"kind":"REWORK","_number":1,"created":"2018-11-26 14:26:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/1"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 14:26:32.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca467a8cd59867c4eae00e2fc0eb6ca363cb8d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca467a8cd59867c4eae00e2fc0eb6ca363cb8d1f"}]},"branch":"refs/heads/master"},"70baca231cb04d27e83a9762be914708db5110f3":{"kind":"REWORK","_number":2,"created":"2018-11-26 16:00:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/2"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 16:00:45.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70baca231cb04d27e83a9762be914708db5110f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70baca231cb04d27e83a9762be914708db5110f3"}]},"branch":"refs/heads/master"},"fb1ad87398081f174f9dbb8b8b982994334a04de":{"kind":"REWORK","_number":3,"created":"2018-11-26 19:16:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/3"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 19:16:25.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb1ad87398081f174f9dbb8b8b982994334a04de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb1ad87398081f174f9dbb8b8b982994334a04de"}]},"branch":"refs/heads/master"},"ce4b402cbb66144163678281c0314951ee440436":{"kind":"REWORK","_number":4,"created":"2018-11-27 14:08:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/4"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-27 14:08:43.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce4b402cbb66144163678281c0314951ee440436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce4b402cbb66144163678281c0314951ee440436"}]},"branch":"refs/heads/master"},"0f642e6ab0165d46c904dfd198f5de0482d6f465":{"kind":"REWORK","_number":5,"created":"2018-11-27 15:20:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/5"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-27 15:20:28.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f642e6ab0165d46c904dfd198f5de0482d6f465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f642e6ab0165d46c904dfd198f5de0482d6f465"}]},"branch":"refs/heads/master"},"7a2e0a7314ee6a19b560e3fadb532d6a60d5e674":{"kind":"REWORK","_number":6,"created":"2018-11-27 19:10:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/6"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-27 19:10:25.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a2e0a7314ee6a19b560e3fadb532d6a60d5e674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a2e0a7314ee6a19b560e3fadb532d6a60d5e674"}]},"branch":"refs/heads/master"},"10ad4ba11de4f1e9de1f4f1945a70ee756f0e479":{"kind":"REWORK","_number":7,"created":"2018-11-28 13:52:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/7"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-28 13:52:45.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10ad4ba11de4f1e9de1f4f1945a70ee756f0e479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10ad4ba11de4f1e9de1f4f1945a70ee756f0e479"}]},"branch":"refs/heads/master"},"134837a7c5b32295f166f31dfc775dadb8000dc5":{"kind":"REWORK","_number":8,"created":"2018-11-29 09:41:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/8"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-29 09:41:02.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/134837a7c5b32295f166f31dfc775dadb8000dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/134837a7c5b32295f166f31dfc775dadb8000dc5"}]},"branch":"refs/heads/master"},"65ce8cc60ef0a64ea341cd70071b62eb4a5d0f4b":{"kind":"REWORK","_number":9,"created":"2018-11-29 17:41:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/9"}}},"commit":{"parents":[{"commit":"e172a4254fed9aa1831baee19b3f535cf86c152f","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e172a4254fed9aa1831baee19b3f535cf86c152f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-29 17:41:33.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65ce8cc60ef0a64ea341cd70071b62eb4a5d0f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65ce8cc60ef0a64ea341cd70071b62eb4a5d0f4b"}]},"branch":"refs/heads/master"},"802e39fec597b1c8bfa8c3b10583d19dc65a106e":{"kind":"REWORK","_number":10,"created":"2018-11-29 18:03:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/10"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-29 18:02:46.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/802e39fec597b1c8bfa8c3b10583d19dc65a106e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/802e39fec597b1c8bfa8c3b10583d19dc65a106e"}]},"branch":"refs/heads/master"},"ed6462cf79ab1dc9869c36d12e459cf4430b4665":{"kind":"REWORK","_number":11,"created":"2018-12-02 21:38:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/11"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-02 21:38:31.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed6462cf79ab1dc9869c36d12e459cf4430b4665"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed6462cf79ab1dc9869c36d12e459cf4430b4665"}]},"branch":"refs/heads/master"},"636b21a667c3fb085e8c65c44718c214096c4fa9":{"kind":"REWORK","_number":12,"created":"2018-12-02 21:41:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/12"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-02 21:41:34.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/636b21a667c3fb085e8c65c44718c214096c4fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/636b21a667c3fb085e8c65c44718c214096c4fa9"}]},"branch":"refs/heads/master"},"901f9a7676b1b1106b60104e1d82a3efc9df8f8a":{"kind":"REWORK","_number":13,"created":"2018-12-07 14:45:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/13"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-07 14:45:40.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901f9a7676b1b1106b60104e1d82a3efc9df8f8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901f9a7676b1b1106b60104e1d82a3efc9df8f8a"}]},"branch":"refs/heads/master"},"91eee39f244db799679d2c1e89104724c5f72f79":{"kind":"REWORK","_number":14,"created":"2018-12-07 14:47:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/14"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-07 14:47:21.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91eee39f244db799679d2c1e89104724c5f72f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91eee39f244db799679d2c1e89104724c5f72f79"}]},"branch":"refs/heads/master"},"1410012e48eac3f72ba2bd933a96f59eb5d860ea":{"kind":"REWORK","_number":15,"created":"2018-12-08 15:17:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/15"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-08 15:16:57.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1410012e48eac3f72ba2bd933a96f59eb5d860ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1410012e48eac3f72ba2bd933a96f59eb5d860ea"}]},"branch":"refs/heads/master"},"b26e6f2f29442ef703b70293c57ac7a0873c43f5":{"kind":"REWORK","_number":16,"created":"2018-12-08 16:33:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/16"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-08 16:33:08.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b26e6f2f29442ef703b70293c57ac7a0873c43f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b26e6f2f29442ef703b70293c57ac7a0873c43f5"}]},"branch":"refs/heads/master"},"a592cac56bab3da30e78c1b240477064589fe6ac":{"kind":"REWORK","_number":17,"created":"2018-12-09 10:28:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/17"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-09 10:26:13.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a592cac56bab3da30e78c1b240477064589fe6ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a592cac56bab3da30e78c1b240477064589fe6ac"}]},"branch":"refs/heads/master"},"fc5e26abd96632ee6c164ce88676c39f939f0e23":{"kind":"REWORK","_number":18,"created":"2018-12-13 10:17:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/18"}}},"commit":{"parents":[{"commit":"76daa4e6de1233ebe799224dd671222277ee11ef","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76daa4e6de1233ebe799224dd671222277ee11ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-09 14:26:11.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc5e26abd96632ee6c164ce88676c39f939f0e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc5e26abd96632ee6c164ce88676c39f939f0e23"}]},"branch":"refs/heads/master"},"fceb7fd9b9d23171416581fd2acbbb9f5542d1cf":{"kind":"REWORK","_number":19,"created":"2018-12-20 17:23:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/19"}}},"commit":{"parents":[{"commit":"bff985330af53f65b2ff2328acb76a835866387b","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff985330af53f65b2ff2328acb76a835866387b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-20 17:21:53.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620052\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fceb7fd9b9d23171416581fd2acbbb9f5542d1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fceb7fd9b9d23171416581fd2acbbb9f5542d1cf"}]},"branch":"refs/heads/master"},"3d35413be64693d7611d4ff06b10101c5c44e488":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2018-12-20 17:56:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/20"}}},"commit":{"parents":[{"commit":"bff985330af53f65b2ff2328acb76a835866387b","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff985330af53f65b2ff2328acb76a835866387b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-20 17:56:07.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d35413be64693d7611d4ff06b10101c5c44e488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d35413be64693d7611d4ff06b10101c5c44e488"}]},"branch":"refs/heads/master"},"52514527a6b7f5b3b0b4532bfad26133cae4c24e":{"kind":"REWORK","_number":21,"created":"2018-12-20 18:58:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/21"}}},"commit":{"parents":[{"commit":"bff985330af53f65b2ff2328acb76a835866387b","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff985330af53f65b2ff2328acb76a835866387b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-20 18:57:41.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52514527a6b7f5b3b0b4532bfad26133cae4c24e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52514527a6b7f5b3b0b4532bfad26133cae4c24e"}]},"branch":"refs/heads/master"},"289e1f6bb46e0ca9bb6fe6811f5f44ffd4dec420":{"kind":"REWORK","_number":22,"created":"2018-12-20 18:59:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/22"}}},"commit":{"parents":[{"commit":"bff985330af53f65b2ff2328acb76a835866387b","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff985330af53f65b2ff2328acb76a835866387b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-20 18:59:08.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/289e1f6bb46e0ca9bb6fe6811f5f44ffd4dec420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/289e1f6bb46e0ca9bb6fe6811f5f44ffd4dec420"}]},"branch":"refs/heads/master"},"05eeb926a723e1f87b35b49abbaecd2915dc1fd8":{"kind":"REWORK","_number":23,"created":"2018-12-20 19:05:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/23"}}},"commit":{"parents":[{"commit":"bff985330af53f65b2ff2328acb76a835866387b","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff985330af53f65b2ff2328acb76a835866387b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-20 19:05:24.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05eeb926a723e1f87b35b49abbaecd2915dc1fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05eeb926a723e1f87b35b49abbaecd2915dc1fd8"}]},"branch":"refs/heads/master"},"d0b5b2781f3b717a7ab29962434cffd6b154dd27":{"kind":"REWORK","_number":24,"created":"2018-12-21 09:14:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/24"}}},"commit":{"parents":[{"commit":"bff985330af53f65b2ff2328acb76a835866387b","subject":"Generate bifrost hostvars prior to provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bff985330af53f65b2ff2328acb76a835866387b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-12-21 09:14:07.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0b5b2781f3b717a7ab29962434cffd6b154dd27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0b5b2781f3b717a7ab29962434cffd6b154dd27"}]},"branch":"refs/heads/master"},"1305298635967f0a4e1fdbacbbbb3a97e2f58ee8":{"kind":"REWORK","_number":25,"created":"2019-01-11 10:42:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/25"}}},"commit":{"parents":[{"commit":"bf3d31aaa273af38693ea3aba96f6e56117b1c5e","subject":"Merge \"Perform ironic online data migrations prior to seed upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf3d31aaa273af38693ea3aba96f6e56117b1c5e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-11 10:42:22.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1305298635967f0a4e1fdbacbbbb3a97e2f58ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1305298635967f0a4e1fdbacbbbb3a97e2f58ee8"}]},"branch":"refs/heads/master"},"9285beef45a0c3009bedd012f2b26627b1ea871d":{"kind":"REWORK","_number":26,"created":"2019-01-11 16:45:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/26"}}},"commit":{"parents":[{"commit":"bf3d31aaa273af38693ea3aba96f6e56117b1c5e","subject":"Merge \"Perform ironic online data migrations prior to seed upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf3d31aaa273af38693ea3aba96f6e56117b1c5e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-11 16:45:34.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9285beef45a0c3009bedd012f2b26627b1ea871d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9285beef45a0c3009bedd012f2b26627b1ea871d"}]},"branch":"refs/heads/master"},"79a77b4bc8a5445044c240d9d8c53855fff5e872":{"kind":"REWORK","_number":27,"created":"2019-01-29 10:54:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/27"}}},"commit":{"parents":[{"commit":"9db9c413d4e6669d22b78f2f5188e4cd7fa085a2","subject":"Merge \"Allow Kayobe to wipe NVMe disks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9db9c413d4e6669d22b78f2f5188e4cd7fa085a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 10:54:05.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a77b4bc8a5445044c240d9d8c53855fff5e872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a77b4bc8a5445044c240d9d8c53855fff5e872"}]},"branch":"refs/heads/master"},"015b73ff452c419b6680e0c290f719a77f527e5c":{"kind":"REWORK","_number":28,"created":"2019-01-29 12:16:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/28"}}},"commit":{"parents":[{"commit":"9db9c413d4e6669d22b78f2f5188e4cd7fa085a2","subject":"Merge \"Allow Kayobe to wipe NVMe disks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9db9c413d4e6669d22b78f2f5188e4cd7fa085a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-11-26 13:58:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 12:16:24.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/626635\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/015b73ff452c419b6680e0c290f719a77f527e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/015b73ff452c419b6680e0c290f719a77f527e5c"}]},"branch":"refs/heads/master"},"3f05a363b88e34df1b302b198cb9fdcd57ad2a1e":{"kind":"REWORK","_number":29,"created":"2019-01-31 16:58:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/29"}}},"commit":{"parents":[{"commit":"e238ad5e0cb5f60ff0b12a66b4fcf8ff3ee43003","subject":"Update development scripts for control plane deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e238ad5e0cb5f60ff0b12a66b4fcf8ff3ee43003"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-31 16:58:30.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f05a363b88e34df1b302b198cb9fdcd57ad2a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f05a363b88e34df1b302b198cb9fdcd57ad2a1e"}]},"branch":"refs/heads/master"},"499af0e31917cb7dd6b6516274b7bc854fe3a119":{"kind":"REWORK","_number":30,"created":"2019-03-29 17:28:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/30"}}},"commit":{"parents":[{"commit":"08bb1441ebc46beeeb2359b434dc99559adb6540","subject":"Prevent use of KVM for Tenks VMs in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08bb1441ebc46beeeb2359b434dc99559adb6540"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-29 17:28:34.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/499af0e31917cb7dd6b6516274b7bc854fe3a119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/499af0e31917cb7dd6b6516274b7bc854fe3a119"}]},"branch":"refs/heads/master"},"e8e673346398b93d228d561174af05cf5e9ba739":{"kind":"REWORK","_number":31,"created":"2019-03-29 18:31:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/31"}}},"commit":{"parents":[{"commit":"08bb1441ebc46beeeb2359b434dc99559adb6540","subject":"Prevent use of KVM for Tenks VMs in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08bb1441ebc46beeeb2359b434dc99559adb6540"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-29 18:31:33.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8e673346398b93d228d561174af05cf5e9ba739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8e673346398b93d228d561174af05cf5e9ba739"}]},"branch":"refs/heads/master"},"4082b833c618db6ba8e9b1b1313c31bbd8cb35e1":{"kind":"REWORK","_number":32,"created":"2019-03-30 08:51:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/32"}}},"commit":{"parents":[{"commit":"08bb1441ebc46beeeb2359b434dc99559adb6540","subject":"Prevent use of KVM for Tenks VMs in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08bb1441ebc46beeeb2359b434dc99559adb6540"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-30 08:51:47.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4082b833c618db6ba8e9b1b1313c31bbd8cb35e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4082b833c618db6ba8e9b1b1313c31bbd8cb35e1"}]},"branch":"refs/heads/master"},"b4b84b24faf05ec6af4abf1f4337ffc103749469":{"kind":"REWORK","_number":33,"created":"2019-12-12 17:57:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/33"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-12 17:56:52.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b84b24faf05ec6af4abf1f4337ffc103749469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b84b24faf05ec6af4abf1f4337ffc103749469"}]},"branch":"refs/heads/master"},"36368a2abb91684970e206605bd50e2590cfbf89":{"kind":"REWORK","_number":34,"created":"2019-12-13 10:48:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/34"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 10:48:22.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36368a2abb91684970e206605bd50e2590cfbf89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36368a2abb91684970e206605bd50e2590cfbf89"}]},"branch":"refs/heads/master"},"6199da7eea3ac05f66e05b94f5a8863c3a8f2084":{"kind":"REWORK","_number":35,"created":"2019-12-13 11:29:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/35"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 11:29:34.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6199da7eea3ac05f66e05b94f5a8863c3a8f2084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6199da7eea3ac05f66e05b94f5a8863c3a8f2084"}]},"branch":"refs/heads/master"},"6db9f6def04bd93d9f4d5add4a7629635fc9e52d":{"kind":"REWORK","_number":36,"created":"2019-12-13 12:33:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/36"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-13 12:33:20.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6db9f6def04bd93d9f4d5add4a7629635fc9e52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6db9f6def04bd93d9f4d5add4a7629635fc9e52d"}]},"branch":"refs/heads/master"},"3993b5aed9ff1dfbbf7e0a436428a50c3b6b12d7":{"kind":"REWORK","_number":37,"created":"2020-01-03 12:16:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/37"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 12:15:56.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.openstack.org/620071\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3993b5aed9ff1dfbbf7e0a436428a50c3b6b12d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3993b5aed9ff1dfbbf7e0a436428a50c3b6b12d7"}]},"branch":"refs/heads/master"},"8a61408cd3379136af71c74ff878f557079971e4":{"kind":"NO_CODE_CHANGE","_number":38,"created":"2020-01-03 15:18:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/38"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 15:18:20.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/620071\nDepends-On: https://review.opendev.org/701032\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a61408cd3379136af71c74ff878f557079971e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a61408cd3379136af71c74ff878f557079971e4"}]},"branch":"refs/heads/master"},"89eccff39f2c4b5760b6aa34e44d1c3beec6d621":{"kind":"REWORK","_number":39,"created":"2020-01-03 16:58:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/39"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 16:57:58.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/620071\nDepends-On: https://review.opendev.org/701032\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89eccff39f2c4b5760b6aa34e44d1c3beec6d621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89eccff39f2c4b5760b6aa34e44d1c3beec6d621"}]},"branch":"refs/heads/master"},"405af3e46683bb8ca4ebbdb270709d459ef04001":{"kind":"REWORK","_number":40,"created":"2021-06-07 13:56:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/40"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 13:55:54.000000000","tz":60},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/405af3e46683bb8ca4ebbdb270709d459ef04001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/405af3e46683bb8ca4ebbdb270709d459ef04001"}]},"branch":"refs/heads/master"},"65f7f04433f6dfabc43631aadc52a547094ed936":{"kind":"REWORK","_number":41,"created":"2021-06-07 14:46:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/41"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:46:07.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65f7f04433f6dfabc43631aadc52a547094ed936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65f7f04433f6dfabc43631aadc52a547094ed936"}]},"branch":"refs/heads/master"},"4b1a8c3314b8c6a083688eb6f5161ccd173a34e6":{"kind":"REWORK","_number":42,"created":"2021-06-08 08:15:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/42"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 08:15:07.000000000","tz":0},"subject":"WIP: Test control plane deployment using tenks","message":"WIP: Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b1a8c3314b8c6a083688eb6f5161ccd173a34e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b1a8c3314b8c6a083688eb6f5161ccd173a34e6"}]},"branch":"refs/heads/master"},"4c7a49a3adf28800866e76385a2a7ec00cec107c":{"kind":"REWORK","_number":43,"created":"2022-02-02 14:30:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/620074/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/43"}}},"commit":{"parents":[{"commit":"d6e836586762e41211940532b736967bad631cc3","subject":"Merge \"Support for untemplated dirs in kolla config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6e836586762e41211940532b736967bad631cc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 14:25:59.000000000","tz":0},"subject":"Test control plane deployment using tenks","message":"Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c7a49a3adf28800866e76385a2a7ec00cec107c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c7a49a3adf28800866e76385a2a7ec00cec107c"}]},"branch":"refs/heads/master"},"1428ed0f91a4d9db34ffa31c0ea52ac00caf8b09":{"kind":"REWORK","_number":44,"created":"2023-12-12 16:56:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/620074/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/44"}}},"commit":{"parents":[{"commit":"bb2af07567152832700a2c2971cd72ff4e3b77d3","subject":"Merge \"Revert \"Use OpenStack 2023.2 release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb2af07567152832700a2c2971cd72ff4e3b77d3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-12 16:56:00.000000000","tz":60},"subject":"Test control plane deployment using tenks","message":"Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1428ed0f91a4d9db34ffa31c0ea52ac00caf8b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1428ed0f91a4d9db34ffa31c0ea52ac00caf8b09"}]},"branch":"refs/heads/master"},"87e21b657eb9133fb805df1d87814c377683e8af":{"kind":"REWORK","_number":45,"created":"2024-05-16 11:37:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/74/620074/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/45"}}},"commit":{"parents":[{"commit":"b9ca63dbea737f41cec20914591bc0906bd4c6fb","subject":"Document required tags for Bifrost-only deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9ca63dbea737f41cec20914591bc0906bd4c6fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-16 11:36:53.000000000","tz":60},"subject":"Test control plane deployment using tenks","message":"Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87e21b657eb9133fb805df1d87814c377683e8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87e21b657eb9133fb805df1d87814c377683e8af"}]},"branch":"refs/heads/master"},"8bbd465ca7baa6ede9fde467243453673597f9a9":{"kind":"REWORK","_number":46,"created":"2024-05-16 11:43:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/74/620074/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/620074/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/620074/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/620074/46"}}},"commit":{"parents":[{"commit":"b9ca63dbea737f41cec20914591bc0906bd4c6fb","subject":"Document required tags for Bifrost-only deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9ca63dbea737f41cec20914591bc0906bd4c6fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-01-29 17:28:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-16 11:42:16.000000000","tz":0},"subject":"Test control plane deployment using tenks","message":"Test control plane deployment using tenks\n\nUses the seed to deploy a \u0027bare metal\u0027 control plane.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/620071/\nChange-Id: I0665806708d507d9c76f7b5075ceed5837af1597\nStory: 2004297\nTask: 27853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bbd465ca7baa6ede9fde467243453673597f9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bbd465ca7baa6ede9fde467243453673597f9a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
