)]}'
{"id":"openstack%2Fkayobe~635586","triplet_id":"openstack%2Fkayobe~master~I2a7a82d7f576739c5516a0072f953712ffa5c233","project":"openstack/kayobe","branch":"master","topic":"story/2004959","hashtags":[],"change_id":"I2a7a82d7f576739c5516a0072f953712ffa5c233","subject":"Make Kayobe code compatible with Python 3","status":"MERGED","created":"2019-02-07 17:43:44.000000000","updated":"2020-02-28 21:21:02.000000000","submitted":"2020-02-28 21:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"635586-1582924729966-d353a4dd","meta_rev_id":"ec0f8af65a77c0811b0e519327d3f1321959efd5","_number":635586,"virtual_id_number":635586,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"date":"2020-02-27 15:28:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 21:18:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-02-28 19:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 11:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-02-28 19:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-07 22:11:24.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2019-02-08 12:03:36.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-02-27 15:28:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-28 19:15:35.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-28 21:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"befeae46bc00790cef3a408fbc240c37c2fc06ec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-02-07 17:43:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1bdd03671f0e6b9225f3208c9ee4dd5db5bb542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 19:18:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/86/635586/1/check/build-openstack-sphinx-docs/3251b2a/html/ : SUCCESS in 6m 23s\n- openstack-tox-cover http://logs.openstack.org/86/635586/1/check/openstack-tox-cover/b5dd547/cover/ : SUCCESS in 7m 51s\n- openstack-tox-py35 http://logs.openstack.org/86/635586/1/check/openstack-tox-py35/732e380/ : SUCCESS in 7m 21s\n- openstack-tox-pep8 http://logs.openstack.org/86/635586/1/check/openstack-tox-pep8/0308e6d/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/86/635586/1/check/openstack-tox-py27/bfc0ed5/ : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/86/635586/1/check/kayobe-tox-ansible-syntax/d53f14d/ : SUCCESS in 8m 10s\n- kayobe-tox-ansible http://logs.openstack.org/86/635586/1/check/kayobe-tox-ansible/b19095a/ : SUCCESS in 12m 23s\n- kayobe-tox-molecule http://logs.openstack.org/86/635586/1/check/kayobe-tox-molecule/bc4481d/ : SUCCESS in 25m 35s\n- kayobe-overcloud-centos http://logs.openstack.org/86/635586/1/check/kayobe-overcloud-centos/9e74fc5/ : SUCCESS in 56m 05s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/86/635586/1/check/kayobe-overcloud-upgrade-centos/8f5e158/ : SUCCESS in 1h 22m 51s\n- kayobe-seed-centos http://logs.openstack.org/86/635586/1/check/kayobe-seed-centos/180fc76/ : SUCCESS in 18m 55s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/86/635586/1/check/kayobe-seed-upgrade-centos/67689f7/ : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"10c5ae1faedcddbfbc1fad916f510deed60a2e4e","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-02-07 22:11:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a5330f4986196c432d2d2296d545d4a79ff3ca4a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-02-08 09:34:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"28c251afa4d7e02a086e6939452425d4a44fcba8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-02-08 09:35:38.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c3eecfc9c0a80a52048f81fd40fc8f9c6b699aca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-02-08 09:46:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f4dac2eb42bab28034a4d929265deae756a406e4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-02-08 12:03:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74f6df6e999ae5ea1407a6a4ede2c480fcf55262","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-02-08 13:36:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6c2eb9ab37f93925b66caa88eabd0097a08bd93","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-06-18 16:26:04.000000000","message":"Removed Code-Review+1 by Bharat Kunwar \u003cbharat@stackhpc.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"ea1080a655c27ea15424a5dae1f90e56244e65a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 17:10:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f26c1b1cfc9c0f1ace01d9dbd6c4253f53e7fce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 18:33:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a1c9e583f154def8501d0835676e5f3 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2189e33126ca41849b58934c00538c0c : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e8e0bd26f2247e6b98b36d6cef37985 : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee4da20b7f2946bba0cdfae059458a30 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19894b0bfa6b47c1ad7d741534549b72 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0765ebce2f741c9a7696961b62674e5 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f483b1ea13b6435ca4944d2a97dabccb : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb787271eb4842cdb2ca68c6ca62f2dd : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/61e2d908c33e4942bd069a94d190073b : SUCCESS in 33m 09s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ecce27318682438eab10a55d44ca316a : FAILURE in 10m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/518a350c952b4e8ab5ae1a4cec3d5660 : FAILURE in 1h 19m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7aa7d1f5c3614f19a9df6111da29d6d2 : FAILURE in 9m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f29164d51c2945aaaf7e7087453d63e3 : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1236a5ab0b952d52f46eda4698bf33798005bad2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 13:47:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"00f0b7995d78f1b424ec0cdfc67a4e87417a445b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 15:39:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e9e1ce8c62ea8cc470b9810369c44b8083e5e797","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 17:02:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"04b3ea9738176720751199fa2fa2ed9bdd638ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 19:01:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/637b3602947e454a8a7b3da210d27458 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fabe8076e7c4b2da6aa5933929ab388 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b6e2e1d3fa410f900ebc6bf57654f6 : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/556c8a9b0c674cff9b9e9797817c938b : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/018bd111b8164e3c9e577e5c4e79485f : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e936c901ed80446ba9004d34d8f4a676 : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6aaa24264f6344178ece7947ca717adf : SUCCESS in 8m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/415629cf63a44d488a671a86d356b8e2 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c35346524f543569a07ec3c30d6a295 : SUCCESS in 40m 56s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/27364ef99b8e40208c624595f4821a56 : SUCCESS in 1h 07m 42s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0a8bea3f64ed46a0bb5e082e8b2221bb : SUCCESS in 1h 53m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/35534a76b3024b4986af600d092a4f13 : SUCCESS in 21m 52s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b5cdf7ec507346ccbf6dadcc45572cb7 : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":5},{"id":"47259aa2e9af2acf2b0d71581de80f971a900e2b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 15:43:01.000000000","message":"Topic changed from python3 to story/2004959","accounts_in_message":[],"_revision_number":5},{"id":"8b5966fdd207e6e96173955a6e60c45e6d9efa78","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 16:51:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"690d85d2747b53ffe6ae9c7d3eef13f83e1ecaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 18:48:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b600a7914924a9d9dd54bd58394328d : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7c403cc22b46b287e599f885b3a29e : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37d7594119ab4d978676a344028adfde : SUCCESS in 9m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66acc8df85cf4aefa9547083282a40c3 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/403d7c6633dc402e8afe5bd0b259b66e : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60fbbae5b5f14e338a1a7cdcd0d05983 : SUCCESS in 9m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b86a1a6fe75343778cecc9d2133a1b1a : SUCCESS in 10m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e26d5bf1a6840b19ded34f03bee8ad6 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2926722d2dd349c4abdad81693d6bd13 : SUCCESS in 40m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9da257e273a2442aa0f53f9b44913084 : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3e6f117b83234f3382975e021197f7e3 : SUCCESS in 1h 52m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b63d3276b7a74bb6aacd125286b69b70 : SUCCESS in 25m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/52301925519e4c6db3964d75d7c68646 : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":6},{"id":"8e5b524516453dc2eb961e8edd622f026be7990d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-26 16:23:20.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks sane, leaning on CI for this.","accounts_in_message":[],"_revision_number":6},{"id":"bd89823a6746951774b99cbd74ba582158cb6be5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-26 16:38:05.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)\n\nOverall looks good, minor comments.","accounts_in_message":[],"_revision_number":6},{"id":"2ec8c988752c96e90c1632b1f2831f4c25501311","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-26 16:50:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a677e637701381d189b51c4b59307b1575bbc8a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-26 17:02:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f144641e36b2212cb3dc2d7b7fbab06389cb5caa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 11:14:37.000000000","message":"Patch Set 6: Workflow-1\n\n(2 comments)\n\nThanks","accounts_in_message":[],"_revision_number":6},{"id":"bb6a7aea693de08e4fcf46dd8e66b1013a722797","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 11:24:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"671d5b97af495fccdfbb3e9490728a1924db7b63","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-27 11:38:57.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1818fe3bce9b0dae4f1cc8036cf8ff8187ed6d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:41:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/105e5e87fd6145cc875e7d347804ddcc : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc52a018c344f0f9c2a32a648985a74 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/300338156f53467fb2ee18cbe34ef2b5 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f995064481649b284ba15da5d2202e5 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6374565a9b3a4da183dcfa822275f0f9 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1acf71b99a4502a5b10567aa2b9865 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58510a89e02a47b9bc7c78756be4ec9a : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1c33ce79bba4583aef8f4520a6febb3 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d019f4a8e2da42fdbb3d007702e452ae : SUCCESS in 41m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/533aae698380444e89aef51d118eab0c : SUCCESS in 1h 15m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fc6085f88a814082a7369d9bb5773de7 : RETRY_LIMIT in 1h 02m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9f38f94a266844cd96d0b245c37ba553 : SUCCESS in 27m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/12145888dab84da88b8a4742865843a9 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":7},{"id":"c3ae2e3312d08211c9462084962d5ce87890a05e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:28:17.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8cb56eae608f26823727d95b7cb49ce51b3e1221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:02:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1edc4da985be46568c36d07aafa2e99c : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346094dc24594fd5ba086821b7d30c56 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f6d502a21f14f07b8eabd1a0a05b085 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98e053d3da874ac4b4ca7f93dbf91c02 : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ba956149f14c6884fc91e231b77bd9 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf130f32b824bb188eb45cc7e6b6bf4 : SUCCESS in 13m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c13a7b4e74884a8584b97c3ef255489a : SUCCESS in 7m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31da29ec220048bbabd6c95299cd8ecf : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b432a0102b9b4f8bb57ae88105b82cea : SUCCESS in 40m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8778a5aeefe94d88bc66949530c672b5 : SUCCESS in 1h 49m 56s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c543da1cc4b44ed3ae83ef52c60e1941 : SUCCESS in 2h 29m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e95921224dd547418331b91c4358dbc4 : SUCCESS in 35m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9a08fc97e44244f681ea8a2718e9eda7 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":7},{"id":"d0d81dca132f1b037c6373f1c11d06bc36a3ffc2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-28 19:15:35.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5cb390db3acd046f239797328e32543bb43c8f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:15:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9ea340e3793c936072621a26b33615ef742fd645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:18:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9d591f9b7867274c19542b11b21276b1e53e4cc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:18:49.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3748ab0b8fe74472b9b263e914e994ef : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40a6a0de3e7943febcb41c6a5cbf4640 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20404530f46c4d579c9dffea60d5a1dc : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa117529140d4954aa27749c553ad65e : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73d4af34655044c79d8e0a4de0d9a861 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6153784f6106415e8f3075b9b2933ee0 : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2f7e6f9a4a04ed3853a7bc2b0e1f1d9 : SUCCESS in 33m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/33937992d2a9464a8f0185ec8abc1525 : SUCCESS in 1h 15m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/41fe5c2f0e6c4882aeaf837839b9fd75 : SUCCESS in 2h 01m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9a841154cd924d809a55e3e06f00bf9d : SUCCESS in 24m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a7e3e76e28a24d5c85a8bf3368793ad2 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":7},{"id":"ec0f8af65a77c0811b0e519327d3f1321959efd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:21:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19531ad8b02948ee97bb72aa64b87e02 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"15109ccb547e603c0aca2a73e43f4db523fa1bcd","revisions":{"e6b9b2ce6db45c2bb5251d701716fc70c48c5e2a":{"kind":"REWORK","_number":1,"created":"2019-02-07 17:43:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/635586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/1"}}},"commit":{"parents":[{"commit":"288338d6b1bdbe830fed23b15a9a965f06ebdd28","subject":"Merge \"Remove support for deploying cAdvisor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/288338d6b1bdbe830fed23b15a9a965f06ebdd28"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"subject":"Make more Kayobe code compatible with Python 3","message":"Make more Kayobe code compatible with Python 3\n\nThis allows a `kayobe overcloud host configure` to complete (although\nKolla-Ansible was still using Python 2.7).\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6b9b2ce6db45c2bb5251d701716fc70c48c5e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6b9b2ce6db45c2bb5251d701716fc70c48c5e2a"}]},"branch":"refs/heads/master"},"8a0d2dc66e9c4d07188a2b41fd47efc6574d4ce9":{"kind":"REWORK","_number":2,"created":"2020-02-20 17:10:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/635586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/2"}}},"commit":{"parents":[{"commit":"4b6a079dcaaa81fa62ada445ff720abd9074585e","subject":"Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b6a079dcaaa81fa62ada445ff720abd9074585e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 17:10:24.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a0d2dc66e9c4d07188a2b41fd47efc6574d4ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a0d2dc66e9c4d07188a2b41fd47efc6574d4ce9"}]},"branch":"refs/heads/master"},"41d4768155711c3a95e5b13739e2fc3c1e0c25fa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-21 13:47:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/635586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/3"}}},"commit":{"parents":[{"commit":"80b336b35ee7036cd1874a0cecf83ed6d5fa37f2","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80b336b35ee7036cd1874a0cecf83ed6d5fa37f2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 13:44:42.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41d4768155711c3a95e5b13739e2fc3c1e0c25fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41d4768155711c3a95e5b13739e2fc3c1e0c25fa"}]},"branch":"refs/heads/master"},"6698f433ec39554c0ddb70105317ea323a209e2e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-21 15:39:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/635586/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/4"}}},"commit":{"parents":[{"commit":"393456f867386f7edbe5fa05a68380939b9b58d6","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/393456f867386f7edbe5fa05a68380939b9b58d6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 15:38:58.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6698f433ec39554c0ddb70105317ea323a209e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6698f433ec39554c0ddb70105317ea323a209e2e"}]},"branch":"refs/heads/master"},"87eeebf26d550c307ef396e5cb139c767d1b6cdb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-21 17:02:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/635586/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/5"}}},"commit":{"parents":[{"commit":"1f482a6d6419759d3d1d025fd81fe216342d9daf","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f482a6d6419759d3d1d025fd81fe216342d9daf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 17:01:50.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87eeebf26d550c307ef396e5cb139c767d1b6cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87eeebf26d550c307ef396e5cb139c767d1b6cdb"}]},"branch":"refs/heads/master"},"78d2fdfdbabd3a4904dbfbf53eaa5b02d084bdb7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-02-25 16:51:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/635586/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/6"}}},"commit":{"parents":[{"commit":"a91135179dc1ad2c3a657e19f7655207f8fdd0db","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a91135179dc1ad2c3a657e19f7655207f8fdd0db"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:15:09.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d2fdfdbabd3a4904dbfbf53eaa5b02d084bdb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d2fdfdbabd3a4904dbfbf53eaa5b02d084bdb7"}]},"branch":"refs/heads/master"},"15109ccb547e603c0aca2a73e43f4db523fa1bcd":{"kind":"REWORK","_number":7,"created":"2020-02-27 11:24:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/635586/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/635586/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/635586/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/635586/7"}}},"commit":{"parents":[{"commit":"9eb02542758c7ee6051456a59294375d6a16ffbb","subject":"Merge \"CI: set previous_release to train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eb02542758c7ee6051456a59294375d6a16ffbb"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 11:10:29.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15109ccb547e603c0aca2a73e43f4db523fa1bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15109ccb547e603c0aca2a73e43f4db523fa1bcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
