)]}'
{"id":"openstack%2Fkayobe~643572","triplet_id":"openstack%2Fkayobe~master~I05d5d6f5718b39dc95b40efe44e9ebb3fb5b1311","project":"openstack/kayobe","branch":"master","topic":"python3.6","hashtags":[],"change_id":"I05d5d6f5718b39dc95b40efe44e9ebb3fb5b1311","subject":"Add python 3.6 jobs","status":"MERGED","created":"2019-03-15 12:59:04.000000000","updated":"2019-04-26 17:57:36.000000000","submitted":"2019-04-26 17:57:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"643572-1556301456970-3a5dd909","meta_rev_id":"5f69370649bfbc48c8398a1e679c6a7b8f8b2246","_number":643572,"virtual_id_number":643572,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2019-03-18 13:30:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-04-26 17:57:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-04-26 14:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-03-15 16:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-04-26 14:13:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-15 16:39:10.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-04-26 14:13:27.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-04-26 17:57:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9465d1787e8840fff4f66f7add0c0c41669b5054","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-15 12:59:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd9a3c290150c4a2f28e9056786dd46464cf8cfd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-15 13:00:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a6553fdb37b46e00aa7ebcd8c5ed97f2abc347a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 15:21:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/643572/2/check/build-openstack-sphinx-docs/22387bb/html/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/72/643572/2/check/openstack-tox-docs/8af20cb/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/72/643572/2/check/openstack-tox-cover/3310325/cover/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/72/643572/2/check/openstack-tox-py35/865aed6/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/72/643572/2/check/openstack-tox-py36/b9a2c6d/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/72/643572/2/check/openstack-tox-pep8/2e14ac8/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/72/643572/2/check/openstack-tox-py27/3ec3504/ : SUCCESS in 5m 59s\n- build-openstack-releasenotes http://logs.openstack.org/72/643572/2/check/build-openstack-releasenotes/54cc248/html/ : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible-syntax/812f6a3/ : SUCCESS in 6m 02s\n- kayobe-tox-ansible http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible/f6eb808/ : SUCCESS in 9m 32s\n- kayobe-tox-molecule http://logs.openstack.org/72/643572/2/check/kayobe-tox-molecule/7599d7c/ : SUCCESS in 28m 00s\n- kayobe-overcloud-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-centos/4b3b19e/ : TIMED_OUT in 2h 01m 31s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-upgrade-centos/5cadaa6/ : SUCCESS in 1h 33m 45s\n- kayobe-seed-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-centos/f5d7700/ : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-upgrade-centos/b1580ca/ : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":2},{"id":"81d79ac2a0f35d0591b39f7276936097137f9c89","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-03-15 16:36:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d571fec4723dcd7dfd72c56e932a1583b6bc4d47","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-03-15 16:39:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eab846c8d642ac0d6c70c685402139995b93fa4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 18:03:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/643572/2/check/build-openstack-sphinx-docs/d7363d5/html/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/72/643572/2/check/openstack-tox-docs/597cadf/html/ : SUCCESS in 5m 08s\n- openstack-tox-cover http://logs.openstack.org/72/643572/2/check/openstack-tox-cover/13682d1/cover/ : SUCCESS in 5m 35s\n- openstack-tox-py35 http://logs.openstack.org/72/643572/2/check/openstack-tox-py35/7696e95/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/72/643572/2/check/openstack-tox-py36/340d069/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/72/643572/2/check/openstack-tox-pep8/3b7d9e7/ : SUCCESS in 8m 39s\n- openstack-tox-py27 http://logs.openstack.org/72/643572/2/check/openstack-tox-py27/aea224b/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/72/643572/2/check/build-openstack-releasenotes/09dd97d/html/ : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible-syntax/7035599/ : SUCCESS in 5m 49s\n- kayobe-tox-ansible http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible/256508a/ : SUCCESS in 9m 25s\n- kayobe-tox-molecule http://logs.openstack.org/72/643572/2/check/kayobe-tox-molecule/54f72a7/ : SUCCESS in 31m 11s\n- kayobe-overcloud-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-centos/b0d0f07/ : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-upgrade-centos/871d371/ : FAILURE in 1h 23m 33s\n- kayobe-seed-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-centos/7920c23/ : SUCCESS in 21m 10s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-upgrade-centos/3603ebe/ : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":2},{"id":"56fb1e04c5d435a38fd87dbc9ad2e2dcefe85451","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-18 09:46:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"706558cfda1634be52b4aafb84ee0aaf2fb87222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 11:13:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/643572/2/check/build-openstack-sphinx-docs/c2682b0/html/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/72/643572/2/check/openstack-tox-docs/86f6522/html/ : SUCCESS in 3m 57s\n- openstack-tox-cover http://logs.openstack.org/72/643572/2/check/openstack-tox-cover/b380d8b/cover/ : SUCCESS in 7m 34s\n- openstack-tox-py35 http://logs.openstack.org/72/643572/2/check/openstack-tox-py35/48ffa91/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/72/643572/2/check/openstack-tox-py36/99ac51a/ : SUCCESS in 7m 17s\n- openstack-tox-pep8 http://logs.openstack.org/72/643572/2/check/openstack-tox-pep8/1ae0716/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/72/643572/2/check/openstack-tox-py27/a09999b/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/72/643572/2/check/build-openstack-releasenotes/3f49b2b/html/ : SUCCESS in 5m 36s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible-syntax/22dee5f/ : SUCCESS in 7m 42s\n- kayobe-tox-ansible http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible/30e93c1/ : SUCCESS in 9m 08s\n- kayobe-tox-molecule http://logs.openstack.org/72/643572/2/check/kayobe-tox-molecule/6600b09/ : SUCCESS in 27m 01s\n- kayobe-overcloud-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-centos/fb82dbe/ : SUCCESS in 56m 49s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-upgrade-centos/f74e213/ : FAILURE in 1h 26m 00s\n- kayobe-seed-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-centos/9623765/ : SUCCESS in 19m 30s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-upgrade-centos/6dd0dab/ : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":2},{"id":"bb368fc5480956e2a88189ef5e8481a1b832cdda","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-18 11:17:29.000000000","message":"Patch Set 2:\n\nrecheck\n\nBare metal deploy in the upgrade job seems a bit unreliable.","accounts_in_message":[],"_revision_number":2},{"id":"2ea5ad77102ca759950a01d4830adb5af7b9e3f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 13:20:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/643572/2/check/build-openstack-sphinx-docs/ea49024/html/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/72/643572/2/check/openstack-tox-docs/d0b031f/html/ : SUCCESS in 7m 08s\n- openstack-tox-cover http://logs.openstack.org/72/643572/2/check/openstack-tox-cover/7b7dba7/cover/ : SUCCESS in 6m 19s\n- openstack-tox-py35 http://logs.openstack.org/72/643572/2/check/openstack-tox-py35/161af7b/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/72/643572/2/check/openstack-tox-py36/7584421/ : SUCCESS in 7m 42s\n- openstack-tox-pep8 http://logs.openstack.org/72/643572/2/check/openstack-tox-pep8/3696126/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/72/643572/2/check/openstack-tox-py27/219ad11/ : SUCCESS in 8m 22s\n- build-openstack-releasenotes http://logs.openstack.org/72/643572/2/check/build-openstack-releasenotes/11ebab3/html/ : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible-syntax/70ba0c0/ : SUCCESS in 5m 58s\n- kayobe-tox-ansible http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible/5c25bb4/ : SUCCESS in 9m 20s\n- kayobe-tox-molecule http://logs.openstack.org/72/643572/2/check/kayobe-tox-molecule/46d2fd9/ : SUCCESS in 29m 05s\n- kayobe-overcloud-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-centos/a887885/ : TIMED_OUT in 2h 01m 07s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-upgrade-centos/97942e7/ : SUCCESS in 1h 26m 42s\n- kayobe-seed-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-centos/3da9ed4/ : SUCCESS in 18m 20s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-upgrade-centos/a0f60e5/ : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9727ec7a083582875b120920916f130d52f019d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-18 13:30:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"427ee9e7e606886e2d860260fef6c77ab8f6b4e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 15:07:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/643572/2/check/build-openstack-sphinx-docs/6dfa54a/html/ : SUCCESS in 8m 01s\n- openstack-tox-docs http://logs.openstack.org/72/643572/2/check/openstack-tox-docs/cf4cf26/html/ : SUCCESS in 7m 34s\n- openstack-tox-cover http://logs.openstack.org/72/643572/2/check/openstack-tox-cover/6701a20/cover/ : SUCCESS in 9m 18s\n- openstack-tox-py35 http://logs.openstack.org/72/643572/2/check/openstack-tox-py35/a7772d3/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/72/643572/2/check/openstack-tox-py36/56d9697/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/72/643572/2/check/openstack-tox-pep8/a466396/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/72/643572/2/check/openstack-tox-py27/669d632/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/72/643572/2/check/build-openstack-releasenotes/152687e/html/ : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible-syntax/4b74600/ : SUCCESS in 5m 15s\n- kayobe-tox-ansible http://logs.openstack.org/72/643572/2/check/kayobe-tox-ansible/06e2332/ : SUCCESS in 9m 10s\n- kayobe-tox-molecule http://logs.openstack.org/72/643572/2/check/kayobe-tox-molecule/9fd2c13/ : SUCCESS in 32m 26s\n- kayobe-overcloud-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-centos/bd05e7b/ : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-overcloud-upgrade-centos/e75d876/ : SUCCESS in 1h 31m 33s\n- kayobe-seed-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-centos/03db1a2/ : SUCCESS in 22m 51s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/643572/2/check/kayobe-seed-upgrade-centos/86412ab/ : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1a8493cb5b9df3edbedb28404208afde0575ce56","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 14:13:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be679c55870a655525bade960fb4def9e92e003b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 14:13:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"655965b8f64895c5e807f9f6afb4defaabc3712b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:05:31.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b292bc08a9a34c73320985b6dc909327feb7444","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:57:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"df2b47677950727fd9583e70c1e3264340f89204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:57:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/643572/2/gate/build-openstack-sphinx-docs/0001331/html/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/72/643572/2/gate/openstack-tox-docs/228d2a6/html/ : SUCCESS in 7m 08s\n- openstack-tox-py35 http://logs.openstack.org/72/643572/2/gate/openstack-tox-py35/6cad51a/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/72/643572/2/gate/openstack-tox-py36/2287b80/ : SUCCESS in 7m 44s\n- openstack-tox-pep8 http://logs.openstack.org/72/643572/2/gate/openstack-tox-pep8/e28cf52/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/72/643572/2/gate/openstack-tox-py27/c207db5/ : SUCCESS in 8m 17s\n- build-openstack-releasenotes http://logs.openstack.org/72/643572/2/gate/build-openstack-releasenotes/4cc57da/html/ : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/643572/2/gate/kayobe-tox-ansible-syntax/5a6cd00/ : SUCCESS in 9m 25s\n- kayobe-tox-ansible http://logs.openstack.org/72/643572/2/gate/kayobe-tox-ansible/d619215/ : SUCCESS in 12m 25s\n- kayobe-tox-molecule http://logs.openstack.org/72/643572/2/gate/kayobe-tox-molecule/2df38e6/ : SUCCESS in 32m 56s\n- kayobe-overcloud-centos http://logs.openstack.org/72/643572/2/gate/kayobe-overcloud-centos/2df08d0/ : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/643572/2/gate/kayobe-overcloud-upgrade-centos/7995d03/ : SUCCESS in 1h 46m 50s\n- kayobe-seed-centos http://logs.openstack.org/72/643572/2/gate/kayobe-seed-centos/27e9d1b/ : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/643572/2/gate/kayobe-seed-upgrade-centos/e35759a/ : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da8059644c25bf99b60aa547bc82a16f35a17ad3","revisions":{"4676249b86d543404be56bacce4df3512d479d06":{"kind":"REWORK","_number":1,"created":"2019-03-15 12:59:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/643572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/643572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/643572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/643572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/643572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/643572/1"}}},"commit":{"parents":[{"commit":"7ae166851514f653e676e0a88fc446a7fd4706a9","subject":"Fix tox cover environment on python 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ae166851514f653e676e0a88fc446a7fd4706a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-15 12:58:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-15 12:58:17.000000000","tz":0},"subject":"Add python 3.6 jobs","message":"Add python 3.6 jobs\n\nAlso change to use the py3 tox environment rather than specifying a\nversion explicitly.\n\nChange-Id: I05d5d6f5718b39dc95b40efe44e9ebb3fb5b1311\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4676249b86d543404be56bacce4df3512d479d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4676249b86d543404be56bacce4df3512d479d06"}]},"branch":"refs/heads/master"},"da8059644c25bf99b60aa547bc82a16f35a17ad3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-03-15 13:00:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/643572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/643572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/643572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/643572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/643572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/643572/2"}}},"commit":{"parents":[{"commit":"14bfce8a726bdcadb8543ad9e652b535a33c609e","subject":"Fix tox cover environment on python 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bfce8a726bdcadb8543ad9e652b535a33c609e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-15 12:58:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-15 13:00:40.000000000","tz":0},"subject":"Add python 3.6 jobs","message":"Add python 3.6 jobs\n\nAlso change to use the py3 tox environment rather than specifying a\nversion explicitly.\n\nChange-Id: I05d5d6f5718b39dc95b40efe44e9ebb3fb5b1311\nStory: 2004959\nTask: 30035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da8059644c25bf99b60aa547bc82a16f35a17ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da8059644c25bf99b60aa547bc82a16f35a17ad3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
