)]}'
{"id":"openstack%2Fkayobe~647572","triplet_id":"openstack%2Fkayobe~master~I24564160b2e8f7c5feef7745fffb73de871018cb","project":"openstack/kayobe","branch":"master","topic":"fix-zuul-clone","hashtags":[],"change_id":"I24564160b2e8f7c5feef7745fffb73de871018cb","subject":"Fix git cloning previous release repos in CI","status":"ABANDONED","created":"2019-03-25 18:36:04.000000000","updated":"2019-06-17 18:28:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34d4a6718170911804b1ab907627de539fc44fef","_number":647572,"virtual_id_number":647572,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"date":"2019-05-22 08:38:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-05-21 10:00:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-22 08:23:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 13:34:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-05-21 10:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 08:23:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"d6775129880ca16792c9c743bb15cbe0fced7098","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-25 18:36:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b01bf5e1ded8528d8642b0747f8aa595ab17b451","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 20:32:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/647572/1/check/build-openstack-sphinx-docs/7b9ecc3/html/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/72/647572/1/check/openstack-tox-docs/96ac241/html/ : SUCCESS in 4m 51s\n- openstack-tox-cover http://logs.openstack.org/72/647572/1/check/openstack-tox-cover/177f230/cover/ : SUCCESS in 6m 29s\n- openstack-tox-py35 http://logs.openstack.org/72/647572/1/check/openstack-tox-py35/59040d6/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/72/647572/1/check/openstack-tox-pep8/58c6ce2/ : SUCCESS in 7m 37s\n- openstack-tox-py27 http://logs.openstack.org/72/647572/1/check/openstack-tox-py27/0d3d41e/ : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/647572/1/check/kayobe-tox-ansible-syntax/1cc8b80/ : SUCCESS in 7m 06s\n- kayobe-tox-ansible http://logs.openstack.org/72/647572/1/check/kayobe-tox-ansible/3b6ba9f/ : SUCCESS in 8m 34s\n- kayobe-tox-molecule http://logs.openstack.org/72/647572/1/check/kayobe-tox-molecule/a557a33/ : SUCCESS in 30m 33s\n- kayobe-overcloud-centos http://logs.openstack.org/72/647572/1/check/kayobe-overcloud-centos/3aff8f3/ : FAILURE in 1h 54m 58s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/647572/1/check/kayobe-overcloud-upgrade-centos/29e1d68/ : RETRY_LIMIT in 2m 58s\n- kayobe-seed-centos http://logs.openstack.org/72/647572/1/check/kayobe-seed-centos/379f878/ : SUCCESS in 20m 11s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/647572/1/check/kayobe-seed-upgrade-centos/1f33f3d/ : RETRY_LIMIT in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8a42a963a56cb0f91c4f9dc7f69475d9d91d96eb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-26 12:13:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"224ff1a6aecca4eb66dbcf719394d438c4f0fc7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 14:07:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/647572/2/check/build-openstack-sphinx-docs/c9d0d95/html/ : SUCCESS in 19m 12s\n- openstack-tox-docs http://logs.openstack.org/72/647572/2/check/openstack-tox-docs/04acb9c/html/ : SUCCESS in 4m 30s\n- openstack-tox-cover http://logs.openstack.org/72/647572/2/check/openstack-tox-cover/1d0d0a3/cover/ : SUCCESS in 13m 06s\n- openstack-tox-py35 http://logs.openstack.org/72/647572/2/check/openstack-tox-py35/6ff9eb3/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/72/647572/2/check/openstack-tox-pep8/f656fa1/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/72/647572/2/check/openstack-tox-py27/24df8e2/ : SUCCESS in 5m 20s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/647572/2/check/kayobe-tox-ansible-syntax/0d45c5b/ : SUCCESS in 5m 50s\n- kayobe-tox-ansible http://logs.openstack.org/72/647572/2/check/kayobe-tox-ansible/ecb0e63/ : SUCCESS in 9m 23s\n- kayobe-tox-molecule http://logs.openstack.org/72/647572/2/check/kayobe-tox-molecule/00a748d/ : SUCCESS in 35m 47s\n- kayobe-overcloud-centos http://logs.openstack.org/72/647572/2/check/kayobe-overcloud-centos/5a79800/ : SUCCESS in 58m 42s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/647572/2/check/kayobe-overcloud-upgrade-centos/58ddcdd/ : SUCCESS in 1h 45m 35s\n- kayobe-seed-centos http://logs.openstack.org/72/647572/2/check/kayobe-seed-centos/c053e29/ : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/647572/2/check/kayobe-seed-upgrade-centos/0207169/ : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8462cadcde4ae5310dd6139b5a3ee3d567d6b3b9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 13:34:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18561d94556ac93266f6c92a61578ca20b6e5f08","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-29 08:19:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24ec33a5f0f99faf9b9bcb934634dc7d3a46bfd7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 14:18:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"16208787251ac9ea7ca0f3b48dc115c272420f0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 16:50:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/647572/3/check/build-openstack-sphinx-docs/4ec8997/html/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/72/647572/3/check/openstack-tox-docs/323606b/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/72/647572/3/check/openstack-tox-cover/2fe9284/cover/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/72/647572/3/check/openstack-tox-py35/4c138f4/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/72/647572/3/check/openstack-tox-py36/ebeec65/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/72/647572/3/check/openstack-tox-pep8/0bf99ab/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/72/647572/3/check/openstack-tox-py27/52f00dc/ : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/647572/3/check/kayobe-tox-ansible-syntax/af43335/ : SUCCESS in 6m 08s\n- kayobe-tox-ansible http://logs.openstack.org/72/647572/3/check/kayobe-tox-ansible/bf1317a/ : SUCCESS in 9m 44s\n- kayobe-tox-molecule http://logs.openstack.org/72/647572/3/check/kayobe-tox-molecule/69f5cfb/ : SUCCESS in 34m 05s\n- kayobe-overcloud-centos http://logs.openstack.org/72/647572/3/check/kayobe-overcloud-centos/3e9cb0d/ : SUCCESS in 58m 58s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/647572/3/check/kayobe-overcloud-upgrade-centos/d7f5661/ : SUCCESS in 2h 01m 21s\n- kayobe-seed-centos http://logs.openstack.org/72/647572/3/check/kayobe-seed-centos/fb73415/ : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/647572/3/check/kayobe-seed-upgrade-centos/92664c0/ : RETRY_LIMIT in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b5b131e225226940464b1c3c1a38edf0100549aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 17:05:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"16dee1de5674c1f8e3748cdad7f539b4d30c0e5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:16:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/647572/4/check/build-openstack-sphinx-docs/c230d01/html/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/72/647572/4/check/openstack-tox-docs/077d8f4/html/ : SUCCESS in 4m 34s\n- openstack-tox-cover http://logs.openstack.org/72/647572/4/check/openstack-tox-cover/2fb0eb7/cover/ : SUCCESS in 6m 30s\n- openstack-tox-py35 http://logs.openstack.org/72/647572/4/check/openstack-tox-py35/2cbf49a/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/72/647572/4/check/openstack-tox-py36/10ea583/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/72/647572/4/check/openstack-tox-pep8/0d52982/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/72/647572/4/check/openstack-tox-py27/4742f68/ : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/647572/4/check/kayobe-tox-ansible-syntax/721b63b/ : SUCCESS in 7m 59s\n- kayobe-tox-ansible http://logs.openstack.org/72/647572/4/check/kayobe-tox-ansible/0a3b626/ : SUCCESS in 10m 16s\n- kayobe-tox-molecule http://logs.openstack.org/72/647572/4/check/kayobe-tox-molecule/64b05c6/ : SUCCESS in 28m 49s\n- kayobe-overcloud-centos http://logs.openstack.org/72/647572/4/check/kayobe-overcloud-centos/0db521f/ : SUCCESS in 1h 20m 51s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/647572/4/check/kayobe-overcloud-upgrade-centos/85883bc/ : SUCCESS in 2h 03m 02s\n- kayobe-seed-centos http://logs.openstack.org/72/647572/4/check/kayobe-seed-centos/6fab6b2/ : SUCCESS in 23m 23s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/647572/4/check/kayobe-seed-upgrade-centos/cbb37d8/ : FAILURE in 24m 06s","accounts_in_message":[],"_revision_number":4},{"id":"c9dd3ecb23968d3aecf19c0a4d2750d77fc30a38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 08:19:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"168340d10246beb27333bc150e3e45aa0a09d398","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:00:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/647572/4/check/build-openstack-sphinx-docs/2285041/html/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/72/647572/4/check/openstack-tox-docs/a25af00/html/ : SUCCESS in 6m 54s\n- openstack-tox-cover http://logs.openstack.org/72/647572/4/check/openstack-tox-cover/0312774/cover/ : SUCCESS in 6m 11s\n- openstack-tox-py35 http://logs.openstack.org/72/647572/4/check/openstack-tox-py35/1c73996/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/72/647572/4/check/openstack-tox-py36/b11e13c/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/72/647572/4/check/openstack-tox-pep8/516209f/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/72/647572/4/check/openstack-tox-py27/dc52c89/ : SUCCESS in 4m 39s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/647572/4/check/kayobe-tox-ansible-syntax/975fcc8/ : SUCCESS in 6m 09s\n- kayobe-tox-ansible http://logs.openstack.org/72/647572/4/check/kayobe-tox-ansible/138bb11/ : SUCCESS in 10m 03s\n- kayobe-tox-molecule http://logs.openstack.org/72/647572/4/check/kayobe-tox-molecule/4393426/ : SUCCESS in 32m 52s\n- kayobe-overcloud-centos http://logs.openstack.org/72/647572/4/check/kayobe-overcloud-centos/5b7f6a0/ : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/647572/4/check/kayobe-overcloud-upgrade-centos/fe6a004/ : SUCCESS in 1h 35m 59s\n- kayobe-seed-centos http://logs.openstack.org/72/647572/4/check/kayobe-seed-centos/f1811dd/ : SUCCESS in 23m 34s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/647572/4/check/kayobe-seed-upgrade-centos/1462030/ : SUCCESS in 24m 18s","accounts_in_message":[],"_revision_number":4},{"id":"361328fcb0623f40e2dfad44697eea02b9920ea4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-22 08:23:32.000000000","message":"Patch Set 4: Code-Review+2\n\nI assume you\u0027ve tested the Zuul magic that allows this?","accounts_in_message":[],"_revision_number":4},{"id":"74b372278db6eebc8dff485f448d88df1541e009","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-22 08:38:46.000000000","message":"Patch Set 4:\n\n\u003e I assume you\u0027ve tested the Zuul magic that allows this?\n\nYes, I used it in kolla-ansible upgrade jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f3e0348b4269a22d51e07c575df3cb4064383a3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 18:28:14.000000000","message":"Abandoned\n\nCombining with cliff patch to allow it to merge: https://review.opendev.org/#/c/665388/","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f15d34b246b5cb3ab371296b20cbed9244153d73","revisions":{"2783718b4a9689d5c163c37a8e0137af6a5ba05b":{"kind":"REWORK","_number":1,"created":"2019-03-25 18:36:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/647572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/647572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/647572/1"}}},"commit":{"parents":[{"commit":"81e3c8a282901f89cd2138f6a9a658925e8f283a","subject":"Add missing space in argument description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81e3c8a282901f89cd2138f6a9a658925e8f283a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-25 18:33:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-25 18:33:13.000000000","tz":0},"subject":"Fix git cloning previous release repos in CI","message":"Fix git cloning previous release repos in CI\n\nZuul provides repositories with all branches containing all prospective\nchanges, not just the current branch. We can use this in the upgrade\njobs to include changes to the previous release branch when a patch\ndepends on them.\n\nChange-Id: I24564160b2e8f7c5feef7745fffb73de871018cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2783718b4a9689d5c163c37a8e0137af6a5ba05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2783718b4a9689d5c163c37a8e0137af6a5ba05b"}]},"branch":"refs/heads/master"},"3e2327d66cec2c1706b24483ec44184cc2c23b28":{"kind":"REWORK","_number":2,"created":"2019-03-26 12:13:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/647572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/647572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/647572/2"}}},"commit":{"parents":[{"commit":"81e3c8a282901f89cd2138f6a9a658925e8f283a","subject":"Add missing space in argument description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81e3c8a282901f89cd2138f6a9a658925e8f283a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-25 18:33:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-26 12:12:40.000000000","tz":0},"subject":"Fix git cloning previous release repos in CI","message":"Fix git cloning previous release repos in CI\n\nZuul provides repositories with all branches containing all prospective\nchanges, not just the current branch. We can use this in the upgrade\njobs to include changes to the previous release branch when a patch\ndepends on them.\n\nChange-Id: I24564160b2e8f7c5feef7745fffb73de871018cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e2327d66cec2c1706b24483ec44184cc2c23b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e2327d66cec2c1706b24483ec44184cc2c23b28"}]},"branch":"refs/heads/master"},"b26474256bfb3e17e15a97a588924c2dbaccbcb0":{"kind":"REWORK","_number":3,"created":"2019-05-20 14:18:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/647572/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/647572/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/647572/3"}}},"commit":{"parents":[{"commit":"c020751a5d32bb772246f8e8546c625804614188","subject":"Merge \"Support for Ceph and Swift storage networks, and improvements to Swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c020751a5d32bb772246f8e8546c625804614188"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-25 18:33:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 14:18:29.000000000","tz":60},"subject":"Fix git cloning previous release repos in CI","message":"Fix git cloning previous release repos in CI\n\nZuul provides repositories with all branches containing all prospective\nchanges, not just the current branch. We can use this in the upgrade\njobs to include changes to the previous release branch when a patch\ndepends on them.\n\nChange-Id: I24564160b2e8f7c5feef7745fffb73de871018cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b26474256bfb3e17e15a97a588924c2dbaccbcb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b26474256bfb3e17e15a97a588924c2dbaccbcb0"}]},"branch":"refs/heads/master"},"f15d34b246b5cb3ab371296b20cbed9244153d73":{"kind":"REWORK","_number":4,"created":"2019-05-20 17:05:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/647572/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/647572/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/647572/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/647572/4"}}},"commit":{"parents":[{"commit":"ca3b5a770103cc6285f2a9d4d1a3117eff3d1788","subject":"DRY out Zuul job configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca3b5a770103cc6285f2a9d4d1a3117eff3d1788"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-25 18:33:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 17:04:23.000000000","tz":60},"subject":"Fix git cloning previous release repos in CI","message":"Fix git cloning previous release repos in CI\n\nZuul provides repositories with all branches containing all prospective\nchanges, not just the current branch. We can use this in the upgrade\njobs to include changes to the previous release branch when a patch\ndepends on them.\n\nChange-Id: I24564160b2e8f7c5feef7745fffb73de871018cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f15d34b246b5cb3ab371296b20cbed9244153d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f15d34b246b5cb3ab371296b20cbed9244153d73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
