)]}'
{"id":"openstack%2Fkayobe~649595","triplet_id":"openstack%2Fkayobe~master~Ie40f258a31d4b72cd9867d321a3ff712522daec5","project":"openstack/kayobe","branch":"master","topic":"update-dev-doc","hashtags":[],"change_id":"Ie40f258a31d4b72cd9867d321a3ff712522daec5","subject":"Small update to automated development docs","status":"MERGED","created":"2019-04-03 12:32:55.000000000","updated":"2019-05-14 11:33:02.000000000","submitted":"2019-05-14 11:33:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649595-1557833582406-7c1c31b6","meta_rev_id":"91c966b7dcac8c64cad673b40b8af443e56e0bf4","_number":649595,"virtual_id_number":649595,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-14 11:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-05-14 11:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 11:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-05-14 11:22:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 11:22:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-05-14 11:22:53.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-05-14 11:33:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a7c1c7ed9ea7be98f02c68714b84fc17c54a676","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-03 12:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3fc1b33705cbb381e0d80b92391c2a4b4f298ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 12:43:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/95/649595/1/check/build-openstack-sphinx-docs/b77f6cb/html/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/95/649595/1/check/openstack-tox-docs/18f509d/html/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/95/649595/1/check/openstack-tox-pep8/2997d0e/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b2235a003efa708914b7959c63c69b4936e9f535","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 13:26:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"170d06e811de50ac5b3c1da5c426456aee0a0a80","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-14 06:57:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0b82ad4068a19619e40cb28d1a08b119a158419","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-14 11:10:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0be69aea2f52a4ebed5b5e7f50313c557a9072c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-14 11:10:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3099ac0d46e0725724bcbfff6d859a0a83a36de4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:20:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/95/649595/2/check/build-openstack-sphinx-docs/735df9d/html/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/95/649595/2/check/openstack-tox-docs/813553b/html/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/95/649595/2/check/openstack-tox-pep8/b7e00d2/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2730e9db30fa1e98b47bc3ce604479c341a7381e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-14 11:22:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25b9437098382715877c6ffa0c3bc043d4498592","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-05-14 11:22:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac301c8e869339598a2916c41edd8e193c9b5a77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:23:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09075d8143f0c1b522fab205f177b8c72e979b24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:33:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f19cc0ef3ef408c5ce12b8a35ad06b399dec8e4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:33:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/95/649595/2/gate/build-openstack-sphinx-docs/cda9932/html/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/95/649595/2/gate/openstack-tox-docs/09a201e/html/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/95/649595/2/gate/openstack-tox-pep8/bff63d8/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bec3fe38327a231256c8cce353246d8bc492f40","revisions":{"6b3e8b4d9b7f48f096fc1498c2c60eba630a4f4e":{"kind":"REWORK","_number":1,"created":"2019-04-03 12:32:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/95/649595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/649595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/649595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/649595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/649595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/649595/1"}}},"commit":{"parents":[{"commit":"44430a42421a7c59807a58cbbd67f90a154fd30a","subject":"Support custom keepalived config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44430a42421a7c59807a58cbbd67f90a154fd30a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-03 12:32:13.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-03 12:32:13.000000000","tz":60},"subject":"Small update to automated development docs","message":"Small update to automated development docs\n\nInclude information on how to test bare metal instance creation.\n\nChange-Id: Ie40f258a31d4b72cd9867d321a3ff712522daec5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b3e8b4d9b7f48f096fc1498c2c60eba630a4f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b3e8b4d9b7f48f096fc1498c2c60eba630a4f4e"}]},"branch":"refs/heads/master"},"7bec3fe38327a231256c8cce353246d8bc492f40":{"kind":"REWORK","_number":2,"created":"2019-05-14 11:10:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/95/649595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/649595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/649595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/649595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/649595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/649595/2"}}},"commit":{"parents":[{"commit":"05cdbeb270775919828735046ad23041ccdea783","subject":"Merge \"Use x/kayobe for storyboard link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05cdbeb270775919828735046ad23041ccdea783"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-03 12:32:13.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-14 11:10:13.000000000","tz":60},"subject":"Small update to automated development docs","message":"Small update to automated development docs\n\nInclude information on how to test bare metal instance creation.\n\nChange-Id: Ie40f258a31d4b72cd9867d321a3ff712522daec5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bec3fe38327a231256c8cce353246d8bc492f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bec3fe38327a231256c8cce353246d8bc492f40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
