)]}'
{"id":"openstack%2Fkayobe~653798","triplet_id":"openstack%2Fkayobe~master~I7a297236d184967fcb48187a64e366617b689e9f","project":"openstack/kayobe","branch":"master","topic":"kenv","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-27 00:18:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7a297236d184967fcb48187a64e366617b689e9f","subject":"WIP: Provide a mechanism for virtual kayobe enviroments","status":"NEW","created":"2019-04-18 15:45:14.000000000","updated":"2023-08-27 00:18:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a729395c807ea67a3c9ad5596aabca0d23ad99bb","_number":653798,"virtual_id_number":653798,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2019-04-24 19:16:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 19:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-27 00:18:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"7d128c2de475508f0efd035814ea327e218b3841","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-18 15:45:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e9decd07319946246ff84f333224c3cc9e73f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 17:01:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/98/653798/1/check/build-openstack-sphinx-docs/7e78f50/html/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/98/653798/1/check/openstack-tox-docs/04cbdfb/html/ : SUCCESS in 5m 49s\n- openstack-tox-cover http://logs.openstack.org/98/653798/1/check/openstack-tox-cover/65124c7/cover/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/98/653798/1/check/openstack-tox-py35/9c5ac75/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/98/653798/1/check/openstack-tox-pep8/667f138/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/98/653798/1/check/openstack-tox-py27/a854a10/ : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/98/653798/1/check/kayobe-tox-ansible-syntax/72265ca/ : SUCCESS in 5m 30s\n- kayobe-tox-ansible http://logs.openstack.org/98/653798/1/check/kayobe-tox-ansible/729e0bf/ : SUCCESS in 8m 59s\n- kayobe-tox-molecule http://logs.openstack.org/98/653798/1/check/kayobe-tox-molecule/ff68207/ : SUCCESS in 32m 41s\n- kayobe-overcloud-centos http://logs.openstack.org/98/653798/1/check/kayobe-overcloud-centos/460dec9/ : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/98/653798/1/check/kayobe-overcloud-upgrade-centos/963c649/ : FAILURE in 30m 46s\n- kayobe-seed-centos http://logs.openstack.org/98/653798/1/check/kayobe-seed-centos/4844c48/ : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/98/653798/1/check/kayobe-seed-upgrade-centos/c5e447d/ : FAILURE in 9m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1c74ad3804fe5cec305c2d48a7353b4f31e847f5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-23 09:09:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54390b1047fd2cec6579896c7c6e30ce40566faf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-23 09:28:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e5f39a58f721f7dee637e1e1320df1e9bd908832","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:15:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/98/653798/3/check/build-openstack-sphinx-docs/c4e4d4a/html/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/98/653798/3/check/openstack-tox-docs/be04dcb/html/ : SUCCESS in 4m 17s\n- openstack-tox-cover http://logs.openstack.org/98/653798/3/check/openstack-tox-cover/6d5bb1a/cover/ : SUCCESS in 5m 31s\n- openstack-tox-py35 http://logs.openstack.org/98/653798/3/check/openstack-tox-py35/dfed5f3/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/98/653798/3/check/openstack-tox-pep8/3da24ce/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/98/653798/3/check/openstack-tox-py27/9ac8766/ : SUCCESS in 5m 15s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/98/653798/3/check/kayobe-tox-ansible-syntax/066c18c/ : SUCCESS in 6m 15s\n- kayobe-tox-ansible http://logs.openstack.org/98/653798/3/check/kayobe-tox-ansible/8db6628/ : SUCCESS in 13m 07s\n- kayobe-tox-molecule http://logs.openstack.org/98/653798/3/check/kayobe-tox-molecule/75823f1/ : SUCCESS in 31m 59s\n- kayobe-overcloud-centos http://logs.openstack.org/98/653798/3/check/kayobe-overcloud-centos/ce1ff37/ : RETRY_LIMIT in 2m 23s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/98/653798/3/check/kayobe-overcloud-upgrade-centos/78cb543/ : FAILURE in 33m 36s\n- kayobe-seed-centos http://logs.openstack.org/98/653798/3/check/kayobe-seed-centos/dfb5745/ : RETRY_LIMIT in 3m 57s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/98/653798/3/check/kayobe-seed-upgrade-centos/9c6c7da/ : FAILURE in 24m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c5029cadc70d608a7ad83910291358a1314e1070","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-23 10:45:22.000000000","message":"Abandoned\n\nProbably better solved with something like virtualenvwrapper","accounts_in_message":[],"_revision_number":3},{"id":"553a226d904e17fc0be2233c43d52e2e95b37395","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-24 14:25:28.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"d1d6d6d3765967fea925396804ce2e334e9014ab","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-24 14:26:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"150b859a3b1cb0acf91f43fa518cc2afc8f42e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 16:38:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/98/653798/4/check/test-release-openstack/6617716/ : SUCCESS in 3m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/653798/4/check/build-openstack-sphinx-docs/5ca83d6/html/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/98/653798/4/check/openstack-tox-docs/d34a230/html/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/98/653798/4/check/openstack-tox-cover/bdcfdb1/cover/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/98/653798/4/check/openstack-tox-py35/177a1dc/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/98/653798/4/check/openstack-tox-pep8/ce1882b/ : FAILURE in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/98/653798/4/check/openstack-tox-py27/2216297/ : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/98/653798/4/check/kayobe-tox-ansible-syntax/2113186/ : SUCCESS in 5m 36s\n- kayobe-tox-ansible http://logs.openstack.org/98/653798/4/check/kayobe-tox-ansible/23284df/ : SUCCESS in 9m 11s\n- kayobe-tox-molecule http://logs.openstack.org/98/653798/4/check/kayobe-tox-molecule/8dcf46b/ : SUCCESS in 31m 27s\n- kayobe-overcloud-centos http://logs.openstack.org/98/653798/4/check/kayobe-overcloud-centos/23451e0/ : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/98/653798/4/check/kayobe-overcloud-upgrade-centos/f2c3ec1/ : SUCCESS in 2h 01m 53s\n- kayobe-seed-centos http://logs.openstack.org/98/653798/4/check/kayobe-seed-centos/50ca764/ : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/98/653798/4/check/kayobe-seed-upgrade-centos/acdbd57/ : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":4},{"id":"b36ad3a9bebe54dc8fbf92a94b27261435b8c3f1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-24 17:11:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f61bb053f85c643759cf3ba8b6066d92dc41320f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:16:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/653798/5/check/test-release-openstack/781ad68/ : SUCCESS in 4m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/653798/5/check/build-openstack-sphinx-docs/e02cf52/html/ : SUCCESS in 8m 02s\n- openstack-tox-docs http://logs.openstack.org/98/653798/5/check/openstack-tox-docs/7d6980b/html/ : SUCCESS in 4m 00s\n- openstack-tox-cover http://logs.openstack.org/98/653798/5/check/openstack-tox-cover/a45a311/cover/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/98/653798/5/check/openstack-tox-py35/c251d82/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/98/653798/5/check/openstack-tox-pep8/d00b936/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/98/653798/5/check/openstack-tox-py27/454d834/ : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/98/653798/5/check/kayobe-tox-ansible-syntax/4eb7a13/ : SUCCESS in 5m 16s\n- kayobe-tox-ansible http://logs.openstack.org/98/653798/5/check/kayobe-tox-ansible/54cdb1d/ : SUCCESS in 10m 06s\n- kayobe-tox-molecule http://logs.openstack.org/98/653798/5/check/kayobe-tox-molecule/b6d01dd/ : SUCCESS in 34m 49s\n- kayobe-overcloud-centos http://logs.openstack.org/98/653798/5/check/kayobe-overcloud-centos/d3d093e/ : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/98/653798/5/check/kayobe-overcloud-upgrade-centos/e8f51a2/ : SUCCESS in 1h 57m 32s\n- kayobe-seed-centos http://logs.openstack.org/98/653798/5/check/kayobe-seed-centos/d96f8fd/ : SUCCESS in 27m 53s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/98/653798/5/check/kayobe-seed-upgrade-centos/5f9e346/ : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dbff7aaa71f3d32f2fbcb971618fcfb72e69c0aa","revisions":{"ad58f60287111e6a7067daec69636151ff2c5762":{"kind":"REWORK","_number":1,"created":"2019-04-18 15:45:14.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/98/653798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/653798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/653798/1"}}},"commit":{"parents":[{"commit":"272709d54aff0ff731feb227ea51627c18ede423","subject":"Merge \"Add openstack_release and openstack_branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/272709d54aff0ff731feb227ea51627c18ede423"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-18 15:35:01.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-18 15:35:01.000000000","tz":60},"subject":"WIP: Provide a mechanism for virtual kayobe enviroments","message":"WIP: Provide a mechanism for virtual kayobe enviroments\n\nWhen you pip install your config with the -e flag we can detect the config\nsource location. This reduces the number of steps needed to use kayobe i.e you\ndon\u0027t need to set KAYOBE_CONFIG_PATH and allows you to switch kayobe\nenvironments by simply changing virtualenvs.\n\nChange-Id: I7a297236d184967fcb48187a64e366617b689e9f\nStory: 2004367\nTask: 30587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad58f60287111e6a7067daec69636151ff2c5762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad58f60287111e6a7067daec69636151ff2c5762"}]},"branch":"refs/heads/master"},"0f4f9fc5f964ace3669eda055cd7022e76ed444b":{"kind":"REWORK","_number":2,"created":"2019-04-23 09:09:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/98/653798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/653798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/653798/2"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-18 15:35:01.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-23 09:11:53.000000000","tz":60},"subject":"WIP: Provide a mechanism for virtual kayobe enviroments","message":"WIP: Provide a mechanism for virtual kayobe enviroments\n\nWhen you pip install your config with the -e flag we can detect the config\nsource location. This reduces the number of steps needed to use kayobe i.e you\ndon\u0027t need to set KAYOBE_CONFIG_PATH and allows you to switch kayobe\nenvironments by simply changing virtualenvs.\n\nChange-Id: I7a297236d184967fcb48187a64e366617b689e9f\nStory: 2004367\nTask: 30587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f4f9fc5f964ace3669eda055cd7022e76ed444b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f4f9fc5f964ace3669eda055cd7022e76ed444b"}]},"branch":"refs/heads/master"},"5f7bd290b406e46f80ea4f2985c510001e37a501":{"kind":"REWORK","_number":3,"created":"2019-04-23 09:28:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/98/653798/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/653798/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/653798/3"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-18 15:35:01.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-23 09:32:25.000000000","tz":60},"subject":"Provide a mechanism for virtual kayobe enviroments","message":"Provide a mechanism for virtual kayobe enviroments\n\nWhen you pip install your config with the -e flag we can detect the config\nsource location. This reduces the number of steps needed to use kayobe i.e you\ndon\u0027t need to set KAYOBE_CONFIG_PATH and allows you to switch kayobe\nenvironments by simply changing virtualenvs.\n\nChange-Id: I7a297236d184967fcb48187a64e366617b689e9f\nStory: 2004367\nTask: 30587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f7bd290b406e46f80ea4f2985c510001e37a501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f7bd290b406e46f80ea4f2985c510001e37a501"}]},"branch":"refs/heads/master"},"6764d1d95380c1e8f4a6e9d7e548acbb5873259a":{"kind":"REWORK","_number":4,"created":"2019-04-24 14:26:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/98/653798/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/653798/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/653798/4"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-18 15:35:01.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-24 14:22:05.000000000","tz":60},"subject":"WIP: Provide a mechanism for virtual kayobe enviroments","message":"WIP: Provide a mechanism for virtual kayobe enviroments\n\nThis adds an env-path option to kayobe. Kayobe will source this file\ninto its enviroment prior to running any commands.\n\nWhen you pip install your config with the -e flag we can detect the config\nsource location. This reduces the number of steps needed to use kayobe i.e you\ndon\u0027t need to set KAYOBE_CONFIG_PATH and allows you to switch kayobe\nenvironments by simply changing virtualenvs.\n\nChange-Id: I7a297236d184967fcb48187a64e366617b689e9f\nDepends-On: https://review.opendev.org/655424\nStory: 2004367\nTask: 30587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6764d1d95380c1e8f4a6e9d7e548acbb5873259a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6764d1d95380c1e8f4a6e9d7e548acbb5873259a"}]},"branch":"refs/heads/master"},"dbff7aaa71f3d32f2fbcb971618fcfb72e69c0aa":{"kind":"REWORK","_number":5,"created":"2019-04-24 17:11:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/98/653798/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/653798/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/653798/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/653798/5"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-18 15:35:01.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-24 17:06:27.000000000","tz":60},"subject":"WIP: Provide a mechanism for virtual kayobe enviroments","message":"WIP: Provide a mechanism for virtual kayobe enviroments\n\nThis adds an env-path option to kayobe. Kayobe will source this file\ninto its enviroment prior to running any commands.\n\nWhen you pip install your config with the -e flag we can detect the config\nsource location. This reduces the number of steps needed to use kayobe i.e you\ndon\u0027t need to set KAYOBE_CONFIG_PATH and allows you to switch kayobe\nenvironments by simply changing virtualenvs.\n\nChange-Id: I7a297236d184967fcb48187a64e366617b689e9f\nDepends-On: https://review.opendev.org/655424\nStory: 2004367\nTask: 30587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbff7aaa71f3d32f2fbcb971618fcfb72e69c0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbff7aaa71f3d32f2fbcb971618fcfb72e69c0aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
