)]}'
{"id":"openstack%2Fkayobe~655669","triplet_id":"openstack%2Fkayobe~master~I9fe28a658dc35adb5358d9a3a96bda9245404257","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I9fe28a658dc35adb5358d9a3a96bda9245404257","subject":"Support customising Prometheus config","status":"MERGED","created":"2019-04-25 11:00:36.000000000","updated":"2019-07-08 16:14:15.000000000","submitted":"2019-07-08 16:14:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655669-1562602455406-29425bd4","meta_rev_id":"d7a6749d413f444a843d3f2b89f4703015643069","_number":655669,"virtual_id_number":655669,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"date":"2019-06-17 16:52:19.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-08 16:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-06-17 16:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-17 17:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-07-08 14:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 04:23:17.000000000","updated_by":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"reviewer":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"state":"REVIEWER"},{"updated":"2019-06-17 17:20:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-08 14:16:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-08 16:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec061a043e9be7baf89f1811901db1554dcd00c1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-25 11:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92e066668afce2c12142ca8d1fe9d660f9f0deeb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 11:12:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8ee2dead17ce9e34c93aa78d7f30e751cef1606","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-25 11:19:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6249032a3c52ae6635e0731cfe9e70362007003c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-25 11:21:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44f999355bd24bafdefb50b1f7ab166e36a2913b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 11:28:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49a98f2da107acd1202a30d7c0fe79f1fa1ed785","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-25 12:09:33.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"0dd9f003ad24516be232be709278c8c8eeb80af6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:17:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/69/655669/2/check/build-openstack-sphinx-docs/7a2d631/html/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/69/655669/2/check/openstack-tox-docs/4e9270c/html/ : SUCCESS in 4m 25s\n- openstack-tox-cover http://logs.openstack.org/69/655669/2/check/openstack-tox-cover/99aefcb/cover/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/69/655669/2/check/openstack-tox-py35/c77f8d7/ : SUCCESS in 10m 30s\n- openstack-tox-pep8 http://logs.openstack.org/69/655669/2/check/openstack-tox-pep8/077704c/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/69/655669/2/check/openstack-tox-py27/a5b7638/ : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/655669/2/check/kayobe-tox-ansible-syntax/641ef6d/ : SUCCESS in 5m 29s\n- kayobe-tox-ansible http://logs.openstack.org/69/655669/2/check/kayobe-tox-ansible/c77d8c8/ : SUCCESS in 8m 58s\n- kayobe-tox-molecule http://logs.openstack.org/69/655669/2/check/kayobe-tox-molecule/19b8334/ : FAILURE in 14m 47s\n- kayobe-overcloud-centos http://logs.openstack.org/69/655669/2/check/kayobe-overcloud-centos/24bdc6f/ : RETRY_LIMIT in 2m 18s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/655669/2/check/kayobe-overcloud-upgrade-centos/162ff4d/ : FAILURE in 48m 36s\n- kayobe-seed-centos http://logs.openstack.org/69/655669/2/check/kayobe-seed-centos/fc88ecf/ : RETRY_LIMIT in 2m 22s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/655669/2/check/kayobe-seed-upgrade-centos/a475666/ : FAILURE in 24m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d24115ffc25fc033479337e5b779bb0cbe6d0e89","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-25 13:44:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e7b28939c72f4306fd66430ab905454765b2b2b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:41:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/69/655669/3/check/build-openstack-sphinx-docs/30779fc/html/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/69/655669/3/check/openstack-tox-docs/77a3165/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/69/655669/3/check/openstack-tox-cover/2f1ca62/cover/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/69/655669/3/check/openstack-tox-py35/fa6dc92/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/69/655669/3/check/openstack-tox-pep8/ca8ca0e/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/69/655669/3/check/openstack-tox-py27/0937952/ : SUCCESS in 4m 40s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/655669/3/check/kayobe-tox-ansible-syntax/1a8c0ec/ : SUCCESS in 5m 24s\n- kayobe-tox-ansible http://logs.openstack.org/69/655669/3/check/kayobe-tox-ansible/e7a03df/ : SUCCESS in 8m 21s\n- kayobe-tox-molecule http://logs.openstack.org/69/655669/3/check/kayobe-tox-molecule/fe114c4/ : SUCCESS in 31m 56s\n- kayobe-overcloud-centos http://logs.openstack.org/69/655669/3/check/kayobe-overcloud-centos/80fb504/ : RETRY_LIMIT in 2m 22s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/655669/3/check/kayobe-overcloud-upgrade-centos/78eacc4/ : FAILURE in 47m 54s\n- kayobe-seed-centos http://logs.openstack.org/69/655669/3/check/kayobe-seed-centos/5acb84f/ : RETRY_LIMIT in 2m 12s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/655669/3/check/kayobe-seed-upgrade-centos/7bf087b/ : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":3},{"id":"66966c3deca5fb963072b31e96c7c0241f3e3f7a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-17 13:30:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b75268e26a517845a792ea69414dcd4d43d0caa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 15:25:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/69/655669/3/check/build-openstack-sphinx-docs/179a592/html/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/69/655669/3/check/openstack-tox-docs/70c03bb/html/ : SUCCESS in 4m 41s\n- openstack-tox-cover http://logs.openstack.org/69/655669/3/check/openstack-tox-cover/1bfd800/cover/ : SUCCESS in 6m 08s\n- openstack-tox-py35 http://logs.openstack.org/69/655669/3/check/openstack-tox-py35/7f5153a/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/69/655669/3/check/openstack-tox-py36/9c5f32b/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/69/655669/3/check/openstack-tox-pep8/01409dd/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/69/655669/3/check/openstack-tox-py27/875bc55/ : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/655669/3/check/kayobe-tox-ansible-syntax/16b47a8/ : SUCCESS in 5m 37s\n- kayobe-tox-ansible http://logs.openstack.org/69/655669/3/check/kayobe-tox-ansible/0c13079/ : SUCCESS in 10m 39s\n- kayobe-tox-molecule http://logs.openstack.org/69/655669/3/check/kayobe-tox-molecule/38751b2/ : SUCCESS in 31m 47s\n- kayobe-overcloud-centos http://logs.openstack.org/69/655669/3/check/kayobe-overcloud-centos/1c990b5/ : SUCCESS in 1h 14m 04s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/655669/3/check/kayobe-overcloud-upgrade-centos/1cbdbda/ : SUCCESS in 1h 52m 45s\n- kayobe-seed-centos http://logs.openstack.org/69/655669/3/check/kayobe-seed-centos/aed09b7/ : SUCCESS in 17m 29s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/655669/3/check/kayobe-seed-upgrade-centos/f508dac/ : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":3},{"id":"91bfd83a422dfcc0cc0e31c3ab6dcfc67e59975a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-17 15:27:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6bef486cb120b7445c7f1b147d8ec1bb487e30f9","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-18 04:23:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5ffd43bdea47a9df4e3ca70ff5ef4bb773a52fcb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-22 16:32:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ce48f41d4513b487a7d490073f5e989ae44f2439","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-10 14:31:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"97e045fc7276da6ac94f2ec2701f294fe0295dff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-10 14:43:12.000000000","message":"Patch Set 4:\n\nI noticed that Kolla Ansible config tasks also create configuration files based on the content of \"{{ node_custom_config }}/prometheus-mysqld-exporter/\" and \"{{ node_custom_config }}/prometheus-openstack-exporter/\". Do we want to support these as well?","accounts_in_message":[],"_revision_number":4},{"id":"a7e1c32c56d34c7c70aa75a350f4c991c74f0a52","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-10 15:11:20.000000000","message":"Patch Set 4:\n\nHmm, yeah, fair point. Fortunately there aren\u0027t many of them which use config files.","accounts_in_message":[],"_revision_number":4},{"id":"644075e568770ffe6987a6197a20a880ba1441b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:29:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/69/655669/4/check/openstack-tox-docs/a1f6370/html/ : SUCCESS in 4m 27s\n- openstack-tox-cover http://logs.openstack.org/69/655669/4/check/openstack-tox-cover/1e3c6a8/cover/ : SUCCESS in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/69/655669/4/check/openstack-tox-py35/54024e0/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/69/655669/4/check/openstack-tox-py36/5411c3b/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/69/655669/4/check/openstack-tox-pep8/938d19d/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/69/655669/4/check/openstack-tox-py27/5978969/ : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/655669/4/check/kayobe-tox-ansible-syntax/706efe3/ : SUCCESS in 5m 00s\n- kayobe-tox-ansible http://logs.openstack.org/69/655669/4/check/kayobe-tox-ansible/81a6cf1/ : SUCCESS in 8m 55s\n- kayobe-tox-molecule http://logs.openstack.org/69/655669/4/check/kayobe-tox-molecule/1d77a7f/ : FAILURE in 6m 41s\n- kayobe-overcloud-centos http://logs.openstack.org/69/655669/4/check/kayobe-overcloud-centos/868735c/ : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/655669/4/check/kayobe-overcloud-upgrade-centos/1317382/ : SUCCESS in 1h 52m 22s\n- kayobe-seed-centos http://logs.openstack.org/69/655669/4/check/kayobe-seed-centos/104d317/ : SUCCESS in 21m 16s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/655669/4/check/kayobe-seed-upgrade-centos/1551c9e/ : FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":4},{"id":"13a5c04a56226511aaee4d3c810a031543f10a14","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 16:44:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4e09e08f563b8b221ecf0b0567a6ae0d1f150844","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-17 16:52:19.000000000","message":"Patch Set 4:\n\nThanks for the reviews. A patch for the other services can be added in a different commit.","accounts_in_message":[],"_revision_number":4},{"id":"97b9fb9966b0e39b0b8b3aec8e8a720329214a95","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-17 17:20:22.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e Thanks for the reviews. A patch for the other services can be added\n \u003e in a different commit.\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"a58b4ec6b1a0aa7057e53b5ffb5cd4e8a335866c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 11:53:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1d2cebb556426e8b98ce773d99ddd85523487919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:51:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/655669/4/check/openstack-tox-docs/a63d094/html/ : SUCCESS in 4m 27s\n- openstack-tox-cover http://logs.openstack.org/69/655669/4/check/openstack-tox-cover/3cd8d14/cover/ : SUCCESS in 6m 05s\n- openstack-tox-py35 http://logs.openstack.org/69/655669/4/check/openstack-tox-py35/d8e8118/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/69/655669/4/check/openstack-tox-py36/f17a8bc/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/69/655669/4/check/openstack-tox-pep8/49cce1a/ : SUCCESS in 9m 27s\n- openstack-tox-py27 http://logs.openstack.org/69/655669/4/check/openstack-tox-py27/0422888/ : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/655669/4/check/kayobe-tox-ansible-syntax/5cdcfba/ : SUCCESS in 5m 56s\n- kayobe-tox-ansible http://logs.openstack.org/69/655669/4/check/kayobe-tox-ansible/c71557b/ : SUCCESS in 10m 41s\n- kayobe-tox-molecule http://logs.openstack.org/69/655669/4/check/kayobe-tox-molecule/611904a/ : SUCCESS in 52m 47s\n- kayobe-overcloud-centos http://logs.openstack.org/69/655669/4/check/kayobe-overcloud-centos/72301a4/ : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/655669/4/check/kayobe-overcloud-upgrade-centos/43aba92/ : SUCCESS in 1h 44m 29s\n- kayobe-seed-centos http://logs.openstack.org/69/655669/4/check/kayobe-seed-centos/7cc7d65/ : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/655669/4/check/kayobe-seed-upgrade-centos/1559b8a/ : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6b52528c1e8e553dc1d37863c780c06164456071","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 14:16:18.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b62e1632dbc038ebb56109535343f831662b1ce0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:16:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c8c29c4bc3af2929fb58962ab4ac7081f85efde2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:14:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/655669/4/gate/openstack-tox-docs/b498636/html/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/69/655669/4/gate/openstack-tox-py35/59bcffa/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/69/655669/4/gate/openstack-tox-py36/22668c4/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/69/655669/4/gate/openstack-tox-pep8/c1ef2d4/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/69/655669/4/gate/openstack-tox-py27/27e992b/ : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/655669/4/gate/kayobe-tox-ansible-syntax/c294dbb/ : SUCCESS in 6m 37s\n- kayobe-tox-ansible http://logs.openstack.org/69/655669/4/gate/kayobe-tox-ansible/6225ffa/ : SUCCESS in 9m 00s\n- kayobe-tox-molecule http://logs.openstack.org/69/655669/4/gate/kayobe-tox-molecule/c3828ca/ : SUCCESS in 54m 35s\n- kayobe-overcloud-centos http://logs.openstack.org/69/655669/4/gate/kayobe-overcloud-centos/b9b67d9/ : SUCCESS in 1h 28m 43s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/655669/4/gate/kayobe-overcloud-upgrade-centos/9917665/ : SUCCESS in 1h 53m 30s\n- kayobe-seed-centos http://logs.openstack.org/69/655669/4/gate/kayobe-seed-centos/7e9d954/ : SUCCESS in 23m 40s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/655669/4/gate/kayobe-seed-upgrade-centos/6272e40/ : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d7a6749d413f444a843d3f2b89f4703015643069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:14:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"13350e3f56a7bd8819a1054553ad7c7e71171d8a","revisions":{"7941d5fce458bdde114d49e8b5f560c7e725508e":{"kind":"REWORK","_number":1,"created":"2019-04-25 11:00:36.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/69/655669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/655669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/655669/1"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 10:48:04.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 10:58:36.000000000","tz":60},"subject":"Support customising Prometheus config","message":"Support customising Prometheus config\n\nKolla Ansible supports customising Prometheus config files.\nThis commit allows Kayobe to take advantage of that.\n\nStory: 2005508\nTask: 30617\nChange-Id: I9fe28a658dc35adb5358d9a3a96bda9245404257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7941d5fce458bdde114d49e8b5f560c7e725508e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7941d5fce458bdde114d49e8b5f560c7e725508e"}]},"branch":"refs/heads/master"},"455658249179e85eb266338ef035d1c2e9f8ca60":{"kind":"REWORK","_number":2,"created":"2019-04-25 11:19:28.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/69/655669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/655669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/655669/2"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 10:48:04.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 11:19:11.000000000","tz":60},"subject":"Support customising Prometheus config","message":"Support customising Prometheus config\n\nKolla Ansible supports customising Prometheus config files.\nThis commit allows Kayobe to take advantage of that.\n\nStory: 2005508\nTask: 30617\nChange-Id: I9fe28a658dc35adb5358d9a3a96bda9245404257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/455658249179e85eb266338ef035d1c2e9f8ca60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/455658249179e85eb266338ef035d1c2e9f8ca60"}]},"branch":"refs/heads/master"},"37194666aefd87d5f65e1bfb856ebe1fc1470e95":{"kind":"REWORK","_number":3,"created":"2019-04-25 13:44:30.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/69/655669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/655669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/655669/3"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 10:48:04.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 13:11:47.000000000","tz":60},"subject":"Support customising Prometheus config","message":"Support customising Prometheus config\n\nKolla Ansible supports customising Prometheus config files.\nThis commit allows Kayobe to take advantage of that.\n\nStory: 2005508\nTask: 30617\nChange-Id: I9fe28a658dc35adb5358d9a3a96bda9245404257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37194666aefd87d5f65e1bfb856ebe1fc1470e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37194666aefd87d5f65e1bfb856ebe1fc1470e95"}]},"branch":"refs/heads/master"},"13350e3f56a7bd8819a1054553ad7c7e71171d8a":{"kind":"REWORK","_number":4,"created":"2019-06-10 14:31:26.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/69/655669/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/655669/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/655669/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/655669/4"}}},"commit":{"parents":[{"commit":"a65511a3d8902fdf84bbb91fd2e86237a43795c5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a65511a3d8902fdf84bbb91fd2e86237a43795c5"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-04-25 10:48:04.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-06-10 14:31:15.000000000","tz":60},"subject":"Support customising Prometheus config","message":"Support customising Prometheus config\n\nKolla Ansible supports customising Prometheus config files.\nThis commit allows Kayobe to take advantage of that.\n\nStory: 2005508\nTask: 30617\nChange-Id: I9fe28a658dc35adb5358d9a3a96bda9245404257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13350e3f56a7bd8819a1054553ad7c7e71171d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13350e3f56a7bd8819a1054553ad7c7e71171d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
